)]}'
{"id":"openstack%2Ftripleo-common~770671","triplet_id":"openstack%2Ftripleo-common~master~Ia244d0fe5964fb8dd43b9cf9d4242a0e2f4abee8","project":"openstack/tripleo-common","branch":"master","topic":"ipa","hashtags":[],"change_id":"Ia244d0fe5964fb8dd43b9cf9d4242a0e2f4abee8","subject":"Use the new IPA-builder element for installing python-hardware","status":"MERGED","created":"2021-01-13 21:58:57.000000000","updated":"2021-01-29 14:43:22.000000000","submitted":"2021-01-26 02:57:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"770671-ipa","meta_rev_id":"b474ab8871422db16b028d1e2d03e742076b2b0f","_number":770671,"virtual_id_number":770671,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 02:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-21 16:43:58.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2021-01-25 15:09:23.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-01-21 13:47:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-21 13:36:20.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2021-01-25 15:09:23.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 22:01:57.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-01-13 22:04:51.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2021-01-14 00:24:46.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-14 12:42:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-20 22:22:46.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2021-01-21 13:35:13.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-01-21 13:36:08.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-01-22 06:02:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"1b885b60fe55238b74f0a3d3e9aa0b7603cea5c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-13 21:58:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"afc21aa5fceef1b11e5cd827db6e465a639e705f","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-13 22:01:57.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, Thanks!","accounts_in_message":[],"_revision_number":1},{"id":"e4846fbcf0b0da02887a966b7ff0945f35472f7f","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-01-13 22:04:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"edc38460124d864b5ed1174ad65d71329db8b244","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-14 00:24:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/15aa3463cbee4abe93921978e4ec1684 : SUCCESS in 2h 24m 24s","accounts_in_message":[],"_revision_number":1},{"id":"aed284559fcfb003552e5ba7a95f3558150d23f6","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-01-14 12:20:43.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"80d1171191a293ea7f11d1bff5e49c91717d7cd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 12:42:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1abed2e40e8344f6a5812e5db4d1c3f6 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a59d8811de94b44814d900b8cbdce13 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/318416269a4749579808d0555f5ddfca : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46acf00548ba4e8780fdcd134855c9b0 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8ae665722374890876a904b4b9591f2 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15733589df3e42e1afb07da446645093 : SUCCESS in 5m 34s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4f765de1f34b4c9a877ac0f008746f8f : SUCCESS in 4h 29m 36s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/7ec690bdade842db87efb3356a9453e8 : SUCCESS in 4h 41m 53s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/90adff99c79a42c095f4b52452936c2f : SUCCESS in 1h 00m 05s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/ec85b9abd8864543a35616d7ce10de63 : SUCCESS in 51m 44s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9eb8239f12d54517a91eeeeaf0ccc57d : SUCCESS in 1h 49m 19s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/885cda1a1a5f49dca0b772501ef5cdb3 : FAILURE in 1h 50m 04s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c3e426e9df314c91bd20ae1fc2a147ca : SUCCESS in 1h 41m 13s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/78e2d3c453684028acc0c77b1b603411 : SUCCESS in 1h 09m 52s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/b26089d5b1a2416ca719f43e1c52df7e : FAILURE in 1h 39m 05s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/7336cbf259c54fc887ac8f5db928b4d0 : SUCCESS in 2h 26m 28s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/34943ded63ad4761b64de79e6533b6c6 : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29f054eb94dc4292abbc42bf7b4606d4 : SUCCESS in 5m 34s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/401185fa6d0e4bc085f666239456d91c : SUCCESS in 5m 32s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/1b66a7e19e47419ca43e07d27172cf5c : SUCCESS in 1h 00m 23s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/a0baa3b079f540b59eea9d211f5cc82a : FAILURE in 21m 10s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/906d849cdc2245438fdb4cf3e3245c00 : SUCCESS in 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"17fb7afb4f353d07c6330e9c8a225776ba272775","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-01-14 13:29:47.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"ebb5047c0368fd5ee0b3e7e742168674fad50325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-01-14 13:34:55.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"44c4a87488cb34decd7a7ae49a20f00782d92199","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-01-14 13:35:58.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cf57da30c275226ac888044593b37a590df198c2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-14 14:32:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4db3245080e087033d7f3d77d2a5d67e33c17373","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-14 16:10:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b01f1d1e17f8405ab49c7424ba2db228 : SUCCESS in 2h 28m 43s","accounts_in_message":[],"_revision_number":2},{"id":"a5c24f740c48d8d12ef295891633e5af53c913d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 04:43:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31bee3b5eb8848b880a986284f18d787 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a894832c89284823adb344870e993d71 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e275e0f4b81844878aa1cfba3fc1fb9e : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d35e01c23b24e59bcd434ca68f2dede : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbfa0b1c0f0b4b59b7699c44bc0f52ef : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc9b311612ee4d0f9cfbaea8b5c18f83 : SUCCESS in 4m 55s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/803093e4f267413e918eccf17b7c0695 : SUCCESS in 4h 05m 54s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/67a68df6ad2545d5b4565fb48928437c : SUCCESS in 4h 36m 14s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/dfbd71232be04afa92d991b982a5679f : SUCCESS in 1h 04m 06s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/28edc1d41e7347fea001c4133b097020 : SUCCESS in 55m 11s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a13ed9dde7df497b9b488b6ae6bae904 : SUCCESS in 2h 03m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/eb81f7356c84493f9d5e5b7fe1e746d6 : FAILURE in 2h 01m 09s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c16bfaabcf3f4200af72b7043962cc36 : SUCCESS in 1h 21m 18s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/adecbaa4b4eb48098affc03566219b9e : SUCCESS in 1h 22m 04s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/ea66e3ae9ab746c38dff999d974b4540 : FAILURE in 1h 51m 52s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/37b837b0d5654be2860f6873660d36d7 : SUCCESS in 2h 49m 22s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/69f0ae88d3734748a64df8056665accb : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e4e1ae9b529b46b2b35a641564de6538 : SUCCESS in 3m 54s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/260bd8ae9b3e4caeaa717782fcec4781 : SUCCESS in 4m 26s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/7c955a375dab46ff86abb942a3840de5 : SUCCESS in 48m 43s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/5dfe22c01a0d48a88cbfc6ba461a5a24 : FAILURE in 23m 00s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/314ead48be574759a1bd1898a9edc563 : SUCCESS in 32m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ff9e8a9109b8b8b4e72db052a00a8dca6bb74cac","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-01-15 10:40:13.000000000","message":"Patch Set 2: -Code-Review\n\nThis is still failing, now with element dependency issue.\n\n  2021-01-15 00:58:27.652 | diskimage-builder version 3.5.1\n  2021-01-15 00:58:27.654 | Building elements: base centos ironic-agent ironic-agent-multipath dynamic-login element-manifest network-gateway enable-packages-install override-pip-and-virtualenv extra-hardware selinux-permissive interface-names\n  2021-01-15 00:58:27.736 | DEBUG diskimage_builder.element_dependencies [-] ELEMENTS_PATH is: /home/zuul/workspace/venv/lib64/python3.6/site-packages/diskimage_builder/elements:/home/zuul/src/opendev.org/openstack/ironic-python-agent-builder/dib/:/home/zuul/src/opendev.org/openstack/heat-agents:/home/zuul/src/opendev.org/openstack/tripleo-image-elements/elements:/home/zuul/src/opendev.org/openstack/tripleo-puppet-elements/elements _find_all_elements /home/zuul/workspace/venv/lib64/python3.6/site-packages/diskimage_builder/element_dependencies.py:209\n  2021-01-15 00:58:27.779 | Traceback (most recent call last):\n  2021-01-15 00:58:27.779 |   File \"/home/zuul/workspace/venv/lib64/python3.6/site-packages/diskimage_builder/lib/element-info.py\", line 7, in \u003cmodule\u003e\n  2021-01-15 00:58:27.779 |     sys.exit(main())\n  2021-01-15 00:58:27.780 |   File \"/home/zuul/workspace/venv/lib64/python3.6/site-packages/diskimage_builder/element_dependencies.py\", line 343, in main\n  2021-01-15 00:58:27.780 |     elements \u003d _get_elements(args.elements)\n  2021-01-15 00:58:27.780 |   File \"/home/zuul/workspace/venv/lib64/python3.6/site-packages/diskimage_builder/element_dependencies.py\", line 254, in _get_elements\n  2021-01-15 00:58:27.780 |     return _expand_element_dependencies(elements, all_elements)\n  2021-01-15 00:58:27.780 |   File \"/home/zuul/workspace/venv/lib64/python3.6/site-packages/diskimage_builder/element_dependencies.py\", line 162, in _expand_element_dependencies\n  2021-01-15 00:58:27.780 |     (provide, provided_by[provide]))\n  2021-01-15 00:58:27.780 | diskimage_builder.element_dependencies.AlreadyProvidedException: ironic-python-agent: already provided by [\u0027ironic-agent\u0027]","accounts_in_message":[],"_revision_number":2},{"id":"a84a446a9962fc6765f6aced033225e4acc53df6","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-15 20:38:50.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: -Code-Review\n\u003e \n\u003e This is still failing, now with element dependency issue.\n\u003e \n\u003e   2021-01-15 00:58:27.652 | diskimage-builder version 3.5.1\n\u003e   2021-01-15 00:58:27.654 | Building elements: base centos ironic-agent ironic-agent-multipath dynamic-login element-manifest network-gateway enable-packages-install override-pip-and-virtualenv extra-hardware selinux-permissive interface-names\n\u003e   2021-01-15 00:58:27.736 | DEBUG diskimage_builder.element_dependencies [-] ELEMENTS_PATH is: /home/zuul/workspace/venv/lib64/python3.6/site-packages/diskimage_builder/elements:/home/zuul/src/opendev.org/openstack/ironic-python-agent-builder/dib/:/home/zuul/src/opendev.org/openstack/heat-agents:/home/zuul/src/opendev.org/openstack/tripleo-image-elements/elements:/home/zuul/src/opendev.org/openstack/tripleo-puppet-elements/elements _find_all_elements /home/zuul/workspace/venv/lib64/python3.6/site-packages/diskimage_builder/element_dependencies.py:209\n\u003e   2021-01-15 00:58:27.779 | Traceback (most recent call last):\n\u003e   2021-01-15 00:58:27.779 |   File \"/home/zuul/workspace/venv/lib64/python3.6/site-packages/diskimage_builder/lib/element-info.py\", line 7, in \u003cmodule\u003e\n\u003e   2021-01-15 00:58:27.779 |     sys.exit(main())\n\u003e   2021-01-15 00:58:27.780 |   File \"/home/zuul/workspace/venv/lib64/python3.6/site-packages/diskimage_builder/element_dependencies.py\", line 343, in main\n\u003e   2021-01-15 00:58:27.780 |     elements \u003d _get_elements(args.elements)\n\u003e   2021-01-15 00:58:27.780 |   File \"/home/zuul/workspace/venv/lib64/python3.6/site-packages/diskimage_builder/element_dependencies.py\", line 254, in _get_elements\n\u003e   2021-01-15 00:58:27.780 |     return _expand_element_dependencies(elements, all_elements)\n\u003e   2021-01-15 00:58:27.780 |   File \"/home/zuul/workspace/venv/lib64/python3.6/site-packages/diskimage_builder/element_dependencies.py\", line 162, in _expand_element_dependencies\n\u003e   2021-01-15 00:58:27.780 |     (provide, provided_by[provide]))\n\u003e   2021-01-15 00:58:27.780 | diskimage_builder.element_dependencies.AlreadyProvidedException: ironic-python-agent: already provided by [\u0027ironic-agent\u0027]\n\nSo I guess diskimage-builder became more strict at some point?! :(","accounts_in_message":[],"_revision_number":2},{"id":"ac2342536cbef1e2c7e1aa55c167d10e50a2beed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-15 20:43:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f0cf73f4de7148cefdc3ea104b134d339d053ee1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-15 20:43:45.000000000","message":"Patch Set 3:\n\nlet\u0027s try ditching the ironic-agent in favor of the newer ironic-python-agent","accounts_in_message":[],"_revision_number":3},{"id":"2c3e75e9aea0ea961aba40dd1b989071dd2a9575","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-15 23:10:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/26d6c311a9f84f46bb1f86b056b41026 : SUCCESS in 2h 25m 50s","accounts_in_message":[],"_revision_number":3},{"id":"da45553833d712ed6a251fb489e8d7302b4877ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-16 00:32:01.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0d2132652684886822990a9cfa45d96 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e713325b8b3e44a2b68e51e4f8ded5ae : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/331517b06dd740d1b615467f92bed9c8 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5da098bf81c445a9a9c0b21a510526fd : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc146e006ef64841bb9a3770375ba458 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac8079ad352f4841bec287032cccdcd0 : SUCCESS in 4m 09s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7a747f5e2ad2491f938e168e898df978 : SUCCESS in 3h 21m 08s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/307340ae530547adaeee77476f92e1e1 : SUCCESS in 2h 11m 33s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/fc8e2ff3a5084a43a70f7c9a0bbf5303 : SUCCESS in 51m 47s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/d7ba32694257478291ab9f5996e5c57b : SUCCESS in 57m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/21112c9646a743f6b9948522a5066340 : SUCCESS in 2h 07m 14s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/a2665ec6a13944c3b5d50137dd31c283 : SUCCESS in 1h 56m 47s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/dc2d21672d3b4bbbba405e03040871d0 : SUCCESS in 1h 27m 20s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/06d5be7e29c64af28bf5b0cf6c0395d7 : SUCCESS in 1h 15m 37s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/00675fd5a95f4d04b246b1f8c43a2b08 : FAILURE in 33m 42s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/0e389fe8114b4554bfe7bd56dbadb91f : FAILURE in 35m 30s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8c4fff917be6439baad0b400a7945199 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5ff7678b11d45599208ed6af9dfa6db : SUCCESS in 4m 19s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/9167e1614eb04f75825d5e1034dc2c59 : SUCCESS in 4m 44s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/0f72bce430594c35adb3de934da28fb1 : SUCCESS in 57m 04s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/1ddfa31731e1401bb9974ab602f39449 : FAILURE in 18m 34s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/fbc7df14e13f49c7b36ad5a0ae93b122 : SUCCESS in 40m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fafcfdb45e7d56138c665edecdccd9bde034f324","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-01-20 22:22:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"71f380c2d0838bbe04332ba5e26beade7cde3846","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-01-21 13:35:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"87fbd8fca294b455c4f03fef0168b8037117e2d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-01-21 13:36:01.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d7b4a70feb72744fd2f0254816c11508cbb143c2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-01-21 13:36:20.000000000","message":"Patch Set 4: Code-Review+2\n\nFixed a trivial issue, LGTM now.","accounts_in_message":[],"_revision_number":4},{"id":"dd0d6276bd3912fde11db641c773e3847177c11e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-01-21 13:47:03.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"11ba5fb3ee927949c026eb0a0432d6f6999a4c67","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-21 16:43:58.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/64f2937c6d44404ab4f20d0f5c3f7fa6 : SUCCESS in 3h 06m 19s","accounts_in_message":[],"_revision_number":4},{"id":"4296280b4dd7609869da85f9479d00d3280ce983","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 21:04:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f773ceaa9acf436bb899a887553e3e43 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b11db42a7a71476482f9235bc7cd3077 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c95e9a9048c4649951b994a59a630fc : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0df40cf7a91458d8d51cc01ba28d918 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/015ee7d990974561b071424240988d8e : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/715f174b8b244b6292833e0d39daeb7b : SUCCESS in 4m 36s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8c913f23ba6042279ef7f048ab252ce2 : SUCCESS in 4h 10m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a8b26734155a4ec9a7c430c7c3d06fae : SUCCESS in 2h 02m 17s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/25aeee6540ce446ab055f9d7a6a0dd38 : SUCCESS in 1h 34m 24s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2b77d4bfe1eb4ad0b7edd70ec9d671d7 : SUCCESS in 1h 15m 35s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/ed93d7890e1e4e48a896ad1219146dcf : SUCCESS in 4h 25m 12s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/ed1c327d2e684c7f847f15673b4dc32b : SUCCESS in 2h 31m 50s (non-voting)\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/00a0f9c95670456e8497205fd2b82445 : SUCCESS in 1h 46m 36s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/afa688e9023b4716b1bb62b6b9199ce7 : SUCCESS in 1h 53m 13s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2bbe0d3b051640aa872629b9148dad6f : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3d3da099a084e419cf234302a12b5d1 : SUCCESS in 5m 30s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/90df02a807e1414999d90746f5947d8f : SUCCESS in 5m 37s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/e57a8ce030984f4a8b817b38a6bf2a63 : SUCCESS in 57m 02s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/d913fc30191c449b9b4f996fb0c2d0d6 : FAILURE in 26m 34s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/a4c47c2368154c61a6352f0b9b955da6 : SUCCESS in 35m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e3d29d0e19d6295b435e7c1bc54dab71cdf7ae2c","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-01-22 06:02:19.000000000","message":"Patch Set 4:\n\nIPA build fails with, i see that rpm is not in CentOS8 base repos, it used to be in CentOS7 though:- 2021-01-21 17:18:53.412 | \u003e Error: Unable to find a match: sdparm\n\nhttps://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_d91/770671/4/check/tripleo-buildimage-ironic-python-agent-centos-8/d913fc3/build.log","accounts_in_message":[],"_revision_number":4},{"id":"6471a052a3ffe17e6cbfa4249fdef03c44d2a045","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-01-22 10:01:33.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e IPA build fails with, i see that rpm is not in CentOS8 base repos, it used to be in CentOS7 though:- 2021-01-21 17:18:53.412 | \u003e Error: Unable to find a match: sdparm\n\u003e \n\u003e https://storage.bhs.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_d91/770671/4/check/tripleo-buildimage-ironic-python-agent-centos-8/d913fc3/build.log\n\nI think sdparm (the missing package) can be found only in epel now and we\u0027re disabling the epel repo by default.\nIf it was in centos7, maybe we can get it for centos8 as well ?\nOtherwise we\u0027ll have to install it from epel in some way.","accounts_in_message":[],"_revision_number":4},{"id":"259dcd66eca31dce8226d6c783613d8cbd914583","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-01-22 11:37:31.000000000","message":"Patch Set 4:\n\nI\u0027m looking at the recent python-hardware sources, and it doesn\u0027t seem to use sdparm. Let me look into it.","accounts_in_message":[],"_revision_number":4},{"id":"b892e71d23e8a1d047805aa3dfa92a20d1a33db1","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-01-22 11:42:43.000000000","message":"Patch Set 4:\n\nFix: https://review.opendev.org/c/openstack/ironic-python-agent-builder/+/771965","accounts_in_message":[],"_revision_number":4},{"id":"ccca30027e81e1f2c94d9c8f195f50b5c8d03355","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-01-23 17:25:38.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"b571206ce1590200681671222c7cd7ebb11659ee","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-23 20:30:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/84b3dd1b88604b82ab7d89e570c867a1 : SUCCESS in 3h 02m 53s","accounts_in_message":[],"_revision_number":4},{"id":"835b659604a247b44ef79bbe9a1a88193fd35428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-24 02:19:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e40cbb8ba3f34483b342226fff39631d : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cfa51d04fb149ddadaf266385b820c0 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92cbe83c1fa440ebad77ab94628502cf : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5b71b45042148358e75df0ea1c7fc91 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88cd0c045b534b8cab8a11a9bbbcf57a : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b526d65b9ae4baea207adb404adae42 : SUCCESS in 3m 35s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f45c1178dba5403d8f1b367841fe0f70 : SUCCESS in 8h 43m 24s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d3968e697ab640c49fe41d2a5d4f816a : SUCCESS in 1h 59m 20s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/483e9d23d4ea4c5681764f2eb1c711e2 : SUCCESS in 1h 25m 50s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/56fbbc9f70c84879abfe9c099b4b04ec : SUCCESS in 1h 04m 18s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/d116f18f20574ecba49b6d16ba8f8456 : SUCCESS in 8h 46m 25s\n- tripleo-ci-centos-8-standalone-upgrade https://zuul.opendev.org/t/openstack/build/111fbbc431914f1a91debe17605c45d5 : SUCCESS in 2h 14m 07s (non-voting)\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/2d1bdbcd61f54dc99061d3e8ed0e1249 : SUCCESS in 1h 26m 14s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades https://zuul.opendev.org/t/openstack/build/1e110ca76f9442ea8c7d6c8d389bfb26 : SUCCESS in 1h 31m 44s (non-voting)\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8a01eb54d414400f91b3576b659ea9ae : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22dc386979a24f71b620b5a96ae5859b : SUCCESS in 7m 13s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/cbd16aeb5dc4483eb6a6192d732a4d72 : SUCCESS in 4m 52s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/99dc1df0936244b881bcd20718552c52 : SUCCESS in 54m 46s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/742c57fa4a5c4143b01f8f33356623f7 : SUCCESS in 32m 26s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/7bace6e4cfd145e3afbd1c3cce96a599 : SUCCESS in 37m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"77c97cf4c9c8f2bd8400f87917f32b7100aeed0b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-25 15:09:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"944fe6e16cd9a4530519da7e6582245c441a0e30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 15:09:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dba7ba0eeb7574d6e2a8b3f12b3faaa7981d26fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 02:57:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1a6b7bd81f04c14a611535d0eaea3e1 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f01dce383964422bfe51fe8ec3ec39f : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7cb82c6071a94b69bf51b7acb2358bf1 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/867bec4e1cd84ee689c21a883e5d3f3c : SUCCESS in 5m 11s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d70776e7d3ae40e6acc36b1bc998670b : SUCCESS in 5h 10m 12s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b03068d2b9c541fbb3a4fcbfac8cd9ae : SUCCESS in 1h 45m 42s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/22b9b4bec0b0424cb9c455fe52ee4fb7 : SUCCESS in 1h 18m 49s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9535aae8578d4115bcdf48fcb1cdc04d : SUCCESS in 1h 03m 55s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/311f521a013c4b288016541872dd806a : SUCCESS in 1h 46m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3aa24ce5074c4791a1fc5e1d1cc0e09e : SUCCESS in 5m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4f8ba287ffc341359ed5c3d2fe120656 : SUCCESS in 5m 08s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/35a4ebc81c1843b394cef11b95cbe12e : SUCCESS in 4m 52s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/a5f1cdc865e04560a7c81b68df059969 : SUCCESS in 54m 59s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/b5d6ea6ccc744f74a7bb446aa1d3ff8b : SUCCESS in 33m 24s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/9a84d6e4e22b42bd98a54d4a430762b1 : SUCCESS in 33m 59s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3dc3237aec95e0735ec28854b25f416317184aac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 02:57:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ec96a3005466b17978ccdf38e0a460c32aa58513","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 03:00:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d60aad0eee543a1ac94aec6e53fd779 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":4},{"id":"d7ff39821fd3b0574449b76d4cbe148970f5b710","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-01-29 14:40:58.000000000","message":"Patch Set 4:\n\nBackport candidate considering https://review.opendev.org/c/openstack/diskimage-builder/+/771808?\nWould also need to include https://review.opendev.org/c/openstack/tripleo-common/+/753972","accounts_in_message":[],"_revision_number":4},{"id":"b474ab8871422db16b028d1e2d03e742076b2b0f","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-01-29 14:43:22.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e Backport candidate considering https://review.opendev.org/c/openstack/diskimage-builder/+/771808?\n\u003e Would also need to include https://review.opendev.org/c/openstack/tripleo-common/+/753972\n\nhttps://review.opendev.org/c/openstack/tripleo-common/+/753972 should be applicable only to train as centos7?","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f938e6ee9f1d399c99db3b3dbd718a31587f4aea","revisions":{"18f85c9afc889b2361246db4cd2c4c8bba60f901":{"kind":"REWORK","_number":1,"created":"2021-01-13 21:58:57.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/71/770671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/71/770671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/1"}}},"commit":{"parents":[{"commit":"5fb9312eda189c63a50b7bddb8154a52ef867a62","subject":"Merge \"pylint: fixed logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5fb9312eda189c63a50b7bddb8154a52ef867a62"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-01-13 21:58:11.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-01-13 21:58:11.000000000","tz":-420},"subject":"Use the new IPA-builder element for installing python-hardware","message":"Use the new IPA-builder element for installing python-hardware\n\nThe python3 file was mieed in Id75c11e5b7c4f3bf73c962bc7e46f5a6021efc5f\n\nChange-Id: Ia244d0fe5964fb8dd43b9cf9d4242a0e2f4abee8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/18f85c9afc889b2361246db4cd2c4c8bba60f901"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/18f85c9afc889b2361246db4cd2c4c8bba60f901"}]},"branch":"refs/heads/master"},"076d5d9a5640825351683ec94e8aa6415d733127":{"kind":"REWORK","_number":2,"created":"2021-01-14 13:34:55.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/71/770671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/71/770671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/2"}}},"commit":{"parents":[{"commit":"5fb9312eda189c63a50b7bddb8154a52ef867a62","subject":"Merge \"pylint: fixed logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5fb9312eda189c63a50b7bddb8154a52ef867a62"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-01-13 21:58:11.000000000","tz":-420},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2021-01-14 13:34:33.000000000","tz":0},"subject":"Use the new IPA-builder element for installing python-hardware","message":"Use the new IPA-builder element for installing python-hardware\n\nThe python3 file was mieed in Id75c11e5b7c4f3bf73c962bc7e46f5a6021efc5f\n\nChange-Id: Ia244d0fe5964fb8dd43b9cf9d4242a0e2f4abee8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/076d5d9a5640825351683ec94e8aa6415d733127"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/076d5d9a5640825351683ec94e8aa6415d733127"}]},"branch":"refs/heads/master"},"e9ab0eb9790fe89dfb06a050b806af66fa710d06":{"kind":"REWORK","_number":3,"created":"2021-01-15 20:43:22.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/71/770671/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/71/770671/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/3"}}},"commit":{"parents":[{"commit":"5fb9312eda189c63a50b7bddb8154a52ef867a62","subject":"Merge \"pylint: fixed logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5fb9312eda189c63a50b7bddb8154a52ef867a62"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-01-13 21:58:11.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-01-15 20:43:19.000000000","tz":-420},"subject":"Use the new IPA-builder element for installing python-hardware","message":"Use the new IPA-builder element for installing python-hardware\n\nThe python3 file was mieed in Id75c11e5b7c4f3bf73c962bc7e46f5a6021efc5f\n\nChange-Id: Ia244d0fe5964fb8dd43b9cf9d4242a0e2f4abee8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e9ab0eb9790fe89dfb06a050b806af66fa710d06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e9ab0eb9790fe89dfb06a050b806af66fa710d06"}]},"branch":"refs/heads/master"},"f938e6ee9f1d399c99db3b3dbd718a31587f4aea":{"kind":"REWORK","_number":4,"created":"2021-01-21 13:36:01.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/71/770671/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/71/770671/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/71/770671/4"}}},"commit":{"parents":[{"commit":"5fb9312eda189c63a50b7bddb8154a52ef867a62","subject":"Merge \"pylint: fixed logging\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5fb9312eda189c63a50b7bddb8154a52ef867a62"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-01-13 21:58:11.000000000","tz":-420},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-01-21 13:35:41.000000000","tz":0},"subject":"Use the new IPA-builder element for installing python-hardware","message":"Use the new IPA-builder element for installing python-hardware\n\nThe python3 file was mieed in Id75c11e5b7c4f3bf73c962bc7e46f5a6021efc5f\n\nChange-Id: Ia244d0fe5964fb8dd43b9cf9d4242a0e2f4abee8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f938e6ee9f1d399c99db3b3dbd718a31587f4aea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f938e6ee9f1d399c99db3b3dbd718a31587f4aea"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
