)]}'
{"id":"openstack%2Ftripleo-common~772411","triplet_id":"openstack%2Ftripleo-common~stable%2Fvictoria~If56cd0fa986ef193d036b73e0ab84a88d59147e3","project":"openstack/tripleo-common","branch":"stable/victoria","topic":"healthcheck/http-stable/victoria","hashtags":[],"change_id":"If56cd0fa986ef193d036b73e0ab84a88d59147e3","subject":"Conditionally use python instead of cURL","status":"MERGED","created":"2021-01-26 11:52:37.000000000","updated":"2021-01-30 18:33:43.000000000","submitted":"2021-01-30 18:31:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772411-healthcheck/http-stable/victoria","cherry_pick_of_change":771422,"cherry_pick_of_patch_set":10,"meta_rev_id":"81768e27a68af12ea474756897608e05a6771794","_number":772411,"virtual_id_number":772411,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-30 18:31:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-26 14:24:44.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-01-29 21:49:16.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2021-01-29 21:49:16.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 14:24:44.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 17:15:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-29 21:49:16.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"905ad2712f24643f55134910b489a86a24595108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-01-26 11:52:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"01c1627d42e89eb1b7efd36ffd0ebf00e01e7299","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-01-26 14:24:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6194ea0205734d4f9558b77e504356cd : SUCCESS in 2h 29m 56s","accounts_in_message":[],"_revision_number":1},{"id":"ec37ac0d03330cd3e499db43ac1a5dadcfde8efb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 17:15:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/568afcfacdd745f1af23991b805ae61b : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e53f73bbb04b6698d144baecc3881b : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74ce5f4baf5c40dd999d85c65a1a9d5b : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e4d5ea844994d468495ca51bdf6d8cf : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3df530fde372453c8340df5ec740fd54 : SUCCESS in 4m 07s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/48f04e7d3105429283762017a16c2d01 : SUCCESS in 3h 52m 23s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e593e7934026430897fd5c77438daa6d : SUCCESS in 1h 54m 17s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b873e55168b649adb751665875e8dcdc : SUCCESS in 1h 23m 01s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9ab19cbe499545339716fcd52e691458 : SUCCESS in 1h 22m 02s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2cca38a3d0cf400eb1e462e3b5c872a1 : SUCCESS in 5h 18m 02s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/50f87dbbf67141b0858b9da8dd320194 : SUCCESS in 2h 35m 42s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/4972366051e04df99050677a0dcd2656 : SUCCESS in 1h 54m 07s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/c2589e6bd9a94d70bb5011f16e999980 : SUCCESS in 2h 14m 39s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/52eac887fba14f97aea60be977ae8a8b : SUCCESS in 1h 54m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/db8e2f2f09934963bc525d643bc08934 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5f76a038d944e1ebb9d8afe5861a217 : SUCCESS in 4m 51s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/a9046565661840f1818158aa935442cf : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":1},{"id":"d01f393439382e2f66779395f8e766db5173db12","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-29 21:49:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9c4785c75e5fdc15808a4cd97fb67b27c843eec6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 12:31:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f04c7577422d6e26457803d621a59cda2c46b47a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 18:31:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09f6b8a559bf4acf93148697ba31f476 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08bfa930d6624729aa89fbb39fc4cc7d : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d244dc273c7b4749a627aa0e3e63a203 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42653d179f744db6ae4abaf38a414580 : SUCCESS in 4m 11s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/19d26473822a4214b252be2ba053c07d : SUCCESS in 3h 58m 32s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/76ff7e2cb2fd46f1bbb038ce41c53e52 : SUCCESS in 2h 00m 54s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9b75c9e705334f8a8ea4acfc5d507bde : SUCCESS in 1h 42m 08s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d6ce0518ca394a69ba07fe528f14f036 : SUCCESS in 1h 27m 23s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/5b0853c595d74761849172c33c2a3c8b : SUCCESS in 4h 02m 11s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/d6ddd59d43f64d4cac3ae6c0fcf3b565 : SUCCESS in 2h 26m 31s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/679bcea58f2748e5bc83625263af17a5 : SUCCESS in 1h 42m 52s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/9f07e8e5c2b847ff9a39f669bcd66a41 : SUCCESS in 2h 00m 10s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/146fa1f56f4a4ec8a4a755b9858cc6e4 : SUCCESS in 1h 54m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/444aea2aadd4456abf6c95dbf6574e0d : SUCCESS in 4m 38s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/700490ea4e21489497e9f7c37d941787 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"5b7158824eec5cdb96d7bc4a3ccb6999e2a2dacf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 18:31:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"81768e27a68af12ea474756897608e05a6771794","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-30 18:33:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a307cfdeeefe42738ef7b80cbfa5a7a3 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c71108ae5d50fe7d78b81ed770e75a7a293e6f0","revisions":{"7c71108ae5d50fe7d78b81ed770e75a7a293e6f0":{"kind":"REWORK","_number":1,"created":"2021-01-26 11:52:37.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/11/772411/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/11/772411/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/11/772411/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/11/772411/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/11/772411/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/11/772411/1"}}},"commit":{"parents":[{"commit":"5668f2397500847590f6c682681591247c1f141e","subject":"TCIB: Add cinder-volume extend_start.sh script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5668f2397500847590f6c682681591247c1f141e"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-01-19 10:41:30.000000000","tz":60},"committer":{"name":"Cedric Jeanneret","email":"cjeanner@redhat.com","date":"2021-01-26 11:52:37.000000000","tz":0},"subject":"Conditionally use python instead of cURL","message":"Conditionally use python instead of cURL\n\nThe healthcheck_curl has always been a bit bonky with the proxies\nmanagement, especially the \"no_proxy\" environment variable: since there\nisn\u0027t any real RFC describing the content of this variable, there isn\u0027t\nany unified handling.\n\nThis leads to situation where an Operator puts some CIDR notation inside\nthis variable, while curl doesn\u0027t handle it.\n\nThis change adds a new python script. Since it uses python-requests, it\nwill have a proper support for the CIDR notation[1].\n\nIt will be called if and only if either NO_PROXY or no_proxy environment\nvariables are set and non-empty.\n\nOne can also force the script usage by passing HEALTHCHECK_CURL_PY\u003d1,\nsuch as:\npodman exec -e \"HEALTHCHECK_CURL_PY\u003d1\" container /healthchecks/you-healthcheck\n\nThe output is 100% iso-compatible, and if someone wants to know what\nhealthcheck is used, they can check the User-Agent on the target logs.\n\nThis change is motivated by, at least, the following existing issues:\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1837458\nhttps://bugzilla.redhat.com/show_bug.cgi?id\u003d1883657\n\n[1] https://github.com/psf/requests/blob/589c4547338b592b1fb77c65663d8aa6fbb7e38b/requests/utils.py#L663-L684\n\nChange-Id: If56cd0fa986ef193d036b73e0ab84a88d59147e3\n(cherry picked from commit 2185d9a8599f4efb79fe38b188bc48e931eba625)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7c71108ae5d50fe7d78b81ed770e75a7a293e6f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7c71108ae5d50fe7d78b81ed770e75a7a293e6f0"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
