)]}'
{"id":"openstack%2Ftripleo-common~775107","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~I259a0d5fcbae8f8a6bbeffa5fdb744d3291315ad","project":"openstack/tripleo-common","branch":"stable/train","topic":"bug/1913782-stable/train","hashtags":[],"change_id":"I259a0d5fcbae8f8a6bbeffa5fdb744d3291315ad","subject":"Handle containers without a namespace","status":"MERGED","created":"2021-02-11 14:17:38.000000000","updated":"2021-03-04 20:37:44.000000000","submitted":"2021-03-04 20:36:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775107-bug/1913782-stable/train","cherry_pick_of_change":773166,"cherry_pick_of_patch_set":1,"meta_rev_id":"20f70a9d0338d32dfc44d7949162548a4d0daccd","_number":775107,"virtual_id_number":775107,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 20:36:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-15 17:33:38.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":2,"date":"2021-02-16 00:44:42.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2021-03-04 14:54:20.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 20:28:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-15 17:33:38.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-16 00:44:42.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2021-03-04 14:54:20.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"}],"messages":[{"id":"da867c08d30955349dbed9e3ccecd677f536ce3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-11 14:17:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0829a94c7b2df0c7c23861a45e13dde079a7b9aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 20:28:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4795ca657794119afe5a14ce64a6f13 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b2804a8da04215aa93529b87e3e3d5 : SUCCESS in 5m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbaed24de2034c32a7043b5337d24cf2 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/72037a96eec14964ad43417d80bcaf75 : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f95a69ceed44682ab26bbd83d5d9f06 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54c9f8b5076844d7991d53070a75e875 : SUCCESS in 6m 15s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5909ffbb2ba347e38d2c1376e6f59de5 : SUCCESS in 2h 52m 11s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/85e34cc8f258424192869f00bf0030b0 : SUCCESS in 3h 45m 33s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/68210ce10f2f42178c1dd8fb503658e8 : FAILURE in 1h 52m 14s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/06c57186e3c24bd3af269f826ae5ab15 : FAILURE in 56m 30s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c0e50d70cac84b589a023f41b137dfc2 : FAILURE in 2h 06m 50s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/65e25531775f4976b59b2d2b1026a2ac : FAILURE in 1h 15m 07s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/2273a72dab2747e0983dd00c78a87461 : SUCCESS in 59m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7d470b1bbbd84442bfbf954d9576a43c : SUCCESS in 7m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0b64d65d60c0ecb3c88c007041c3c42ce0b3f0eb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-15 14:51:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f7cd6fc97bcde812e5c323cf0404df7057119817","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-15 17:33:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5161ac51ffc04148b344d0c1b1b5c23b : SUCCESS in 2h 39m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9affbe0c255f5a63f127b8c3e1766285e23e7608","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 22:47:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a921f70fdad401f92acee51b7f0b97b : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/287023c382bf479b83c876930e3bb28d : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f667db8b1bae490db9d93f715bb6a149 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4d750450e74403794eff8f13724ec74 : SUCCESS in 5m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19763b4cfa9543969b484e9aed86f404 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/907b34d7b5e147c496812ee74e729c01 : SUCCESS in 4m 18s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0c1ae39032cc4928874bd5236bc4e097 : SUCCESS in 3h 36m 31s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/7d0e8e28e8d34b9399aaffbbacce89b4 : SUCCESS in 3h 58m 01s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/58de31c644994fa5977cacf50ddf6548 : SUCCESS in 2h 36m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/67e83fcd04e147b3b67990b662fa9902 : SUCCESS in 1h 46m 00s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/d0c8180f843e4c1cae92cea358a7784c : SUCCESS in 2h 24m 24s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7d7715ae01454d7e9f3893fdc7035f1d : SUCCESS in 1h 19m 51s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/cbdc47038f574ecfa57d2d78b0b0077c : SUCCESS in 1h 03m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00cdd858baa3489ca7c225164b36b23c : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6e92e3fafa41d320bb4714d94247fa4cc70579d9","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-02-16 00:44:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"42243409a3047a47709d565a653fc3fddcfec26f","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-03-04 14:54:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cbbf429ffd599b0618294e21efe308ad95cbe86f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 15:08:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a585db636a85c8b0c247993fff3795bbaf82686","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 20:36:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4792ed7945b64dac99ab4ca4ebb39d7d : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e016fc3fd4304e8eb1853bd7a02a265e : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96d6b687fb7e4094b1e9ff8576b0d432 : SUCCESS in 6m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf1ad0ec55064084b48f94f2e43eec2f : SUCCESS in 14m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95a962bc5a54fe38e830d069a4b4046 : SUCCESS in 6m 20s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/88013b22f8f64561bc93d4621e7a4684 : SUCCESS in 3h 09m 24s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/c934d56c07cf40c5a323c2913f678855 : SUCCESS in 3h 45m 20s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/59bbb70f55f0424d8db81508b25d7867 : SUCCESS in 2h 03m 43s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/052b04a96ed848cf936442485f8e328a : SUCCESS in 2h 19m 26s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4057b906c837439e8fdfb751b47eeed5 : SUCCESS in 1h 30m 26s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/29ddb9b84ea24c95bb9275432d25e850 : SUCCESS in 1h 01m 00s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/899a480399754f32868b74a903cfff5c : SUCCESS in 1h 11m 39s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/3632619ff214460ebdb0b6ea426bc9d3 : SUCCESS in 1h 58m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dea2fe362c7b40a2b5d5927734b0b109 : SUCCESS in 13m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1f03aa05e1bcf177c2547fb7107d16df72c2db23","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 20:36:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"20f70a9d0338d32dfc44d7949162548a4d0daccd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 20:37:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf7f805b74894053aaad0fd37bb23e74 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d7a983caa44f08ff62c707b942ec186728e2506","revisions":{"7d7a983caa44f08ff62c707b942ec186728e2506":{"kind":"REWORK","_number":1,"created":"2021-02-11 14:17:38.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/07/775107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/07/775107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/07/775107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/07/775107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/07/775107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/07/775107/1"}}},"commit":{"parents":[{"commit":"e5d483b7edb718ab07134be597ee90170c3891c2","subject":"Merge \"Fix placement endpoint check when endpoints are not set\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e5d483b7edb718ab07134be597ee90170c3891c2"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-01-29 18:34:27.000000000","tz":-420},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-02-11 14:17:38.000000000","tz":0},"subject":"Handle containers without a namespace","message":"Handle containers without a namespace\n\nSatellite server does not include a namespace when exposing containers.\nThis change fixes our image-serve metadata creation to understand this\nconcept and build the _catalog we provide correctly. We will only handle\ntwo types of containers path structures with the image-serve.\n\n   * host:port/namespace/container:tag\n   * host:port/container:tag\n\nChange-Id: I259a0d5fcbae8f8a6bbeffa5fdb744d3291315ad\nDepends-On: https://review.opendev.org/c/openstack/tripleo-ansible/+/775158\nCloses-Bug: #1913782\n(cherry picked from commit 7e797317a33a5c5881a1e24b9666ae774ca224c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7d7a983caa44f08ff62c707b942ec186728e2506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7d7a983caa44f08ff62c707b942ec186728e2506"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
