)]}'
{"id":"openstack%2Ftripleo-common~776674","triplet_id":"openstack%2Ftripleo-common~master~I56cabb9aa8483c5039b5920db96a00509dc730cc","project":"openstack/tripleo-common","branch":"master","topic":"tripleo-ansible-ee","attention_set":{},"removed_from_attention_set":{"6816":{"account":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"last_update":"2022-09-15 16:04:56.000000000","reason":"Change was submitted"},"11491":{"account":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"last_update":"2022-09-15 16:04:56.000000000","reason":"Change was submitted"},"32968":{"account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"last_update":"2022-09-15 12:02:02.000000000","reason":"\u003cGERRIT_ACCOUNT_32968\u003e replied on the change","reason_account":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-09-15 16:04:56.000000000","reason":"Change was submitted"},"7353":{"account":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"last_update":"2022-09-15 16:04:56.000000000","reason":"Change was submitted"},"6926":{"account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"last_update":"2022-09-15 12:26:31.000000000","reason":"\u003cGERRIT_ACCOUNT_6926\u003e replied on the change","reason_account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}}},"hashtags":[],"change_id":"I56cabb9aa8483c5039b5920db96a00509dc730cc","subject":"Add an Ansible execution environment to our Build","status":"MERGED","created":"2021-02-19 17:00:54.000000000","updated":"2022-09-16 02:46:19.000000000","submitted":"2022-09-15 16:04:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":44,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"776674-tripleo-ansible-ee","meta_rev_id":"c8b2c3e94ad2361f2d93d32d8801da96a939f80f","_number":776674,"virtual_id_number":776674,"owner":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-15 16:04:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-09-14 15:07:31.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"recommended":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"all":[{"value":1,"date":"2022-09-15 12:02:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2022-09-15 12:26:31.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":2,"date":"2022-09-15 12:07:26.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":1,"date":"2022-09-16 02:46:19.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-15 12:30:54.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"}],"CC":[{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 17:23:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 17:26:50.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 22:16:10.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2021-08-03 16:30:39.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"CC"},{"updated":"2022-02-23 16:52:06.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REMOVED"},{"updated":"2022-05-03 13:58:16.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"CC"},{"updated":"2022-05-24 12:43:16.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2022-07-18 15:14:42.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"CC"},{"updated":"2022-07-18 22:44:44.000000000","updated_by":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"reviewer":{"_account_id":1955,"name":"Alan Pevec","email":"alan.pevec@redhat.com","username":"apevec"},"state":"CC"},{"updated":"2022-07-19 07:47:05.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"CC"},{"updated":"2022-07-25 14:33:34.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2022-09-15 12:02:02.000000000","updated_by":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"reviewer":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"state":"REVIEWER"},{"updated":"2022-09-15 12:07:26.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2022-09-16 02:46:19.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"}],"messages":[{"id":"7b81d863373242019de3033199ffda010ee224fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-02-19 17:00:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc382a54828ffe3bcb3442edea70fe6d63dafc72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 17:23:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04cae156094a4ca893353ea28d32936a : FAILURE in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/684a05452494495388bd65b7e6f92ab9 : POST_FAILURE in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eaea48446fea46cd922260f802437e07 : FAILURE in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26e94025d0394653bf448c5e25e8815c : FAILURE in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8484d1a98c44b7197fac212ea4f9867 : FAILURE in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48e723aee6614c83a0f153edc5a137f4 : SUCCESS in 5m 00s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/394d0b9df4cc41fd879303a8544c857c : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/37138fad3390450b934646e12d8780ff : POST_FAILURE in 6m 09s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/7b6378d1f2f4444c975a5f8496026b56 : SUCCESS in 6m 02s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":1},{"id":"95f04b66548480140394ad0e50bf6d22eb67b88d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-19 17:26:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/57634261d9e540ba9fd8aa0fbcb0eacf : FAILURE in 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"bcdd93a0ff943851bd800f43130db2c7247be83d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-02-22 17:35:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe5d25303c721a74e1a7a556ec73c756ab2aff3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 19:39:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63e8ef6d3e324462a2af42be8d02f141 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a04e04de1b5427a809cfb05418825c7 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29e5cc2bede34e81b45d301c1e03b74a : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/569d7fbf9def4eefa786478587b9a6d7 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da0d2cf9013e424fb9d35911394257f3 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e011d1f39a4aee9c92459526e85cfd : SUCCESS in 4m 56s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0e9dc8a4f3a24409b2b597a8dd762891 : FAILURE in 34m 33s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/82436d715a2248ac8e8cd12c6e63b8e4 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12cd869e01f440a4a167078b3a46bfc4 : SUCCESS in 5m 44s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/0b37df99772545ccae9f0135ef969158 : SUCCESS in 5m 03s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/ddd4f129ab4a47bdbc10d18611458159 : FAILURE in 21m 40s","accounts_in_message":[],"_revision_number":2},{"id":"ccd615b0371dc94950f4aa6223a0f9352969c902","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-22 20:55:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5858529bbb3c4435b5ac56414a8325b6 : SUCCESS in 3h 16m 24s","accounts_in_message":[],"_revision_number":2},{"id":"784a88b9db81ebde99e1558b3b3de685a9f5c62e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-02-24 15:38:22.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3ab696c0d1957efc7d757e40d517373273b4fa89","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-24 19:22:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/75501c151aee413181b677a691caa89d : SUCCESS in 3h 25m 30s","accounts_in_message":[],"_revision_number":3},{"id":"c5fb9dba7db0e543a8f962f11f7a4b88555148f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-02-24 20:29:40.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c2b389bffe4fcabf627aee0d5eb97e467cac2127","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-02-24 23:26:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0ed32a1b8196486fb571dea9f9d64849 : SUCCESS in 2h 55m 43s","accounts_in_message":[],"_revision_number":4},{"id":"eac209b24b5794db7c14071bdf310ecc779c9fea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 06:29:08.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7da4f5b534104ff289445eb0463d22cf : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3987131891a4a9a88d8c9e7476ca8bb : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef53e9890a904b429b2aaa4d8d5e001b : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d6318d669d043b6a15cad073795a674 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/188f5119c8b446c187eb08dab6d3e6cf : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e64850fca6245fab6c34e014dccabf1 : SUCCESS in 4m 38s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9526282a5bdd42c0a1f6804b75b5583f : SUCCESS in 3h 35m 35s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/446c28619db54faebbd5d9f7704ec5ab : SUCCESS in 1h 55m 53s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/07aead9d6ae5441cbcca84a7c041cc8a : SUCCESS in 1h 44m 26s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/eed5f7b3c35349e1a7bebaa07dccc829 : SUCCESS in 1h 12m 24s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9a082ed1ee854b328f8453a1852abfce : SUCCESS in 1h 59m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/932926a22524456c8d32759ca2d4ef32 : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8657a4a66fb4f0c80c9a2d5ec71c19e : SUCCESS in 6m 34s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/916f6d559cd1485e924ef3d812f02ab3 : SUCCESS in 4m 57s\n- tripleo-build-containers-ubi-8 https://zuul.opendev.org/t/openstack/build/ec5866babc8d4899b2fec4272b5b0442 : SUCCESS in 1h 11m 39s","accounts_in_message":[],"_revision_number":4},{"id":"b764096ce80241e7f4d3faad5cb0fd3e21b142d7","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-05 22:16:10.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"52e52cf9daf2082cad1c79730796e42da8ca05ea","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-03-15 13:43:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e1e372289fac0e5dda8b8aa4976868e0dac944f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-03-15 13:44:28.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"e6fb201d25de6ceb59d16b4c9cafdbd1d0babdb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 17:08:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f0f7151770f4032ad5f7cb50009610d : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3945c1aef8348cea8739acb7cb5d21c : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/223398a0eafc410190f20291c32d42ce : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48c4eaa8c0764466ba0b7248f49c175a : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caf2c298149b49ee91f3c9eeda1ecaa6 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42c06e8dac5a4a38b0cd8b8b1ddd5fe8 : SUCCESS in 4m 09s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ad12cfcc497b46019378b0f6f0e2a824 : SUCCESS in 3h 07m 53s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9451092878eb4493a24bf75cda512297 : SUCCESS in 1h 38m 47s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/512eb70ec1dd403a9a2caa8639017d31 : SUCCESS in 1h 00m 30s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/48541dec72644573b6e12595229077a8 : SUCCESS in 1h 14m 11s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/77ba83b8368e4216b73c07f6eb75389d : SUCCESS in 1h 37m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/756297e33e604ad8878e17cc0092ed2d : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6622a0aae18f40d996145898bdb067c2 : SUCCESS in 4m 22s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/310cc1d42a154cbe83d13e45bf146e25 : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":5},{"id":"d9663493dcd6eb8e11b56db5e7abf378ee4691fd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-15 18:36:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/74a48d8edb8240238434bf295b243d84 : FAILURE in 2h 48m 38s","accounts_in_message":[],"_revision_number":5},{"id":"b86d633cfc7a561fa7bf203df3ebe04cb0b968b3","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-08-03 16:30:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b5ff43f4841d8de2243be2a347d5f029ff446064","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-01-11 20:50:03.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d21b2565092f437a513a2add6108a77b463c69b0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-11 23:43:45.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8480997a790b44eebad21b92d536216b : FAILURE in 2h 48m 08s","accounts_in_message":[],"_revision_number":6},{"id":"2d173c9f943b1ae150daab474de73f5b0af6ed56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-12 00:18:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/056259ffb10c463690ce69b30c1aa4c3 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cfc6de8c2d840a6a46c2d134aabb287 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a945a92e1964d009281820dc46b98e4 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce069cf8f2b5493cbdf3b14875cfd87a : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b408ebe0ec044bf1a848b30dc5d94d82 : SUCCESS in 4m 09s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b448fc8cf5034f90822576a045531947 : FAILURE in 46m 43s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/fdd01691f1574f1cb435f329fa32352d : SUCCESS in 3h 15m 47s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2bfd0098d98b45e3bd84c34cf5b89067 : SUCCESS in 1h 34m 34s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/e34c88889eb44d0e9ee73ed01281f698 : SKIPPED (non-voting)\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f1457ca773f3478db959610cae3b6ca1 : SKIPPED (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/21f2b4da0b104acbbeb5d1406e7b444f : SKIPPED (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/78dba3a2b04d4fd09e05083fc513f034 : SUCCESS in 45m 13s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/190f1ef6a8ca470b90ac8252df0ba48c : SKIPPED (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/51873ad29aa94dfda4c4cadb0ee2bbbe : SUCCESS in 1h 27m 51s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/64eca4df1fbc4acf8819c020ab6b1561 : SKIPPED (non-voting)\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/fb5b353d941b47f7af4e474018baa957 : SUCCESS in 2h 57m 44s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/218fa53ed38343239de717ea10fbf7ff : SUCCESS in 1h 23m 27s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/55c16fa156b7444ca98ae5c8fdf9d9b3 : SUCCESS in 1h 19m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9b266670d95d43fdace01c1c196dfac2 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c28d7c23bbce464da0fbdc51c7468e0c : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6161f2aae09744ba8c9c6bcbc42c86ea : SUCCESS in 4m 54s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/a05a24d6b69c4ee19f29a356ee515b7c : SUCCESS in 5m 43s","accounts_in_message":[],"_revision_number":6},{"id":"66f52f3a893fcc64e241f9fcbf1ec74db945fe8a","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-02-23 16:52:06.000000000","message":"Removed reviewer James Slagle.","accounts_in_message":[],"_revision_number":6},{"id":"c9e68a5049a4ed64ea3c9fa50d8ad709f65fcd92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-04-22 21:59:09.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"016b99b445c2b36f73a042dd21f515a94eed3553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 23:10:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/037cb7430ff548978c4d7938e39e85f9 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea64699b451f42b69059b3290d3bde95 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7394e98ceede4aac9e8cacbe7bfdc14f : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e207586ba7234e20838903e333d33645 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe02062f6c0640c7b85295d5cbf448cf : SUCCESS in 3m 45s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b7161915c9ca42589d62d68b6ace7df9 : FAILURE in 42m 41s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/84b04b56ab9a47e98f03fc839f6acac6 : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0662549c32a64f6da766cb84b5ae493a : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/944089f41ee5434398145e3f01a5fbcd : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/e02a27c15ebe465b9c57c39d0b5002e2 : SKIPPED\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/a8446bf631ef4ae3822f03e2e0669a3f : SUCCESS in 1h 10m 30s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6056721106134b57800b41379cd43582 : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/71bbb6394ef74d07bb2be03d8e955675 : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a3dd89d2363a46c3a5ebcdb9d0be80b5 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ddee1bd80f84aad9f0d4affb5cd18c2 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e90abb7c35d423f8be0f2b336c6f089 : SUCCESS in 4m 02s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/b824769286fb472191be10124616b11e : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":7},{"id":"22c92deac36fab027d76efc024c63f444a18f06a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-04-23 00:47:30.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9d28fdc24b7a4332916eaed44e851091 : FAILURE in 2h 47m 22s","accounts_in_message":[],"_revision_number":7},{"id":"baca8bfcb9e3e48cd58f1f98d2d4dfd733f8c521","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-04-25 14:03:58.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3b1d5308ad0fe89a09b8652b19ba0fca0d43c489","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-04-25 15:07:25.000000000","message":"Topic ansible-ee removed","accounts_in_message":[],"_revision_number":8},{"id":"d6a04e2b9a3b7c3facca4958bb9dc589382a00dd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-04-25 15:07:27.000000000","message":"Topic set to tripleo-ansible-ee","accounts_in_message":[],"_revision_number":8},{"id":"171479be7236cc4fb8610df01fb20481bbbd1d41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 15:09:05.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7beb02e178fb43bcb209c1284e07596b : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52629633298e4066b70815fda261213c : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1648f4a9906410ebffef3fd78a3230a : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39ac2526313e46b38265b8e76c394b37 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eec5b8c2bf24096975d1c6486290bd6 : SUCCESS in 3m 58s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/4fc479f7cd0f454a9ec94f8dbed68e30 : FAILURE in 39m 55s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/ad18229e1bac440db072db125b4f7686 : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/110a38f4838046ab87c58c8757005587 : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/9053eff370a54ce1a40e2449c3333219 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/d73655238ce24d47866fa4fd818c3c21 : SKIPPED\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/cb05306eb8f84992bcbfe045ec68043d : SUCCESS in 1h 01m 19s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/de496e7419d442d1b22e769a02dbc96e : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/4caee72939cf4b35858373795d168fbf : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/56af9a22d3dd41edbdba5d9aeeb39dc4 : SUCCESS in 4m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c21e67c0d1364df7a7e9cffebd0f23e4 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69d1e02ff59a40f0b2313f22234bbb90 : SUCCESS in 3m 57s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/81c261302e3a44cf914e89aaef50d089 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":8},{"id":"3d24371288e0951e785768ffda4a6a903ce3222a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-04-25 15:46:46.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"352a9fc8947389d44427b72cec919086bc0a4e1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-25 16:55:45.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a02c4081c509405893e35f49dffd92ed : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0745932236c4693be737b3d5c454144 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ca5ac384f0d408293ac9d15e4bddd3c : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e6379ffebe9470ca41ee6e5b5084a1f : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddaccfc2879245f0baed82ad59c47ce7 : SUCCESS in 4m 02s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/6315eafe5edf4cf6be688fc8ac24a143 : FAILURE in 43m 05s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/5a490359169d43e5a31735960bbf1bd3 : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1a5ccd6a886b419481c96f9ffe5ac1f9 : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/4f3aa2aff60d4e5283709cae61968f75 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/adc0e2a3dfe843ee85a7ef8571f7d469 : SKIPPED\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3c85007b71d149a38ebfe6bc78cc204d : SUCCESS in 1h 02m 35s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3eeb9ff9a2f44c4096121919ffb5ccdf : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/a0ee8283ae20426b86af00f9489fef40 : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4d914ad6abce43f29ade0060d4b9e12c : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/20145611d48a49ac884730261996eaf5 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73ad353c3e9d40d490e76833745acf83 : SUCCESS in 4m 02s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/7f3d60182681443b9c747ec80c16a45c : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":9},{"id":"95ec583112d9216d2388005b0d4c271357be6038","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-04-25 18:04:33.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/cab5bad014cc414dac3e4eefa40268ff : FAILURE in 2h 16m 27s","accounts_in_message":[],"_revision_number":9},{"id":"cacb97e776743f4d927bf48bab51ebe58adeb4e6","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-04-28 15:46:06.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a34d97f4a24a1f6b2c38a46607c98f2d9172d78b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-04-28 15:57:40.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"1f064b7db1b1371d4b1d0c9e4d98a71d4299134f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-28 16:55:32.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfe2c06b34fd4d6999ef63af854d1373 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bda359f281042e494e04e5289486871 : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00d46df4ab6a4130b4fa437c7a495de7 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee69b7214c904952bfca18b439ad463b : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb0a24de7db49a4ae21f80839a0d2ed : SUCCESS in 4m 45s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/6177dc0f392945d9acd2b5e660b5fbde : FAILURE in 35m 26s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/c23c14b313034234bdb8d12b32beeee3 : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/0dc73e1d17184d61bb5aca696db1e7a3 : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/4414cb9bc10741c9bc69dc80cff11904 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/7d8c16e02ab14900a0117331ce41e080 : SKIPPED\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/a3791abf69154199902e6f06b9fc7d31 : SUCCESS in 56m 01s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/39d9365c65394f7c89d2f7e6c282a14d : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/2d64408486b44903ba4ce68e00632c2b : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fd62495f40f24a38acbb1e2d697e521a : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4604d8c1bff5403b806e5cc83e73a3ca : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76f218bd82894b71adc495287885d6d7 : SUCCESS in 4m 10s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/418f3dcc01b4473abd3c85d120261bc0 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":10},{"id":"5decb9ef1c51a71b680db8e877354b16a19c271f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-04-28 19:50:08.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a38a57599b5c434aab8d1c9c2700fbc1 : FAILURE in 3h 51m 03s","accounts_in_message":[],"_revision_number":10},{"id":"c48f36cd8b2197fb9bb465e22ce3eb8b9594b537","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-05-02 21:29:04.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"7fc79cdb53161971c81d9d446c200e241b1f311e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 22:33:37.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19becb6c13b54e08aff3c0312d1428dc : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c58a7aca1eaa411ea98a77b5b02358e7 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/044c87022dc4419d9f4b00384c4d5fb7 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5545b2c91954416ab2f36dc07285147 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c14ee5ded3d4b6da7fb6205e9ebee8c : SUCCESS in 3m 54s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/e07c4d2db818493fb9dc385c3789e625 : SKIPPED\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/3aa3910e1e884e2c9b38ea0ed77a2a2f : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4330e9aaf74a4516bbff44785771e0b9 : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/23c14407c66b46c39a77b416a2b7d539 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/9a884a4ef4e9411e8e7a1d7a12b2eb70 : SKIPPED\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/6152b61900d54bba9d0ef19af240d9c7 : SUCCESS in 1h 03m 08s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/fe62693f04174efe8fccd4d819da13ac : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/253dbb43324a4e2daa773fbc5f814bb0 : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e23d94c1f984424bb715f6f53ad64e90 : FAILURE in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3b9fea20da7f4879a6ee9acf0a304a8c : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e6e27bd3dd746068f5bccd4f88fdc62 : SUCCESS in 3m 22s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/790dde0e34714ed8a14e353a94a352e1 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":11},{"id":"12614d126a2227057ec7229485a60e215ee99300","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-05-03 00:17:50.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/de4d47aefdbe424386ecea5461f36a11 : FAILURE in 2h 46m 58s","accounts_in_message":[],"_revision_number":11},{"id":"791e28437329619ee420cef8754c046e80bc3835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-05-03 13:59:10.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"026e62f45fe5f4ca42d4ab35a1accfc22a9a1870","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-05-03 15:43:02.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/674626d505b046fca42f581ba47eecfd : FAILURE in 1h 42m 33s","accounts_in_message":[],"_revision_number":12},{"id":"78c352c81ec060431c340bc6d8b9e189dad05ff1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-03 17:06:50.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc0c02e4dd4c4f55b1b95eb296e21bb8 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ed9892913684706bc0911da4cb75538 : SUCCESS in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74c006f0321b475d86d6104ec20be88c : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3239f0879bbf4f47ac7657387df72d16 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/184e043c74af4b09b30c31a314bb685d : SUCCESS in 3m 33s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/8d047f445a34401bb7a9e28fce4dcba3 : SUCCESS in 2h 57m 18s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/ab09e22a837f4a39a125e610691b1b82 : SUCCESS in 1h 36m 09s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d6f12fa1a7434a779583b31864812d67 : SUCCESS in 1h 43m 09s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/dc01e810cefb42fb8fd891ac2159cc64 : SUCCESS in 47m 13s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/03aaa43532fe473e91bd5c51752a7d2b : SUCCESS in 1h 27m 12s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/8549fb9de31548eaa4c48714a4ffcb9f : SUCCESS in 2h 26m 57s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0a2a633102114e769cfa31bb1485d425 : SUCCESS in 1h 16m 59s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/27244a7aae26422f98905c9085dc6ca8 : SUCCESS in 1h 10m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0bb0fd4b6943487f9ec2d370fa4db675 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17b7db723920406f974059334f772b74 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd6576196ea34e869af1a705f0221ec8 : SUCCESS in 3m 38s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/d2174979845b45f3863d34865ba01d8c : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":12},{"id":"e454e4ca5548cefede71f6ece898c14ba8d455a3","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-05-24 12:43:16.000000000","message":"Patch Set 12: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"c3dc207cb8fcff2722a36bf591e5252bc8d772d3","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2022-05-24 17:42:43.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d8a3991613e29686e9ecfef74700693b2dd8eefb","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-05-30 14:40:36.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"7574223030a20b2b3a0cb3524173ebb4d7005ff2","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2022-07-18 15:14:42.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"34d7aaf847ae9890786b2849e84bd18b8b83e286","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-18 15:23:42.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"191fffa310e7af3fa608e93d419afe27a3e431e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-19 09:12:53.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"a53ec3579ba02209f9a9b9411e845514305a8462","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-19 09:15:12.000000000","message":"Patch Set 12: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"0f5be537e2d4d658075ec9e63d7e14191708f2d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-19 10:02:21.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c0755cbb58f48feb3521f5d04276d9e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8621d0addd884ed6ae20d7f4418c6955 : FAILURE in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a03e16e59264d92ba5bcad8f74dbb57 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fec1626a7aa48b6943b97be742cb93a : FAILURE in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b433dbfcf18c4aeda81c98ec89775156 : FAILURE in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/978f1535739c45b98506e7dd851ce527 : FAILURE in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c88e4df6424845f589cab2129e14d90f : SUCCESS in 4m 46s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/448ed43135ea4181a8484d1e32fd6aaa : SKIPPED\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/7490686f813c4a7eabc410121e40d6c7 : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ce072ea7c8e5481d8b6b9b91d4b26234 : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/aaf3fdf05e4a439496fa989f54f70295 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/d4ea4fd34e294f1d9b079157eec275ec : SKIPPED\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/bd6b6a0f70734b2ab6dd70f18ef3ed22 : SKIPPED (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/8f47f1741267416996832e72bf64c37d : SUCCESS in 47m 40s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d1e31258a83042069f16f8cb5e82e6a3 : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/5ecbb66da606451b9f70a94b6422c996 : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a4a4283e4ff34b4ca3613117b1b63aa3 : SUCCESS in 4m 27s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/98410bf336d2471695d77debe859c079 : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":13},{"id":"6fef390427bc44a49e571b1fecc8f9ed068c9b98","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-19 11:26:15.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/587b4d3efb47446685058b52d333d674 : FAILURE in 26m 15s","accounts_in_message":[],"_revision_number":13},{"id":"919092da878ccef8a078cbdf4f4cfa24caafd8b4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-20 12:07:37.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"4f5a9aea086bad5e23cdf9a03fab0da3323ccc2f","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2022-07-21 10:07:04.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"515eabea4dada5edb74f7c91bbcc5a636776cd23","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-22 06:44:54.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ba86548a9f8ac0e3c2e508f5583d6ef79f5e6b01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-22 07:34:09.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"c33df583c3b37d357099ac59358a030328ead372","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-22 08:11:04.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7bc5fab311704b4e86f407baec517f7a : FAILURE in 28m 14s","accounts_in_message":[],"_revision_number":14},{"id":"837959168d2a70620f8b0d271ab123c3d58d6821","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-22 08:30:59.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"fa3ebc959119eccff069d6b25c73b3cabe1cfd8c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-22 08:59:51.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/834f948977fe426891c58248765068bb : FAILURE in 27m 40s","accounts_in_message":[],"_revision_number":15},{"id":"51ca25acdc834b031ea82dd1693f1a237cbb00ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 09:22:33.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/631be46ce32741658560cb4339277d4f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0ca2b9fc2374a95a3ebfa1e33318de6 : FAILURE in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b787523d4aa4eadac983a4b6ad5fafd : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1646007372d5430591f7cdd0a123faba : FAILURE in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/672812e74394465690a364faafac7fd9 : FAILURE in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/00b2eeabbd1948aaba55313d324c15fd : FAILURE in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e8fd0d87b24a1f83093b9b3eb4ad8a : SUCCESS in 4m 43s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/8604ea0185e04e26a68c6ee2845425e5 : SKIPPED\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/0610900c43394c109ddcdf36ed7abcf8 : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e3aac9906ac34f58a06bbe1247bcf2c4 : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/687b0034af254e0db3fde604803dfde7 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/6ff1e209414f43568c0c03caa6910cf3 : SKIPPED\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/b460d0bd7d6441cf88214885e01ccb69 : SKIPPED (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/587f2ae263e345438d91dbc4c5a589be : SUCCESS in 49m 38s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/82e5f5961e074641a4bc925de24763a6 : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/75920610ed504e2295f201fa41b44aec : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/de87f4fb2aa84e00a175651ce7514666 : SUCCESS in 4m 59s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/99afc841c92e4eeea8b6e0cf1ba5bd9e : SUCCESS in 4m 44s","accounts_in_message":[],"_revision_number":15},{"id":"7ceed0f61114b4e9ddb0a411f1068297059494cd","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2022-07-25 09:15:00.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"e6ffb3af03b8e2619cf763b474fe1dc680a1ba47","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2022-07-25 10:22:14.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"7e508bc44ca770c2c108f4e5b8ce3c36c1167948","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-25 10:35:09.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"0118c9620829d0ed26055d6a8d45a82895f66921","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2022-07-25 14:50:27.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"284eba09dbc2811da36923a57761fce23f8f7703","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-26 07:35:34.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"1bd30ed465b9b1601f21035dbe186484a265cf45","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-26 08:01:44.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0e8d860ea1574a7c92b66cf670d1d6ef : FAILURE in 25m 10s","accounts_in_message":[],"_revision_number":16},{"id":"5dd5de2d139c768dd035e821bbc672f14147f5af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 08:32:06.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c39d7506571e4a4681cdd1d5e5f0a866\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95c047a821444415aa436bfc6ddacd0b : FAILURE in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/287e1e49958d44669c4abf69726dcc78 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efb32ff6bf8d434eabf796a487dfe1c7 : FAILURE in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8b3a73c035845ed98b1ea379eceb537 : FAILURE in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab894256d4fa432487825d6fc7c44cb1 : FAILURE in 3m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f0566b0ff4499db8e49fc2ecd69257 : SUCCESS in 3m 08s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/6d20478742ac4c3fa6d8f4b6d61ca20a : SUCCESS in 54m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9944971e4dc5468484bbc513da5440f3 : FAILURE in 4m 06s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/985df3451ccf458bb087d427e892a48c : SUCCESS in 7m 35s\nSkipped 8 jobs","accounts_in_message":[],"_revision_number":16},{"id":"6291003a333dd2530bac296a6c74e0f4e708ac62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-26 09:40:33.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"75645a8261298fef113388063562aa7fb2700491","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-26 10:02:01.000000000","message":"Patch Set 18: Patch Set 17 was rebased","accounts_in_message":[],"_revision_number":18},{"id":"da1a3dc2dd911f65306df8a7a9a48c39d61b643b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-26 10:33:30.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/375d1b46c1d047b6a6a802616d1f2d83 : POST_FAILURE in 24m 16s","accounts_in_message":[],"_revision_number":18},{"id":"7c7493032c870063e29ae1cabc1778befa2eb0ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-26 10:35:27.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"c07b8a0863097a56926c32f497a3cff1a25d91f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-26 10:59:45.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"ebcf2fb8f76b2b912af1f58888c9e930f483eb33","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-26 11:27:44.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b8b086720fe84607a4e1d779b9bfa3f1 : FAILURE in 26m 59s","accounts_in_message":[],"_revision_number":20},{"id":"43e28a720362b30c99636c2be8d19bfeaa7640a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 11:42:53.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/763f5f7743c44ea6a2f85291e01e5156\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/604608fdf9b14e50bfea7c08dcf1a586 : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3191d40a4c43aa81779cd1771ad9e4 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9252b0edb47f4e039e2e886eec22258d : FAILURE in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbdb90198f0f48b497ad9cc33916bd1a : FAILURE in 4m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3311bda503c433da6cbbfeea99efd48 : FAILURE in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4a342b49b48498eb582c98906edd727 : SUCCESS in 4m 05s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3530954a8e304dc0a9b776ac55384b7b : SUCCESS in 41m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2f3fd63f44dc4efc8c31134b700e197f : SUCCESS in 4m 04s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/c1669e7365c64e5593b2ea68dcd99a73 : SUCCESS in 4m 36s\nSkipped 8 jobs","accounts_in_message":[],"_revision_number":20},{"id":"6b77950e3945b96bc01078d6b72542d0bad6c4db","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2022-07-28 13:05:58.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"1a7c6c6c88aec5bc7cc846970875f6581275eb44","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2022-07-28 13:08:27.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"af0ee15a9dfd2b9e5c76cdb41aa0b6812d122c2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-29 04:00:26.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"7b50c47c404f6a154af87d2f4c6658545412cd3f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-29 04:42:16.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c2642c02f6d04f498e4fc622c3759194 : FAILURE in 34m 30s","accounts_in_message":[],"_revision_number":21},{"id":"c51357d580fd4f78a0c71299fe2e9cbb81a119ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-29 04:42:59.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"25dd44f97f61cecbeeeea82062b1d7c36f3186e2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-29 05:10:23.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/73be2484e75c46c0a91bff640801f939 : FAILURE in 26m 23s","accounts_in_message":[],"_revision_number":22},{"id":"64524a75542822261a5b617d59c830535e4d4ba2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-29 05:12:03.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"d27cf8d50dcce53718c07006fd31d89598219013","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-29 05:40:09.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f29bfdd2cad94729a47eb3377e06c3d0 : FAILURE in 26m 51s","accounts_in_message":[],"_revision_number":23},{"id":"91ba6a928d7926e30f582ef2437cb0df1f478428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 05:55:30.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08897dad28354ae996c0999f19300ff2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/948ac58daefc412cac73a5fac5ec9c42 : FAILURE in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf8a6f31987a4aa68b0e603024da1bdf : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d63df3fb8b5a4ac9badeeb605c7dd656 : FAILURE in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf0340f0128142d0a03b4c0c2b6db1d1 : FAILURE in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6aa85123099c44749bab881845c5f56d : FAILURE in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae34059c64a4a489a6f591e9df043b2 : SUCCESS in 4m 36s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/40c5b9b7f3a04104a3d16233b582713d : SUCCESS in 42m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c20e2ade051c45d8b5d74c05e00fa25e : SUCCESS in 4m 14s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/c33a80f4c7934ae588caaa458f7a85e8 : SUCCESS in 4m 51s\nSkipped 8 jobs","accounts_in_message":[],"_revision_number":23},{"id":"7d6ea1e4b6a2b1daa3b4f550c1ce2af88a8b88b9","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2022-07-29 08:46:19.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"fa899ce7f8ed7d5c91b5fbb0558669b2f6f45f6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-29 11:39:59.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"061ce266f44b4109cedada839b86a2753d95a8bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 12:38:27.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f5d0a35016a42ac8d33eb497b867110\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab04f725e120468faa12256f763d35e1 : FAILURE in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/975ac2998a5845138f0d2a21d9781aaa : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c38b7dd8229849e0b49619f0477ed715 : FAILURE in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68f770c9e05f4f3aaf624da53034f588 : FAILURE in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd3f04fe12c34f3ca8fd34d58e9c8a71 : FAILURE in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b48ef7e7c0a54ee4b6b32e8f206902f1 : SUCCESS in 7m 10s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/4937e2e61b284352b2769ab036f49902 : SUCCESS in 56m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2de72a47170845ee99ee8b36951306dc : SUCCESS in 4m 35s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/6566c05401474f83ad87b911d0977134 : SUCCESS in 5m 28s\nSkipped 8 jobs","accounts_in_message":[],"_revision_number":24},{"id":"5b112b20f475c6c901bf7f57032e6c00d0fa6346","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-29 12:38:45.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b9b4e40ada464bf9b842e7b20ce27159 : FAILURE in 29m 09s","accounts_in_message":[],"_revision_number":24},{"id":"33e98ed824d443da66a27f5a9b842951b07986e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-07-29 12:51:01.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"01205158e9424edaa2251b7178148ed74c36380f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-29 13:38:45.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/15ce04e314b34fab840687e0cdfd1c3c : FAILURE in 31m 44s","accounts_in_message":[],"_revision_number":25},{"id":"a0409bb332f7beb22121caaee977612b4c695d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 13:48:27.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc53d81a16cc484b904ed80ef499f295\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5faaeda19cb45c481672b0dc997889f : FAILURE in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/845fcc58ea634e64a2caaf9c388eac21 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b9879c564544fafa660d862ec5d8800 : FAILURE in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21b5d9d6d0654dc283bb07e21c63e86f : FAILURE in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62bc682882194beaa94cce56c1e9c423 : FAILURE in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b5f7dd71d6a41dcb7d7e06c08285276 : SUCCESS in 7m 32s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/acf4c45784e14904beef0ab2a83dcacc : SUCCESS in 55m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3d348aa5006842d39e65310597e37e2d : SUCCESS in 6m 18s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/606321235b7a4ba882392899ffd493bd : SUCCESS in 7m 31s\nSkipped 8 jobs","accounts_in_message":[],"_revision_number":25},{"id":"cd5af7ced77dc179a696133f134d05fe2815a7f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-01 06:02:14.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"986a4b98e328f171dffcdaadc1e63517c047f1ea","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-01 06:29:31.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3397cd54a7464069842805c7fc90e6c6 : FAILURE in 25m 24s","accounts_in_message":[],"_revision_number":26},{"id":"f6e57dc671447df36b3dec2ea37b97a49572c20e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 07:14:23.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d521ba74f4345e2b0f0d16abcaa5c9f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3d5c5dc0f444c7893c3dc27df21549e : FAILURE in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8817f0b714f944bd98fd33650c69caec : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ec3fc851ebc4e9a8def1334f1d525bd : FAILURE in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db178036224d489894e608cffe00a4f6 : FAILURE in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b135803a6d2140a388db23eae9abfd63 : FAILURE in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/085cfe91d9d54e9aa8869f67ac5ae1fa : SUCCESS in 3m 58s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/47e965ca8d9e4375a7edece8ff2f8d5c : SUCCESS in 1h 07m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/49051c2ddf754ed385657e5f952a1af0 : SUCCESS in 4m 22s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/2cb53f9f8c5747c88aeb2a89f464a8e8 : SUCCESS in 4m 49s\nSkipped 8 jobs","accounts_in_message":[],"_revision_number":26},{"id":"7e46ea4bfd5c47f7fb30edd92694c57931b56542","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-01 07:24:02.000000000","message":"Uploaded patch set 27.","accounts_in_message":[],"_revision_number":27},{"id":"a0174210b86296da02f0e933f828a66a885c5bdb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-01 07:49:35.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1bcf490d2a784e6782b7492bf8225e6c : FAILURE in 25m 00s","accounts_in_message":[],"_revision_number":27},{"id":"89ab0bb5690e253df89f084a0c790edda6bb713a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-01 08:14:09.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"c53c4582c01dfa6b981a83300fcbb4d3b7969d24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-01 09:33:17.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd850c154f034ee2a233caf55d42013d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2623ce4e85884f3c85b5e3e51f178232 : FAILURE in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c83c8d1f1314e8fb14408de6fa4c26e : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d69a6472fff4beab05a6556d09a6c8a : FAILURE in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a1fb2f99ea94e0bb0a40353f553566d : FAILURE in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/205fd424725d4a3b93664b6edf23eae5 : FAILURE in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/720859cae62c414b8643f55376ab2224 : SUCCESS in 4m 17s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/0c825bc71f2c4d639808b9a6f85e6572 : SUCCESS in 1h 11m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dc0b53c00ac0439e8a475c940433d037 : SUCCESS in 4m 11s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/bcc253ca7abd4e1788de175399709dcf : SUCCESS in 5m 03s\nSkipped 8 jobs","accounts_in_message":[],"_revision_number":28},{"id":"c597151c57af42cd8ff8bcd90e1b5bc225a2fdb5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-01 10:06:16.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d8748bc97db54ce1be8ee3a0d39b1799 : FAILURE in 30m 26s","accounts_in_message":[],"_revision_number":28},{"id":"c00634ff44596d368d7f5bddf64087eba03ebd0d","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-08-03 12:45:32.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"f01a4122bd6ca60127450a3ce73165fab1f2f7bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-05 08:00:56.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"4967a54ef5a4421b0985f996dbb2539013681adc","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-05 08:06:38.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"40458992060b82eb66eba0511f83ef5bdb958f6b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-05 08:44:05.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/107a1f71614f4812933f8ddb71b23458 : FAILURE in 31m 48s","accounts_in_message":[],"_revision_number":29},{"id":"6722b5fe0f84116ce6f48dad0d2e3409556a2874","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 08:46:39.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/198f38f860bf44efa035d8fe3d4a2609\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1edbf53ead744598ae700f7d1df36ce1 : FAILURE in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00d49b75e2b244fb9eb34b45f8ad3059 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d313c4c6c55040c4a52d82404f124b51 : FAILURE in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f2d5c8e002d4871a95c5432739cfa03 : FAILURE in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd43d2ff4e8f435a9682e55554370509 : FAILURE in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00fa0cf6929f43fa957c48883164351f : SUCCESS in 3m 17s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3ecae1285b4e4bed9473408fb5b75385 : SUCCESS in 44m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/db3aef599615421b88446eda234e5983 : SUCCESS in 4m 08s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/974bbdd885dd48daa7b5005e62f1c955 : SUCCESS in 4m 52s\nSkipped 8 jobs","accounts_in_message":[],"_revision_number":29},{"id":"d1c418b1a50551b06e8f056245ff218ad13c6e49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-05 09:19:31.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"005b6b352f7faa925da9e5a90d00c129a69816d1","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2022-08-05 09:19:34.000000000","message":"Patch Set 29:\n\n(2 comments)","accounts_in_message":[],"_revision_number":29},{"id":"2dba1fbbd4c485c77a8b730cc496c12b3f3680a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-05 09:22:29.000000000","message":"Uploaded patch set 31.","accounts_in_message":[],"_revision_number":31},{"id":"2affd2e34494b8177f3dd185f8198203680b4e0f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-05 11:46:23.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d0a3fcb9300b4576bd9ee52613e73369 : SUCCESS in 2h 17m 15s","accounts_in_message":[],"_revision_number":31},{"id":"bf570c1a245a017272fd0e6e0c6f22d1ade3378d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-05 11:56:16.000000000","message":"Uploaded patch set 32.","accounts_in_message":[],"_revision_number":32},{"id":"92c0b8f5532c4af10474701c0a07168933bb9957","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-05 14:31:33.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9c48d5af8965462f848005ebbb23156b : SUCCESS in 2h 34m 12s","accounts_in_message":[],"_revision_number":32},{"id":"db795f28ba0be21105bc2d625cc38bd637b108c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-05 15:12:15.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef4331e716a44ff28f0e5ef63c15f7c7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b08bb0bd62e14e0b8e5e26e7405171ac : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/315eef814afb47bea67fa44e4eeffe9b : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d8d506eea864e089a763840b875e9b4 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cf72214843a4a6291bf38385a673c99 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adef5a1f6e474515b5ed796d6560b261 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/714a000ee53342fea5fd9a547c4614cd : SUCCESS in 6m 31s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/eddbadac479047f2bf800b6cd1bf18a5 : SUCCESS in 3h 09m 20s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/9b6bf574f4e64fff915e1e4a452a29a4 : SUCCESS in 1h 54m 43s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8491c7c9381948dfa925cac9793d4935 : SUCCESS in 1h 39m 06s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/1586901d7d46416aa8568d9c237a6448 : SUCCESS in 55m 10s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/8b42d1b7616e49f2b424ec9dd5baf035 : SUCCESS in 1h 29m 25s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/919ea2b179484fd8a5358d9e35d8acc6 : SUCCESS in 1h 55m 30s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/c8bf9cd4f4754a23ad0b90d918eda656 : SUCCESS in 2h 40m 50s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ee4456b1c7e14588ab52fc82d171f2d0 : SUCCESS in 1h 48m 09s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/09a0c2e3bbbd40edbb0b10a5d76aff8b : SUCCESS in 1h 19m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/55aa9716d9984222a3cce518a86c518a : SUCCESS in 4m 32s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/33930122d2f24dba98db7cb57e6ddbfd : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":32},{"id":"d01f89c8de78b90aed2e372aff5e8c65ef813dea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-08 05:07:45.000000000","message":"Uploaded patch set 33.","accounts_in_message":[],"_revision_number":33},{"id":"01663d933a667099c7ed454bb2e46616150823c2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-08 07:23:15.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/adbbbed7ce6c470892ba4ac10b5783e8 : SUCCESS in 2h 05m 26s","accounts_in_message":[],"_revision_number":33},{"id":"c21840f0201b02e576ba98b1f2d7e600c5acb2d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 07:58:49.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25e3a17659f441589e1f961f9a9d13ec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e28fd179a474390a91a10f43bcb01d3 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9e60dbdc5714ddab323e4d25fb9bc0b : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/751064f9303647b6a19800027eea1ffd : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84885a8e0d484ead8ad061cd4a1cefb7 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74cbf934150a43b7925197d0e410b678 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b41ca9db64184734987abdf92284fd1b : SUCCESS in 4m 23s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/968e36d7db3342c3aee2cbe1f10c21a2 : SUCCESS in 2h 44m 25s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/25ba5d15edd64032bb5f5b294095fa99 : SUCCESS in 1h 39m 39s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ab604dfe66a34c9d8c7cb477fc5840c9 : SUCCESS in 1h 43m 10s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/304b99b2d21c4c238ef8173e6ca61a62 : SUCCESS in 41m 48s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/201d0571c32e49f2a837139828680483 : SUCCESS in 1h 19m 16s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/112400aee2a2418382830ebf015fe77c : SUCCESS in 1h 35m 54s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/6dd141b67f434dcc8609d243101d7dd1 : SUCCESS in 2h 08m 37s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/106c9afa1cff4478a49f6030718da105 : SUCCESS in 1h 26m 16s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/a8f8be33c611406ea3e436f42a615eb7 : SUCCESS in 1h 09m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3b3bdb7234834e0f9ca1bfb689abc019 : SUCCESS in 4m 19s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/efae17c5830c475590542ef8afb62ee9 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":33},{"id":"23edf45eab9a1a55a7f9d9656a40994ffff4b864","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-09 09:38:32.000000000","message":"Uploaded patch set 34.","accounts_in_message":[],"_revision_number":34},{"id":"0e49aecdc66b947c5773ae63f763547b9c6f3534","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-09 12:12:07.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/836afa76651c4ddeb489bbe63554fc3a : SUCCESS in 2h 31m 03s","accounts_in_message":[],"_revision_number":34},{"id":"8f64b00bcf39c5ee55e31cb7e4925a80c819e40b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 12:48:44.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62e62a6a66704deaadb01441194b01b8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb580a47cb1b4e0f8bedd9ea1a2a42ca : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2bbee779444962a2516f7624ddfc6a : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a98a2d9473ca48bfb1bc571ff686e36b : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7ef987c411643d7b351606484f589f2 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b573c942945f4b64b331372186a824b5 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05a2b039e6f64637bc7e6aaeff00c63a : SUCCESS in 4m 01s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/871c4a43044b449abd8b27dcefc5200a : SUCCESS in 2h 45m 07s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/ccf4c8067bfe4dd788c7fe122c843242 : SUCCESS in 1h 29m 39s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f45a82c231954b7eae71491a47a0de76 : SUCCESS in 1h 34m 21s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/d7ce99c42f264cccb7a1e3cb13a6c44a : SUCCESS in 37m 23s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/782394776898493ab6db2c4f1bf00615 : SUCCESS in 1h 07m 31s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/53a308b62d354b36ab3b637f48e28271 : SUCCESS in 1h 50m 22s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/be5b5e34fc444f2793f1d6b402dc0002 : SUCCESS in 2h 40m 10s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/8bb687704677455ea78c43903a1a05da : SUCCESS in 1h 35m 21s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/16598ef44dba4f4eb567faf33c477052 : SUCCESS in 1h 29m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0ca2fc3a3de14400b52d5aa76155c193 : SUCCESS in 4m 08s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/ff516034362846d6823315140a417aac : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":34},{"id":"836dd43ebbba8a4274a489248503979f114183e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-10 08:37:27.000000000","message":"Uploaded patch set 35.","accounts_in_message":[],"_revision_number":35},{"id":"500717c75ed2dcc3003e25988b0b68becbef7bc9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-10 09:21:25.000000000","message":"Patch Set 35:\n\n(4 comments)","accounts_in_message":[],"_revision_number":35},{"id":"20805170701aa645d722189fe99a225efd273b54","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-10 09:33:27.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/32dcdca5ebe5410ab1c72ee44c25c37f : FAILURE in 53m 24s","accounts_in_message":[],"_revision_number":35},{"id":"d84e61876679c85024eb69bb0102da2ec17705b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-10 09:40:40.000000000","message":"Patch Set 36: Commit message was updated.","accounts_in_message":[],"_revision_number":36},{"id":"61e0188ac4911d99a481d09681f3b07c437ffd9a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-10 10:41:26.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/524d3a5492d84bd2abaa43321c1680ae : FAILURE in 58m 50s","accounts_in_message":[],"_revision_number":36},{"id":"0ed2b93ecd2f72e3cd1071fa659d9d4d60c7e3ec","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2022-08-10 12:33:27.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"bab0329f12bf881681bb1e8de3703e26d8a37f46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 13:06:27.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/690f1f6cb0ad4e659b85c55da6cdb119\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf2c8a3c69a840c59a93c2153b9e257e : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b55323b268a41d091100e7af275c23b : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12f2a6b9528040eba797873bbdf1a91d : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/708f40a090d64d8aa5663d63c20d29ce : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8ee90456c834bb298ba3b0a47f1155c : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60e3851efe50490294430ecb2ac6bfcc : SUCCESS in 4m 39s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/1ad1e2ab68b64f39b8ddd521204f166b : SUCCESS in 3h 13m 21s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/2cccc27d79be4857be66e5bf30d675be : SUCCESS in 1h 51m 07s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c4c648d757f742fca9b2d2093020586e : SUCCESS in 1h 46m 40s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/90f8b052ae474921b9bf934b42245c53 : SUCCESS in 50m 34s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/3d27cdf90c80460aa20c0709e4526554 : SUCCESS in 1h 34m 51s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/7f476d0b2f744cbba77dd8b34c3442a1 : FAILURE in 1h 18m 38s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ad05c2a0e3b945e1891f8d670ddacf15 : SUCCESS in 2h 22m 16s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e8493d500a3845afaabebdc37854f1a5 : SUCCESS in 1h 44m 18s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/27314ffa8bfe47a6aecef308116bbf49 : FAILURE in 53m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3174a27351e44a478ba7373ce399af01 : SUCCESS in 4m 32s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/d7a5134427cd4e55bd8732503d1f6917 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":36},{"id":"1b6d2a8dfe6ee52f0a554fe928e2435ea4b9d214","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-12 09:34:56.000000000","message":"Patch Set 37: Patch Set 36 was rebased","accounts_in_message":[],"_revision_number":37},{"id":"1b9a5c7e394346a6da150048234e1f9b66a7acc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-12 09:40:58.000000000","message":"Uploaded patch set 38.","accounts_in_message":[],"_revision_number":38},{"id":"cc26041be1c367b438411a8857bfeca3a4ff17dd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-12 11:38:54.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d2cb9783b33f4fa687f15d11ded69716 : FAILURE in 1h 56m 24s","accounts_in_message":[],"_revision_number":38},{"id":"40eeaae647ecb806b516e468fb7bb929c9bd7aed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 13:03:10.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b58235379a274d48892f02db8be977fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cfdb688a9da4ba19dec88c9ec4cdc43 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2add159bb9634274a8c11040ab4afc7f : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03718648699e4d8d881276b7e4c67bb3 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c64150ab29a44fe8a7b00b24a424125c : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2dfa2de2d5974283bfef261daa0c5527 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b3ecef45dd4877995f4f5cd69344a5 : SUCCESS in 4m 13s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7b11a084046d410f8f1f424f8d9df7dd : SUCCESS in 3h 08m 39s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/2bbb7305d8d442e786a85454a9ad3547 : SUCCESS in 1h 49m 25s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a01bad6b5a1d48479a1a80ff81498563 : RETRY_LIMIT in 14s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/d4014ccee722487ab2619a0b9335fd78 : SUCCESS in 55m 31s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/adbffa97ae9a401f8b7a1663f75deddf : RETRY_LIMIT in 13s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/dfa32d4de2ae4eb2ae73250288015a32 : FAILURE in 1h 15m 00s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/226a78100a834522a58f135d55b0ed91 : SUCCESS in 1h 29m 20s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a9bbe43bd60940259a669a267820dacf : RETRY_LIMIT in 13s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/ec1f2546e756429b88f085bfbdf51b6a : RETRY_LIMIT in 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3da7636a26424c07aa6594ab10484cce : SUCCESS in 5m 08s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/e05ad5cbf99748978451ae370725bdb7 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":38},{"id":"dfafe5c463fa700eb9dec64a6b69ed112d89a188","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-15 12:23:10.000000000","message":"Patch Set 38:\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"7cdcfeaeb9993e3a577b827800f54c160330b894","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-15 13:17:54.000000000","message":"Patch Set 38:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a03bbd4b9d79489cb147ee95c36a34ec : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":38},{"id":"cb6ffd0481b8bf033c0073e30dec20c960499dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-15 18:16:17.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62c26315c330421989473b40f7ef8192\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aebc470844d642489d0235e9bf3b42a8 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/616da0807275400a94963d15f523ab40 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fb417503cac41e7bb2215c121822e8b : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acc32b2d336b415497fb7d627f229ba0 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b62a0af9e76461a87a00c4e5a24dee4 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2e10f6f966f4156a6db28e77bcd0865 : SUCCESS in 4m 34s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/eb5c29b5fe5942abb706d48331449163 : SUCCESS in 5h 41m 01s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/2334d53c6e8240688efb3d9aa68f99b3 : SUCCESS in 1h 36m 02s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6a2a7c5c9a41407eb525a512597d29f0 : SUCCESS in 1h 30m 10s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/cb0135e2585c4de3b10b7405f88884aa : SUCCESS in 42m 24s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/d94fa14af26242718e89c843a28453ef : SUCCESS in 1h 17m 49s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/864c2c283b204b2385a44ec7ac37575e : FAILURE in 59m 57s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/390d6514f8794829be759296aef5c51b : SUCCESS in 2h 27m 15s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2b697aa96aed49a2bb5f9c1306392ce5 : SUCCESS in 1h 27m 56s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/be48d58fd61046819acf649a0ccebd01 : SUCCESS in 1h 19m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/522a95ebe2a949dba64698c311b5be38 : SUCCESS in 4m 22s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/b20217ba048e49b9a423d72d8f63555e : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":38},{"id":"f190a37f9029cfbf692c3d874c77fba48e4f5a85","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-26 08:43:03.000000000","message":"Patch Set 38: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"526d60e512768c760de634a43770aa7e234a767d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-26 08:49:30.000000000","message":"Uploaded patch set 39.","accounts_in_message":[],"_revision_number":39},{"id":"645d4a4b9db6679506f43df2b579f5e3f2a8f178","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-26 09:17:49.000000000","message":"Uploaded patch set 40: New patch set was added with same tree, parent, and commit message as Patch Set 39.","accounts_in_message":[],"_revision_number":40},{"id":"272322e619c2cd9c5e5eaf98b64a6a48d1e5fb65","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-26 10:36:32.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/042b1c487f1e41a3bf1b6c95409ba1d6 : RETRY_LIMIT in 12m 58s","accounts_in_message":[],"_revision_number":40},{"id":"4f5ec4e6a924b89ea9d2b586cc1ed4d05095d89b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-26 12:08:42.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"fc04da7f851ab0e020999fc1cacdcea5d0b975ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 12:12:18.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0af333c22ef5490393eec1676260c8d8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9345702a4adf4c079ceee21aa65999bc : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97ab5fbf03124aa7a757e4a354323c1e : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79c57f45c4e04c74a77f380b93f105c9 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52b09d9455cc4083a3ade5630d27539d : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ecda087f379b4b8d8fe5bc98bf889137 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0daf7d74171d412b8eee909b2847b22e : SUCCESS in 4m 35s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/9cbee5775b344989a4bd6b627758046c : SUCCESS in 2h 41m 17s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/1d10dc115f4046829918e9c194bd8cad : SUCCESS in 1h 36m 59s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9558bb2f2b54462dad04b90a4798b70e : FAILURE in 1h 12m 13s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/983eba36e3f144be9ab6d48eaf6cfeba : SUCCESS in 41m 38s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/a12c307bbd6c4bb981f91d92feb0bcc0 : SUCCESS in 1h 11m 38s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/f18fb0250c684eac8127b341ecf3cfbc : SUCCESS in 1h 32m 16s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/98611194db2e45038d3d5512bc64072f : SUCCESS in 2h 35m 46s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b68f1ab1678143adb7b04bac291dfe0a : SUCCESS in 1h 22m 58s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/d242c7b8d157457da9bb15087020070e : SUCCESS in 1h 20m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/43fd4b2ffe404d1a9a5fb0809ffeb47e : SUCCESS in 4m 27s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/264feefea6bc4c6aa922eae2d92725c8 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":40},{"id":"783c43d9ce2f2eba6d5ed93eca7269c1774dde2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-29 04:36:40.000000000","message":"Patch Set 41: Patch Set 40 was rebased","accounts_in_message":[],"_revision_number":41},{"id":"edb81d5f951276b2f301e962c86ea8bf1adc040f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-29 06:42:29.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c1ce7e132f754f47ac1a7a82f9a39a0a : SUCCESS in 2h 04m 55s","accounts_in_message":[],"_revision_number":41},{"id":"08263d4e3cfb4b2cb0ddb2fbe6f0d7675dd2f640","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-29 07:29:21.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b73f976eb4934f8cb75a029ce156012c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adabec43e32e485898d4e37ef0d87874 : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb1fcb7d25c8462e877b5998d9959ec2 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c500be5a0ebb446da28827186140030a : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfc7fc7537dc4967844596fd5b6b5339 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c34fed89eac84432920603d3000953ab : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cda3b1f6571747f791ed504e4ddca240 : SUCCESS in 4m 50s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/0607e13cae5e49b8875a038357bc3d36 : SUCCESS in 2h 43m 07s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/03d0be12a2344f8a981a76262f951f61 : SUCCESS in 1h 41m 43s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/fd588b85085f4664874bff36763c14c4 : SUCCESS in 1h 31m 11s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/4fc5faf23db14c09bbdd4b0f839b553a : SUCCESS in 42m 05s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/571a321428f04ec9a0b8dd65e3f38659 : SUCCESS in 1h 15m 22s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/08885f237938493c8ce0fa1075be219b : SUCCESS in 1h 30m 23s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/a1e0f88b3ef444179c998eced8e80ab0 : SUCCESS in 2h 19m 35s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/32763e401d6b4cc9b0451cb8beb1a4be : SUCCESS in 1h 26m 01s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/5f7bfc926fe64e51a7fbdc999b3d4013 : SUCCESS in 1h 08m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9a8830375f9e44939ba7d65e13b6853a : SUCCESS in 3m 30s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/61aaf97b6db540a884081bb9c727a77e : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":41},{"id":"01a172cc84455566196371b08e7fdf9926f63650","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-29 13:45:04.000000000","message":"Topic dnm removed","accounts_in_message":[],"_revision_number":41},{"id":"507958790e2e29ced05cd285f67bfafa7379efff","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-08-29 13:45:11.000000000","message":"Topic set to tripleo-ansible-ee","accounts_in_message":[],"_revision_number":41},{"id":"77244c71fab6a2a266670f421758bad32f075484","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-01 12:15:41.000000000","message":"Patch Set 41:\n\n(2 comments)","accounts_in_message":[],"_revision_number":41},{"id":"469a32b7cfc40efa47d86f9b4f6282e100ac5275","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-05 10:49:19.000000000","message":"Uploaded patch set 42.","accounts_in_message":[],"_revision_number":42},{"id":"f657856fceb1a378e5200f41aeb93648ba726d99","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-05 13:02:35.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"15ec059740f922188d5fc93c0f4c9ababe6cd939","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-05 13:20:16.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5e0dd14490fc48c8abe2e3796e8329f3 : SUCCESS in 2h 08m 58s","accounts_in_message":[],"_revision_number":42},{"id":"c0cc1907ed7c29940234b55c20e02bd90c2b6d82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-05 13:41:45.000000000","message":"Uploaded patch set 43.","accounts_in_message":[],"_revision_number":43},{"id":"ae723346e66c8546ea1a79eb2bd1faf61aa313a4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-05 13:42:41.000000000","message":"Patch Set 43:\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"8b88f7ee317c987122c0b46dec3daeb0eb3054bc","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-05 13:43:08.000000000","message":"Patch Set 43:\n\n(2 comments)","accounts_in_message":[],"_revision_number":43},{"id":"6100cf93dd006467d17240f8f0522a6be5e2f857","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-05 13:45:16.000000000","message":"Uploaded patch set 44: Commit message was updated.","accounts_in_message":[],"_revision_number":44},{"id":"735a2622298277b6909e4dbdd99729ec3f5f7fe5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-05 15:41:40.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/71fe67ac9f414409aa40e63fc0bcaed4 : FAILURE in 1h 50m 36s","accounts_in_message":[],"_revision_number":44},{"id":"63e27234272add460d523c5b59a8563553641718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-05 17:29:35.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78f82d73696840e0965883aec0f743f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db059d0f642d4159b363aa26962b1461 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5fa8d3cbf1e48a8bc74fcca9d00b2a7 : SUCCESS in 3m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b74bc6baa184c5797d954855d3694a0 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bbdf89f4897444dac6df5e0122888f2 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ed449ec015c48f688ce3881eb807ad3 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cfc38fbab344627ba76c22d8ecc1951 : SUCCESS in 4m 00s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7ea668276d5041acb09130a485fc4ea5 : SUCCESS in 3h 28m 53s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/816704bb3d664ea3be1be2e5bc8ca031 : SUCCESS in 1h 52m 36s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7e96319e4bae41968753ee8d53593163 : SUCCESS in 1h 42m 06s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/6d30dd2561f14eacb777bb6d2f26475e : SUCCESS in 54m 33s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/30a2bd2e04e34314bc24a6be32f8c982 : SUCCESS in 1h 29m 16s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/31ebf3fca02d409887c46222182aa583 : SUCCESS in 1h 58m 02s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/0fda217b126e4063bf7fc038ba2917cf : SUCCESS in 3h 01m 04s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/84cf2be54f374357a1c86585a97aad34 : SUCCESS in 1h 47m 06s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/618dfd8143604f98be7cdb82f2d63a40 : SUCCESS in 1h 21m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d5d01e2d1bdc44dc84f981af08a51903 : SUCCESS in 4m 03s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/7c6f2f26d245437d9a40b93b5822a4b3 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":44},{"id":"2fcba90431cf24de63cca1e6a147eb1392b087c9","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-06 08:48:53.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"03ff59752d4c488be359476c8fb742e73989c321","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-12 04:38:46.000000000","message":"Uploaded patch set 45.","accounts_in_message":[],"_revision_number":45},{"id":"a5490b934e3b474c55ac51e22a879ed413bd23a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-12 05:10:16.000000000","message":"Uploaded patch set 46.","accounts_in_message":[],"_revision_number":46},{"id":"eea0ef12342235094ea9eb36ec5eed723cb23b92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 05:55:05.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c58fdaf12eb468a95e9a1696dff564e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fc739b32b934f70aa220b6a38ecd4ef : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b0cf350a74c4da7b29c01553dd02efc : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70edd27761e64ea78c8cb19be18d2f0f : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e8bab3b481f492d8819e1055dbcbbfb : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2836eb54c7a3468ba74471bbb616dcea : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b44a2a60f34cffad46094fd4349ba9 : SUCCESS in 4m 26s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/80b979bbe3b747efaf399b5ce8c20421 : SUCCESS in 43m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9e86f0225a15408b843c89fcb2f93b5f : FAILURE in 6m 36s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/4a2381eec01c45aebd1001a3f9944025 : SUCCESS in 5m 22s\nSkipped 8 jobs","accounts_in_message":[],"_revision_number":46},{"id":"4e4a1d174760b1e016a71784c9707a54d0f5a11c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-12 05:55:09.000000000","message":"Uploaded patch set 47.","accounts_in_message":[],"_revision_number":47},{"id":"36cdc40b86f07363a05151d40c1a23c2dbce8d72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 07:09:25.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e30592552a0040f1b2b480c23bd6ef64\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7d6ca347fbb4a67b90733bfe69b280c : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a7c725414294b0ba51b17eda34872d1 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae4acc63a59c48238c4fb0a86e2bc575 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb196286eea241a997a8cf7e41a53d01 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbff44c82f7f41c99a80ababb8be380b : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77f9237f750c453a969feb4664c64cf9 : SUCCESS in 4m 26s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b9914b9ea09448649c63493608bb8b92 : FAILURE in 44m 32s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/800b5f73f9b44b8e9bf4183034a9ec25 : SUCCESS in 1h 07m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5253a3a2627344c7b24f15dd34aed2a3 : SUCCESS in 4m 04s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/70ace4e5bc81407ab9b346cb26a22cc5 : SUCCESS in 4m 59s\nSkipped 7 jobs","accounts_in_message":[],"_revision_number":47},{"id":"f668516d83e097ea7050e906461e35f13ae6d8f5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-12 07:36:01.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e2c063962a1b4606b14f8b76eb7977d7 : FAILURE in 1h 40m 02s","accounts_in_message":[],"_revision_number":47},{"id":"0a93feccb9a71aba670ed34eded45c5bba221d2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-12 07:57:20.000000000","message":"Uploaded patch set 48.","accounts_in_message":[],"_revision_number":48},{"id":"6f3e3824b3601dd07596547774ea7509a33afc5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 09:17:43.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c14dac5e4584600a092b1e6de67f4e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0c8789c0ab24a10b15e085cfed18d0c : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aacb72dabe74548af2ed4d9ef32f0cb : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fab2ae80e8494362b39999042f8cdccd : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2daba76ebf54a00acc1afa6b2e73816 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9b30f66ce1e4ab38d533a13e46b32fb : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9546fd94d22e460d9b8976ad2a33cdaa : SUCCESS in 3m 53s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/3bd56ad7a4d44179af7a5ebe0d2307f1 : FAILURE in 27m 04s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/436c94ece537499580afb5e68b9c4e8b : SUCCESS in 1h 13m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/830d77800eca4058ab6433fd5d32b332 : SUCCESS in 4m 01s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/66833f2dac9445eabd6bdf97d783543e : SUCCESS in 5m 48s\nSkipped 7 jobs","accounts_in_message":[],"_revision_number":48},{"id":"25ea267cce111a2de8583796a103fff878f71248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-12 09:24:39.000000000","message":"Uploaded patch set 49.","accounts_in_message":[],"_revision_number":49},{"id":"04049c4ac93fc407135cc332a44d8079db883447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 10:34:09.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e648d762fb434910a7e4471d443813db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96a8dab8e85241d4a8e84aa066ce940e : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ada2acdbadbc45c58faffb59b47ae220 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7f3706001d249319afdbf57f95cc8fc : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/259c004ce84b4aa1b18b42cb1ac3fa29 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/85b4c257ec5f4739bf9d24d559cd86c6 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5c0d8d57fa545e99860ed1aa872c8e6 : SUCCESS in 4m 17s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c37c1c0d44dd4eab9031531951bb55a1 : FAILURE in 45m 30s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/91f14583dba14f3dbf61abad33d1f6bf : SUCCESS in 1h 08m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e5d89c775f254144a58587106d790f05 : SUCCESS in 4m 22s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/85044ee233e7420eae22323d7e4a01fb : SUCCESS in 4m 14s\nSkipped 7 jobs","accounts_in_message":[],"_revision_number":49},{"id":"e21d62b03155f27aeba7c6bb98704d4c0329411f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-12 10:48:03.000000000","message":"Uploaded patch set 50.","accounts_in_message":[],"_revision_number":50},{"id":"0c0ccf3b26bd0678c97b4cd70dcd062f5fcf3547","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-12 13:02:57.000000000","message":"Uploaded patch set 51.","accounts_in_message":[],"_revision_number":51},{"id":"8ec78a4f457c26320bd0d4214858ef3104750d12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-12 14:31:36.000000000","message":"Uploaded patch set 52.","accounts_in_message":[],"_revision_number":52},{"id":"7d97c7246deef6ae9ffd86edceaa66e5eff5f9d5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-12 16:29:27.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/02b811897a5a4f09b937e1458958db51 : FAILURE in 1h 56m 48s","accounts_in_message":[],"_revision_number":52},{"id":"71ad3222eb4ba4dceb8343eb21128c1dad9053a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 18:07:18.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8369b20b19d342329e9c2682a0880476\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cda167d41fc44a12871a910597cc000a : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dc3f42a109a412d969d08530363464d : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4509de6b67944f6684a6b28206a796d1 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36eaf25ab1e441289a8d19d3e4f7f7d6 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a40d8225a0144a0a7525da08edc34f4 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/739c4820cce24f698be3feac74a89424 : SUCCESS in 4m 08s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/42994e7085fa42949cac638c8f3fe0be : SUCCESS in 3h 29m 11s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/0b031e70b08f4d43ac40aea1d2b0c5ae : SUCCESS in 1h 53m 17s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2c0528b78e4948c78455a8e28ee0c481 : SUCCESS in 1h 39m 45s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/53746b6d04d246378769b4dfe018b7a9 : SUCCESS in 53m 25s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/2604098000e24ad9a30909c7507cde55 : SUCCESS in 1h 33m 35s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/c940d09c52234af6972e4ec7bd99500b : SUCCESS in 1h 56m 09s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/8f7e96323772458aa1601badc0f3a24b : SUCCESS in 2h 49m 13s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ba5cd4d8de3e4a68be667ace524eb0b7 : SUCCESS in 1h 34m 58s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/33f6f6c07a3f4fefbfcd91bf291498ce : SUCCESS in 1h 12m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aeb0237663bb43949f642d294ef8fd04 : SUCCESS in 4m 30s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/90f6713f75a04b3488dc083e726c4583 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":52},{"id":"4269fd189797d46db2c5446ed5a283332fa2a799","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-14 06:33:30.000000000","message":"Uploaded patch set 53.","accounts_in_message":[],"_revision_number":53},{"id":"264d73dee9011720e197c8d4a3b846643fb98b38","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-14 08:33:30.000000000","message":"Patch Set 53: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/96eeae3a63c84257b1c93c99219a642d : FAILURE in 1h 58m 51s","accounts_in_message":[],"_revision_number":53},{"id":"87ac09b44f9f94f8e708b2d957d7047e75920d9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-09-14 10:15:09.000000000","message":"Uploaded patch set 54.","accounts_in_message":[],"_revision_number":54},{"id":"4bdb40428652d46c8f1f70475c4093332881a13d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 13:57:54.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55ca169422e94087ac5f738de30050ec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fbb63d900394c7aa4c292f945ffe4cd : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a2e7d0c2b54c29a7c6957af3ae0a91 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86aaebba45444135b923932cebac3661 : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83ba0a3056284676bf1d27557f062c68 : SUCCESS in 7m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/747f546d453f4533a82e87050001b032 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf83c50d25cf4294b8ea719c0bcb702b : SUCCESS in 4m 58s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/3a7cb1cd64b64c408876d7f7c9d80465 : SUCCESS in 3h 25m 35s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/1af533b9ca9d436c907d306a81232096 : SUCCESS in 1h 48m 08s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f88a04c2987c45978386e390c1ff3750 : SUCCESS in 1h 45m 57s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/73457eabb07d44aca8e09831be7aaf09 : SUCCESS in 47m 26s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/088a037ba57f489da59e139f1a532158 : SUCCESS in 1h 21m 45s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/b5f6e73a3eef49c78e00eae4e85e8b13 : SUCCESS in 1h 54m 50s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/0fcdcb6614714355a2008107ae28483f : SUCCESS in 3h 03m 03s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/740fd8b88d5a4d9ba4c8d9ac5e9f597d : SUCCESS in 1h 42m 36s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/448b6aa20c5c448c8b1a6d0ccf0648ff : SUCCESS in 1h 17m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9f38510559604c5b979505d5622ad14e : SUCCESS in 4m 41s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/6552692108dd4ae3bd635bf55526ee1c : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":54},{"id":"2ba727701a306687d7231c5ca3a7e2b8f3dd078e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-09-14 15:07:31.000000000","message":"Patch Set 54: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e2cb96b9d47949e08b772f765f054ca9 : FAILURE in 2h 31m 19s","accounts_in_message":[],"_revision_number":54},{"id":"d688e8f53dbcead1d83995045c739fd5fabeaf02","author":{"_account_id":32968,"name":"Juan Larriba","email":"jlarriba@redhat.com","username":"jlarriba"},"date":"2022-09-15 12:02:02.000000000","message":"Patch Set 54: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":54},{"id":"b1416a27f1174167fec8bc079e590c434ae7b846","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-09-15 12:07:26.000000000","message":"Patch Set 54: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":54},{"id":"909c9fec5093e3beb637e43f182458c149211e05","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-09-15 12:26:31.000000000","message":"Patch Set 54: Code-Review+2","accounts_in_message":[],"_revision_number":54},{"id":"381ee37154db4830cf7baf3723400c57b83e296f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-09-15 12:30:54.000000000","message":"Patch Set 54: Workflow+1","accounts_in_message":[],"_revision_number":54},{"id":"85d1bab3ec6df152871cfd6a5d9164e14d103b59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 12:31:41.000000000","message":"Patch Set 54: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":54},{"id":"2642f0f468d0d545b8be46d6a821ebe11d6614b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 16:04:56.000000000","message":"Patch Set 54: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ac343c80f84441d82ce68eeccfb1e39\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51e85eff69b241d18b4c515b448421ae : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f873a4631d42496a9b0f3fd5497461a0 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ade75fc90074328bc3945b631f2957b : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbb88199eaea407bb3f3b976ceff9f0d : SUCCESS in 4m 01s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c0ee424bb48d418eb0d2a6e454e25c2f : SUCCESS in 3h 27m 10s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/e591994f4df042ce905cd024ff853afd : SUCCESS in 1h 56m 23s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/28f4ce401b454b7eac955799b7789af5 : SUCCESS in 52m 16s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/a16b68ad6e334f9a84634a626f160073 : SUCCESS in 1h 24m 13s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f0c9be654c2a44478efdbdae7ebe0a8c : SUCCESS in 2h 45m 57s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/4393e14d34964017bfe812dc2c1e8fa7 : SUCCESS in 1h 14m 51s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c465a997e03d43dfac28c8334a51db35 : SUCCESS in 1h 49m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/744880489965476ab720d8cf6eeb9cf5 : SUCCESS in 3m 59s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/5c1d380d3b304882983ffc4a329a0f54 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":54},{"id":"2af0b2b35711c991965cef9c4ac3458589060593","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 16:04:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":54},{"id":"642af22baa9e4b5423710b240314803ee51e6c31","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 16:06:09.000000000","message":"Patch Set 54:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc646c489f6a4bf2938c179069178a85\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce8a7b4c7c8c4680aac8c129a1d0fd61 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":54},{"id":"c8b2c3e94ad2361f2d93d32d8801da96a939f80f","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2022-09-16 02:46:19.000000000","message":"Patch Set 54: Code-Review+1","accounts_in_message":[],"_revision_number":54}],"current_revision_number":54,"current_revision":"54a6bac5d557aa84c21986e96db0e0911792fe1b","revisions":{"3c33437ba2857ef0ca890ea5dd95bdd72a9524fe":{"kind":"REWORK","_number":1,"created":"2021-02-19 17:00:54.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/74/776674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/1"}}},"commit":{"parents":[{"commit":"64b6411d6dc234fd0d4207acde5b5f894fabd777","subject":"Merge \"Cleanup dead mistral actions and workbooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/64b6411d6dc234fd0d4207acde5b5f894fabd777"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 17:00:52.000000000","tz":-360},"subject":"[WIP] Add Ansible execution environment","message":"[WIP] Add Ansible execution environment\n\nThis change adds an ansible execution environment with all of the\npackages we\u0027ll need to be able to run Ansible from within a\ntripleo compatible container.\n\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3c33437ba2857ef0ca890ea5dd95bdd72a9524fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3c33437ba2857ef0ca890ea5dd95bdd72a9524fe"}]},"branch":"refs/heads/master"},"6d686b995762c9a4b5856dbb2495261c8c063498":{"kind":"REWORK","_number":2,"created":"2021-02-22 17:35:24.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/74/776674/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/2"}}},"commit":{"parents":[{"commit":"64b6411d6dc234fd0d4207acde5b5f894fabd777","subject":"Merge \"Cleanup dead mistral actions and workbooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/64b6411d6dc234fd0d4207acde5b5f894fabd777"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-22 17:35:16.000000000","tz":-360},"subject":"[WIP] Add Ansible execution environment","message":"[WIP] Add Ansible execution environment\n\nThis change adds an ansible execution environment with all of the\npackages we\u0027ll need to be able to run Ansible from within a\ntripleo compatible container.\n\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6d686b995762c9a4b5856dbb2495261c8c063498"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6d686b995762c9a4b5856dbb2495261c8c063498"}]},"branch":"refs/heads/master"},"5feb3886933ed5f4bac89b8e949c0caeebaa81ad":{"kind":"REWORK","_number":3,"created":"2021-02-24 15:38:22.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/74/776674/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/3"}}},"commit":{"parents":[{"commit":"64b6411d6dc234fd0d4207acde5b5f894fabd777","subject":"Merge \"Cleanup dead mistral actions and workbooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/64b6411d6dc234fd0d4207acde5b5f894fabd777"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-24 15:38:15.000000000","tz":-360},"subject":"[WIP] Add Ansible execution environment","message":"[WIP] Add Ansible execution environment\n\nThis change adds an ansible execution environment with all of the\npackages we\u0027ll need to be able to run Ansible from within a\ntripleo compatible container.\n\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5feb3886933ed5f4bac89b8e949c0caeebaa81ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5feb3886933ed5f4bac89b8e949c0caeebaa81ad"}]},"branch":"refs/heads/master"},"cbaf023d3678e8033171f4a7de557660087162bf":{"kind":"REWORK","_number":4,"created":"2021-02-24 20:29:40.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/74/776674/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/4"}}},"commit":{"parents":[{"commit":"64b6411d6dc234fd0d4207acde5b5f894fabd777","subject":"Merge \"Cleanup dead mistral actions and workbooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/64b6411d6dc234fd0d4207acde5b5f894fabd777"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-24 20:29:33.000000000","tz":-360},"subject":"[WIP] Add Ansible execution environment","message":"[WIP] Add Ansible execution environment\n\nThis change adds an ansible execution environment with all of the\npackages we\u0027ll need to be able to run Ansible from within a\ntripleo compatible container.\n\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cbaf023d3678e8033171f4a7de557660087162bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cbaf023d3678e8033171f4a7de557660087162bf"}]},"branch":"refs/heads/master"},"cdfeacd2dee1a01aa8b39a9575f3341dc2876a59":{"kind":"REWORK","_number":5,"created":"2021-03-15 13:44:28.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/74/776674/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/5"}}},"commit":{"parents":[{"commit":"64b6411d6dc234fd0d4207acde5b5f894fabd777","subject":"Merge \"Cleanup dead mistral actions and workbooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/64b6411d6dc234fd0d4207acde5b5f894fabd777"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kevin@cloudnull.com","date":"2021-03-15 13:44:06.000000000","tz":0},"subject":"[WIP] Add Ansible execution environment","message":"[WIP] Add Ansible execution environment\n\nThis change adds an ansible execution environment with all of the\npackages we\u0027ll need to be able to run Ansible from within a\ntripleo compatible container.\n\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cdfeacd2dee1a01aa8b39a9575f3341dc2876a59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cdfeacd2dee1a01aa8b39a9575f3341dc2876a59"}]},"branch":"refs/heads/master"},"a7770b8c9fee28ed65de8228d7959c9cb03a4819":{"kind":"REWORK","_number":6,"created":"2022-01-11 20:50:03.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/74/776674/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/6"}}},"commit":{"parents":[{"commit":"6688b56221842eb40b0191243640dc96f4c14ff9","subject":"Don\u0027t enable module stream in rhel9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6688b56221842eb40b0191243640dc96f4c14ff9"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2022-01-11 20:50:25.000000000","tz":-360},"subject":"[WIP] Add Ansible execution environment","message":"[WIP] Add Ansible execution environment\n\nThis change adds an ansible execution environment with all of the\npackages we\u0027ll need to be able to run Ansible from within a\ntripleo compatible container.\n\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a7770b8c9fee28ed65de8228d7959c9cb03a4819"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a7770b8c9fee28ed65de8228d7959c9cb03a4819"}]},"branch":"refs/heads/master"},"8cc71353b4b3b192226300df9aaf03e44d141071":{"kind":"REWORK","_number":7,"created":"2022-04-22 21:59:09.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/74/776674/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/7"}}},"commit":{"parents":[{"commit":"8898142b3322f01dc60b6b8855c841c65e7712d4","subject":"Merge \"Switch from grub2 to bootloader element for overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8898142b3322f01dc60b6b8855c841c65e7712d4"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2022-04-22 21:58:59.000000000","tz":-300},"subject":"[WIP] Add Ansible execution environment","message":"[WIP] Add Ansible execution environment\n\nThis change adds an ansible execution environment with all of the\npackages we\u0027ll need to be able to run Ansible from within a\ntripleo compatible container.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8cc71353b4b3b192226300df9aaf03e44d141071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8cc71353b4b3b192226300df9aaf03e44d141071"}]},"branch":"refs/heads/master"},"f05aaaf68cb1c577ee24eb33446d302a3b336a65":{"kind":"REWORK","_number":8,"created":"2022-04-25 14:03:58.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/74/776674/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/8"}}},"commit":{"parents":[{"commit":"8898142b3322f01dc60b6b8855c841c65e7712d4","subject":"Merge \"Switch from grub2 to bootloader element for overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8898142b3322f01dc60b6b8855c841c65e7712d4"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2022-04-25 14:03:48.000000000","tz":-300},"subject":"[WIP] Add Ansible execution environment","message":"[WIP] Add Ansible execution environment\n\nThis change adds an ansible execution environment with all of the\npackages we\u0027ll need to be able to run Ansible from within a\ntripleo compatible container.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f05aaaf68cb1c577ee24eb33446d302a3b336a65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f05aaaf68cb1c577ee24eb33446d302a3b336a65"}]},"branch":"refs/heads/master"},"baccba524b2f1015a3088fe5991e5e74ca23e165":{"kind":"REWORK","_number":9,"created":"2022-04-25 15:46:46.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/74/776674/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/9"}}},"commit":{"parents":[{"commit":"8898142b3322f01dc60b6b8855c841c65e7712d4","subject":"Merge \"Switch from grub2 to bootloader element for overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8898142b3322f01dc60b6b8855c841c65e7712d4"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2022-04-25 15:46:39.000000000","tz":-300},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/baccba524b2f1015a3088fe5991e5e74ca23e165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/baccba524b2f1015a3088fe5991e5e74ca23e165"}]},"branch":"refs/heads/master"},"99997eee85589d1431a62b0a4b50608ffdbc147e":{"kind":"REWORK","_number":10,"created":"2022-04-28 15:57:40.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/74/776674/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/10"}}},"commit":{"parents":[{"commit":"8898142b3322f01dc60b6b8855c841c65e7712d4","subject":"Merge \"Switch from grub2 to bootloader element for overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8898142b3322f01dc60b6b8855c841c65e7712d4"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kevin@cloudnull.com","date":"2022-04-28 15:57:39.000000000","tz":0},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/99997eee85589d1431a62b0a4b50608ffdbc147e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/99997eee85589d1431a62b0a4b50608ffdbc147e"}]},"branch":"refs/heads/master"},"a30111b4812c5585972673b858a24628de287e52":{"kind":"REWORK","_number":11,"created":"2022-05-02 21:29:04.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/74/776674/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/11"}}},"commit":{"parents":[{"commit":"8898142b3322f01dc60b6b8855c841c65e7712d4","subject":"Merge \"Switch from grub2 to bootloader element for overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8898142b3322f01dc60b6b8855c841c65e7712d4"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kevin@cloudnull.com","date":"2022-05-02 21:28:20.000000000","tz":0},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a30111b4812c5585972673b858a24628de287e52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a30111b4812c5585972673b858a24628de287e52"}]},"branch":"refs/heads/master"},"b00e6e7d40c01d5288d84997a7a1b5699ff2e433":{"kind":"REWORK","_number":12,"created":"2022-05-03 13:59:10.000000000","uploader":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"ref":"refs/changes/74/776674/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/12"}}},"commit":{"parents":[{"commit":"8898142b3322f01dc60b6b8855c841c65e7712d4","subject":"Merge \"Switch from grub2 to bootloader element for overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8898142b3322f01dc60b6b8855c841c65e7712d4"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Kevin Carter","email":"kevin@cloudnull.com","date":"2022-05-03 13:59:10.000000000","tz":0},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b00e6e7d40c01d5288d84997a7a1b5699ff2e433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b00e6e7d40c01d5288d84997a7a1b5699ff2e433"}]},"branch":"refs/heads/master"},"f8b25b5eb877fb75226a4906d56190e4dd4e9935":{"kind":"REWORK","_number":13,"created":"2022-07-19 09:12:53.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/13"}}},"commit":{"parents":[{"commit":"8898142b3322f01dc60b6b8855c841c65e7712d4","subject":"Merge \"Switch from grub2 to bootloader element for overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8898142b3322f01dc60b6b8855c841c65e7712d4"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-19 09:10:47.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt will also contains tripleo-ansible roles, collections\nand modules by default.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f8b25b5eb877fb75226a4906d56190e4dd4e9935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f8b25b5eb877fb75226a4906d56190e4dd4e9935"}]},"branch":"refs/heads/master"},"ebd60ea59aaa83411c22c67dcb57c8ad8dae123f":{"kind":"REWORK","_number":14,"created":"2022-07-22 07:34:09.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/14"}}},"commit":{"parents":[{"commit":"8898142b3322f01dc60b6b8855c841c65e7712d4","subject":"Merge \"Switch from grub2 to bootloader element for overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8898142b3322f01dc60b6b8855c841c65e7712d4"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-22 07:31:05.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible is commented out.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ebd60ea59aaa83411c22c67dcb57c8ad8dae123f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ebd60ea59aaa83411c22c67dcb57c8ad8dae123f"}]},"branch":"refs/heads/master"},"f4a6ae385db0c8391365a91dd8acd831df6e5e98":{"kind":"REWORK","_number":15,"created":"2022-07-22 08:30:59.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/15"}}},"commit":{"parents":[{"commit":"8898142b3322f01dc60b6b8855c841c65e7712d4","subject":"Merge \"Switch from grub2 to bootloader element for overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8898142b3322f01dc60b6b8855c841c65e7712d4"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-22 08:30:46.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible is commented out.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f4a6ae385db0c8391365a91dd8acd831df6e5e98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f4a6ae385db0c8391365a91dd8acd831df6e5e98"}]},"branch":"refs/heads/master"},"3b346c380ff75397a940864e9bbe797d24bfc9ff":{"kind":"REWORK","_number":16,"created":"2022-07-26 07:35:34.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/16"}}},"commit":{"parents":[{"commit":"8898142b3322f01dc60b6b8855c841c65e7712d4","subject":"Merge \"Switch from grub2 to bootloader element for overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8898142b3322f01dc60b6b8855c841c65e7712d4"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-26 07:33:49.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3b346c380ff75397a940864e9bbe797d24bfc9ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3b346c380ff75397a940864e9bbe797d24bfc9ff"}]},"branch":"refs/heads/master"},"4f231cc758067ed07bb3933a38c7498d442fa2ac":{"kind":"REWORK","_number":17,"created":"2022-07-26 09:40:33.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/17"}}},"commit":{"parents":[{"commit":"8898142b3322f01dc60b6b8855c841c65e7712d4","subject":"Merge \"Switch from grub2 to bootloader element for overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8898142b3322f01dc60b6b8855c841c65e7712d4"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-26 09:40:21.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4f231cc758067ed07bb3933a38c7498d442fa2ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4f231cc758067ed07bb3933a38c7498d442fa2ac"}]},"branch":"refs/heads/master"},"74c5ee162879da95a72729345211614cc9957eca":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2022-07-26 10:02:01.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/18"}}},"commit":{"parents":[{"commit":"0b83227bdbe9e7905daa35f0f88f830a793caccd","subject":"Merge \"Remove unused oslo.rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0b83227bdbe9e7905daa35f0f88f830a793caccd"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2022-07-26 10:02:01.000000000","tz":0},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/74c5ee162879da95a72729345211614cc9957eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/74c5ee162879da95a72729345211614cc9957eca"}]},"branch":"refs/heads/master","description":"Rebase"},"984b5e0b30cb7dc7cf6dc53818cd6123942810ab":{"kind":"REWORK","_number":19,"created":"2022-07-26 10:35:27.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/19"}}},"commit":{"parents":[{"commit":"0b83227bdbe9e7905daa35f0f88f830a793caccd","subject":"Merge \"Remove unused oslo.rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0b83227bdbe9e7905daa35f0f88f830a793caccd"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-26 10:35:00.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/984b5e0b30cb7dc7cf6dc53818cd6123942810ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/984b5e0b30cb7dc7cf6dc53818cd6123942810ab"}]},"branch":"refs/heads/master"},"62284ad2d8d4aee7a07f064ed406fb5e7737ffb3":{"kind":"REWORK","_number":20,"created":"2022-07-26 10:59:45.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/20"}}},"commit":{"parents":[{"commit":"0b83227bdbe9e7905daa35f0f88f830a793caccd","subject":"Merge \"Remove unused oslo.rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0b83227bdbe9e7905daa35f0f88f830a793caccd"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-26 10:59:33.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/62284ad2d8d4aee7a07f064ed406fb5e7737ffb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/62284ad2d8d4aee7a07f064ed406fb5e7737ffb3"}]},"branch":"refs/heads/master"},"ffaacecc2569d3b51dd4dd16c658925776e1b935":{"kind":"REWORK","_number":21,"created":"2022-07-29 04:00:26.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/21"}}},"commit":{"parents":[{"commit":"0b83227bdbe9e7905daa35f0f88f830a793caccd","subject":"Merge \"Remove unused oslo.rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0b83227bdbe9e7905daa35f0f88f830a793caccd"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-29 04:00:14.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ffaacecc2569d3b51dd4dd16c658925776e1b935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ffaacecc2569d3b51dd4dd16c658925776e1b935"}]},"branch":"refs/heads/master"},"3be02542d2d4e485013cc72508dcfbe00fd90602":{"kind":"REWORK","_number":22,"created":"2022-07-29 04:42:59.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/22"}}},"commit":{"parents":[{"commit":"0b83227bdbe9e7905daa35f0f88f830a793caccd","subject":"Merge \"Remove unused oslo.rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0b83227bdbe9e7905daa35f0f88f830a793caccd"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-29 04:42:48.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3be02542d2d4e485013cc72508dcfbe00fd90602"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3be02542d2d4e485013cc72508dcfbe00fd90602"}]},"branch":"refs/heads/master"},"b869c512cb3edf0c7b118db6262fc7e5eaa48d2c":{"kind":"REWORK","_number":23,"created":"2022-07-29 05:12:03.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/23"}}},"commit":{"parents":[{"commit":"0b83227bdbe9e7905daa35f0f88f830a793caccd","subject":"Merge \"Remove unused oslo.rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0b83227bdbe9e7905daa35f0f88f830a793caccd"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-29 05:11:51.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b869c512cb3edf0c7b118db6262fc7e5eaa48d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b869c512cb3edf0c7b118db6262fc7e5eaa48d2c"}]},"branch":"refs/heads/master"},"bcd7b9dc1376cb574e25edf3f5d4bee345f2e9e3":{"kind":"REWORK","_number":24,"created":"2022-07-29 11:39:59.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/24"}}},"commit":{"parents":[{"commit":"0b83227bdbe9e7905daa35f0f88f830a793caccd","subject":"Merge \"Remove unused oslo.rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0b83227bdbe9e7905daa35f0f88f830a793caccd"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-07-29 11:39:48.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bcd7b9dc1376cb574e25edf3f5d4bee345f2e9e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bcd7b9dc1376cb574e25edf3f5d4bee345f2e9e3"}]},"branch":"refs/heads/master"},"fc6b9373512333ffe9d803226a03be9c99ba30ca":{"kind":"REWORK","_number":25,"created":"2022-07-29 12:51:01.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/25"}}},"commit":{"parents":[{"commit":"0b83227bdbe9e7905daa35f0f88f830a793caccd","subject":"Merge \"Remove unused oslo.rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0b83227bdbe9e7905daa35f0f88f830a793caccd"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2022-07-29 12:51:00.000000000","tz":0},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fc6b9373512333ffe9d803226a03be9c99ba30ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fc6b9373512333ffe9d803226a03be9c99ba30ca"}]},"branch":"refs/heads/master"},"f8f4de62a8db084f9029fe34f2b12d6e650ca34d":{"kind":"REWORK","_number":26,"created":"2022-08-01 06:02:14.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/26"}}},"commit":{"parents":[{"commit":"0b83227bdbe9e7905daa35f0f88f830a793caccd","subject":"Merge \"Remove unused oslo.rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0b83227bdbe9e7905daa35f0f88f830a793caccd"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-01 06:02:01.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f8f4de62a8db084f9029fe34f2b12d6e650ca34d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f8f4de62a8db084f9029fe34f2b12d6e650ca34d"}]},"branch":"refs/heads/master"},"1dc8b40e7331e6aefe84d759174d4b0f6317dafd":{"kind":"REWORK","_number":27,"created":"2022-08-01 07:24:02.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/27"}}},"commit":{"parents":[{"commit":"0b83227bdbe9e7905daa35f0f88f830a793caccd","subject":"Merge \"Remove unused oslo.rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0b83227bdbe9e7905daa35f0f88f830a793caccd"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-01 07:23:50.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1dc8b40e7331e6aefe84d759174d4b0f6317dafd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1dc8b40e7331e6aefe84d759174d4b0f6317dafd"}]},"branch":"refs/heads/master"},"fe9b44c7b8ee6f0a552f1f210a694a6dc390485c":{"kind":"REWORK","_number":28,"created":"2022-08-01 08:14:09.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/28"}}},"commit":{"parents":[{"commit":"0b83227bdbe9e7905daa35f0f88f830a793caccd","subject":"Merge \"Remove unused oslo.rootwrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0b83227bdbe9e7905daa35f0f88f830a793caccd"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-01 08:13:57.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nCurrently ansible-runner image is based on CS8 and tripleo-ansible\nstandalone work is based on CS9 master. In order to consume the\ntripleo-ansible-ee image for external compute work, It needs\nto be built on cs8 wallaby.\n\nIn order to avoid mixing of tripleo-ansible cs9 content,\ninstallation of tripleo-ansible from rpm is commented out.\nIt is currently installed from git.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fe9b44c7b8ee6f0a552f1f210a694a6dc390485c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fe9b44c7b8ee6f0a552f1f210a694a6dc390485c"}]},"branch":"refs/heads/master"},"efe421d68c4ed4150249fc826697b055005903bd":{"kind":"REWORK","_number":29,"created":"2022-08-05 08:00:56.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/29"}}},"commit":{"parents":[{"commit":"0100f8b1fb129515351ef125f3c8286a57757656","subject":"Include required files for tripleo-ansible-ee container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0100f8b1fb129515351ef125f3c8286a57757656"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-05 08:00:00.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/efe421d68c4ed4150249fc826697b055005903bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/efe421d68c4ed4150249fc826697b055005903bd"}]},"branch":"refs/heads/master"},"462bb25187e438e1ab8f621aebb5073cec7bf4ff":{"kind":"REWORK","_number":30,"created":"2022-08-05 09:19:31.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/30"}}},"commit":{"parents":[{"commit":"0100f8b1fb129515351ef125f3c8286a57757656","subject":"Include required files for tripleo-ansible-ee container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0100f8b1fb129515351ef125f3c8286a57757656"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-05 09:19:17.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/462bb25187e438e1ab8f621aebb5073cec7bf4ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/462bb25187e438e1ab8f621aebb5073cec7bf4ff"}]},"branch":"refs/heads/master"},"60a44b798501ccd10dc4d5af0b1298f5f5644a73":{"kind":"REWORK","_number":31,"created":"2022-08-05 09:22:29.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/31"}}},"commit":{"parents":[{"commit":"0100f8b1fb129515351ef125f3c8286a57757656","subject":"Include required files for tripleo-ansible-ee container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0100f8b1fb129515351ef125f3c8286a57757656"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-05 09:22:15.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/60a44b798501ccd10dc4d5af0b1298f5f5644a73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/60a44b798501ccd10dc4d5af0b1298f5f5644a73"}]},"branch":"refs/heads/master"},"50e7d3f923ce8c0dce45ad4949d59d681df122e7":{"kind":"REWORK","_number":32,"created":"2022-08-05 11:56:16.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/32"}}},"commit":{"parents":[{"commit":"0100f8b1fb129515351ef125f3c8286a57757656","subject":"Include required files for tripleo-ansible-ee container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0100f8b1fb129515351ef125f3c8286a57757656"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-05 11:55:59.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/50e7d3f923ce8c0dce45ad4949d59d681df122e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/50e7d3f923ce8c0dce45ad4949d59d681df122e7"}]},"branch":"refs/heads/master"},"f9579476067371516e5e9e7020873c0371a9db2c":{"kind":"REWORK","_number":33,"created":"2022-08-08 05:07:45.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/33"}}},"commit":{"parents":[{"commit":"0100f8b1fb129515351ef125f3c8286a57757656","subject":"Include required files for tripleo-ansible-ee container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0100f8b1fb129515351ef125f3c8286a57757656"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-08 05:07:19.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f9579476067371516e5e9e7020873c0371a9db2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f9579476067371516e5e9e7020873c0371a9db2c"}]},"branch":"refs/heads/master"},"187368048723f08c5a01a7af9b72903668fc962e":{"kind":"REWORK","_number":34,"created":"2022-08-09 09:38:32.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/34"}}},"commit":{"parents":[{"commit":"0100f8b1fb129515351ef125f3c8286a57757656","subject":"Include required files for tripleo-ansible-ee container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0100f8b1fb129515351ef125f3c8286a57757656"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-09 09:38:16.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/187368048723f08c5a01a7af9b72903668fc962e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/187368048723f08c5a01a7af9b72903668fc962e"}]},"branch":"refs/heads/master"},"5c7d1c53b8870abec73ba70253e5f8c751b6a07f":{"kind":"REWORK","_number":35,"created":"2022-08-10 08:37:27.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/35"}}},"commit":{"parents":[{"commit":"45e6d58c56f879929d09cdc4f6a2f6df0b14ce23","subject":"Include required files for tripleo-ansible-ee container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/45e6d58c56f879929d09cdc4f6a2f6df0b14ce23"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-10 08:34:33.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5c7d1c53b8870abec73ba70253e5f8c751b6a07f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5c7d1c53b8870abec73ba70253e5f8c751b6a07f"}]},"branch":"refs/heads/master"},"294ae93691acfaeda15edae225f458d48f0076d8":{"kind":"NO_CODE_CHANGE","_number":36,"created":"2022-08-10 09:40:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/36"}}},"commit":{"parents":[{"commit":"45e6d58c56f879929d09cdc4f6a2f6df0b14ce23","subject":"Include required files for tripleo-ansible-ee container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/45e6d58c56f879929d09cdc4f6a2f6df0b14ce23"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2022-08-10 09:40:40.000000000","tz":0},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/294ae93691acfaeda15edae225f458d48f0076d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/294ae93691acfaeda15edae225f458d48f0076d8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"beb2c19dfef262ceb8c0c6e57b6cf22011760c52":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2022-08-12 09:34:56.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/37"}}},"commit":{"parents":[{"commit":"6602511720900041bd3b94e6bdbf479fcb890904","subject":"Merge \"Support standard partition type GUIDs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6602511720900041bd3b94e6bdbf479fcb890904"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2022-08-12 09:34:56.000000000","tz":0},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/beb2c19dfef262ceb8c0c6e57b6cf22011760c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/beb2c19dfef262ceb8c0c6e57b6cf22011760c52"}]},"branch":"refs/heads/master","description":"Rebase"},"55ef2cff789a976a0bc5ef92ba0ac66fcb77e89b":{"kind":"REWORK","_number":38,"created":"2022-08-12 09:40:58.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/38"}}},"commit":{"parents":[{"commit":"6602511720900041bd3b94e6bdbf479fcb890904","subject":"Merge \"Support standard partition type GUIDs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6602511720900041bd3b94e6bdbf479fcb890904"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-12 09:37:30.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nentrypoint.sh file contains the arbitrary uid\u0027s so that it\nwill be consumed with OpenShift Operator.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/55ef2cff789a976a0bc5ef92ba0ac66fcb77e89b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/55ef2cff789a976a0bc5ef92ba0ac66fcb77e89b"}]},"branch":"refs/heads/master"},"3f01abd98d5f1ca7c9d6c4890ebf620927d8cb96":{"kind":"REWORK","_number":39,"created":"2022-08-26 08:49:30.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/39"}}},"commit":{"parents":[{"commit":"6602511720900041bd3b94e6bdbf479fcb890904","subject":"Merge \"Support standard partition type GUIDs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6602511720900041bd3b94e6bdbf479fcb890904"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-26 08:48:55.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3f01abd98d5f1ca7c9d6c4890ebf620927d8cb96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3f01abd98d5f1ca7c9d6c4890ebf620927d8cb96"}]},"branch":"refs/heads/master"},"d26a2047c9dadc5a643d0f375198b1a07ba66fc5":{"kind":"NO_CHANGE","_number":40,"created":"2022-08-26 09:17:49.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/40"}}},"commit":{"parents":[{"commit":"6602511720900041bd3b94e6bdbf479fcb890904","subject":"Merge \"Support standard partition type GUIDs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6602511720900041bd3b94e6bdbf479fcb890904"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-08-26 08:55:43.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d26a2047c9dadc5a643d0f375198b1a07ba66fc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d26a2047c9dadc5a643d0f375198b1a07ba66fc5"}]},"branch":"refs/heads/master"},"3e918194597e7489cd2c56d55830d402038be51c":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2022-08-29 04:36:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/41"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2022-08-29 04:36:40.000000000","tz":0},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3e918194597e7489cd2c56d55830d402038be51c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3e918194597e7489cd2c56d55830d402038be51c"}]},"branch":"refs/heads/master","description":"Rebase"},"94ec595808fcb7f69d9d7aa845997a6058cf6fb4":{"kind":"REWORK","_number":42,"created":"2022-09-05 10:49:19.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/42"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-05 10:49:06.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/94ec595808fcb7f69d9d7aa845997a6058cf6fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/94ec595808fcb7f69d9d7aa845997a6058cf6fb4"}]},"branch":"refs/heads/master"},"5308e8477b78aeb541f97df99f62b3a26305b6f8":{"kind":"REWORK","_number":43,"created":"2022-09-05 13:41:45.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/43"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-05 13:41:19.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5308e8477b78aeb541f97df99f62b3a26305b6f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5308e8477b78aeb541f97df99f62b3a26305b6f8"}]},"branch":"refs/heads/master"},"43466ea170e53cb5edb785fa2f1b0f315cb22377":{"kind":"NO_CODE_CHANGE","_number":44,"created":"2022-09-05 13:45:16.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/44"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-05 13:44:21.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nThis image will be consumed in\ntripleo-ci-centos-9-singlenode-external-compute-via-ee job defined in\nhttps://review.opendev.org/c/openstack/tripleo-ci/+/843836.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/43466ea170e53cb5edb785fa2f1b0f315cb22377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/43466ea170e53cb5edb785fa2f1b0f315cb22377"}]},"branch":"refs/heads/master"},"e378d3e7b4179e8e12753d80503d8a1694db92b5":{"kind":"REWORK","_number":45,"created":"2022-09-12 04:38:46.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/45"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-12 04:38:32.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nThis image will be consumed in\ntripleo-ci-centos-9-singlenode-external-compute-via-ee job defined in\nhttps://review.opendev.org/c/openstack/tripleo-ci/+/843836.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e378d3e7b4179e8e12753d80503d8a1694db92b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e378d3e7b4179e8e12753d80503d8a1694db92b5"}]},"branch":"refs/heads/master"},"e99e0bdd5e531edb3d6d3a7bf740722a8ef088bf":{"kind":"REWORK","_number":46,"created":"2022-09-12 05:10:16.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/46"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-12 05:10:02.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nThis image will be consumed in\ntripleo-ci-centos-9-singlenode-external-compute-via-ee job defined in\nhttps://review.opendev.org/c/openstack/tripleo-ci/+/843836.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e99e0bdd5e531edb3d6d3a7bf740722a8ef088bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e99e0bdd5e531edb3d6d3a7bf740722a8ef088bf"}]},"branch":"refs/heads/master"},"d7eba58f0fd3db0db1fea727b7054206c380ffe4":{"kind":"REWORK","_number":47,"created":"2022-09-12 05:55:09.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/47"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-12 05:54:55.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nThis image will be consumed in\ntripleo-ci-centos-9-singlenode-external-compute-via-ee job defined in\nhttps://review.opendev.org/c/openstack/tripleo-ci/+/843836.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d7eba58f0fd3db0db1fea727b7054206c380ffe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d7eba58f0fd3db0db1fea727b7054206c380ffe4"}]},"branch":"refs/heads/master"},"e9d70fd8b5598c7875b70899ab8df90b877dcbe5":{"kind":"REWORK","_number":48,"created":"2022-09-12 07:57:20.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/48"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-12 07:57:06.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nThis image will be consumed in\ntripleo-ci-centos-9-singlenode-external-compute-via-ee job defined in\nhttps://review.opendev.org/c/openstack/tripleo-ci/+/843836.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e9d70fd8b5598c7875b70899ab8df90b877dcbe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e9d70fd8b5598c7875b70899ab8df90b877dcbe5"}]},"branch":"refs/heads/master"},"cf7774e27e10da625815b3c314c2a1eb03b7049f":{"kind":"REWORK","_number":49,"created":"2022-09-12 09:24:39.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/49"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-12 09:24:24.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nThis image will be consumed in\ntripleo-ci-centos-9-singlenode-external-compute-via-ee job defined in\nhttps://review.opendev.org/c/openstack/tripleo-ci/+/843836.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cf7774e27e10da625815b3c314c2a1eb03b7049f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cf7774e27e10da625815b3c314c2a1eb03b7049f"}]},"branch":"refs/heads/master"},"bdf5f9a6bd88910168d016116cdc04cb453632b2":{"kind":"REWORK","_number":50,"created":"2022-09-12 10:48:03.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/50"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-12 10:47:50.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nThis image will be consumed in\ntripleo-ci-centos-9-singlenode-external-compute-via-ee job defined in\nhttps://review.opendev.org/c/openstack/tripleo-ci/+/843836.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bdf5f9a6bd88910168d016116cdc04cb453632b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bdf5f9a6bd88910168d016116cdc04cb453632b2"}]},"branch":"refs/heads/master"},"95ea4376fa95cfa7e0a01159913c18502e3054c4":{"kind":"REWORK","_number":51,"created":"2022-09-12 13:02:57.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/51"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-12 13:02:43.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nThis image will be consumed in\ntripleo-ci-centos-9-singlenode-external-compute-via-ee job defined in\nhttps://review.opendev.org/c/openstack/tripleo-ci/+/843836.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/95ea4376fa95cfa7e0a01159913c18502e3054c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/95ea4376fa95cfa7e0a01159913c18502e3054c4"}]},"branch":"refs/heads/master"},"a8510ac7165eb3af9bd633edb8f3da4a225ffd18":{"kind":"REWORK","_number":52,"created":"2022-09-12 14:31:36.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/52"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-12 14:31:22.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains the tripleo-ansible roles and collections\ncontents via RPM.\ntripleo-ansible ansible contents are symlinked back with\nthe /runner directory.\n\nBy using ansible-runner, one can easily run those playbooks\nwith in the container.\n\nThis image will be consumed in\ntripleo-ci-centos-9-singlenode-external-compute-via-ee job defined in\nhttps://review.opendev.org/c/openstack/tripleo-ci/+/843836.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a8510ac7165eb3af9bd633edb8f3da4a225ffd18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a8510ac7165eb3af9bd633edb8f3da4a225ffd18"}]},"branch":"refs/heads/master"},"1335961dc6ad89afda8f09daabfdc7ff8df64dd0":{"kind":"REWORK","_number":53,"created":"2022-09-14 06:33:30.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/53"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-14 05:53:25.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains:\n- the tripleo-ansible roles and collections contents via RPM.\n- tripleo-ansible ansible contents are symlinked back with\n  the /runner directory.\n- ansible-runner settings included in /runner/env\n- custom entrypoint for tripleo specific usecase, which later\n  on appends the ansible-runner entrypoint\n\nThis image will be consumed in\ntripleo-ci-centos-9-singlenode-external-compute-via-ee job defined in\nhttps://review.opendev.org/c/openstack/tripleo-ci/+/843836.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1335961dc6ad89afda8f09daabfdc7ff8df64dd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1335961dc6ad89afda8f09daabfdc7ff8df64dd0"}]},"branch":"refs/heads/master"},"54a6bac5d557aa84c21986e96db0e0911792fe1b":{"kind":"REWORK","_number":54,"created":"2022-09-14 10:15:09.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/74/776674/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/776674/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/776674/54"}}},"commit":{"parents":[{"commit":"615cf9ff6ee9cc5ff10ec2b6824862112ee97c03","subject":"Update ovmf-amdsev from ovmf-cc to boot UEFI VMs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/615cf9ff6ee9cc5ff10ec2b6824862112ee97c03"}]}],"author":{"name":"Kevin Carter","email":"kecarter@redhat.com","date":"2021-02-19 16:58:01.000000000","tz":-360},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-09-14 10:14:56.000000000","tz":330},"subject":"Add an Ansible execution environment to our Build","message":"Add an Ansible execution environment to our Build\n\nThis change adds an ansible execution environment to our build tools,\nwhich will allow us to run with execution environments via tripleoclient\nin the future. This build simply pulls in the base ansible-runner\nimage, and provides an interface to download/install roles and\ncollections as needed.\n\nIt also contains:\n- the tripleo-ansible roles and collections contents via RPM.\n- tripleo-ansible ansible contents are symlinked back with\n  the /runner directory.\n- ansible-runner settings included in /runner/env\n- custom entrypoint for tripleo specific usecase, which later\n  on appends the ansible-runner entrypoint\n\nThis image will be consumed in\ntripleo-ci-centos-9-singlenode-external-compute-via-ee job defined in\nhttps://review.opendev.org/c/openstack/tripleo-ci/+/843836.\n\nDepends-On: https://review.opendev.org/c/openstack/tripleo-quickstart-extras/+/852720\n\nChange-Id: I56cabb9aa8483c5039b5920db96a00509dc730cc\nSigned-off-by: Kevin Carter \u003ckecarter@redhat.com\u003e\nCo-authored-by: Chandan Kumar \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/54a6bac5d557aa84c21986e96db0e0911792fe1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/54a6bac5d557aa84c21986e96db0e0911792fe1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}}]}],"submit_requirements":[]}
