)]}'
{"id":"openstack%2Ftripleo-common~778262","triplet_id":"openstack%2Ftripleo-common~master~Id47d6ff94edfb607eb3611aa4cd306c3de450fb4","project":"openstack/tripleo-common","branch":"master","topic":"ephemeral-heat","hashtags":[],"change_id":"Id47d6ff94edfb607eb3611aa4cd306c3de450fb4","subject":"Don\u0027t require heat client for generate_passwords","status":"MERGED","created":"2021-03-02 20:36:55.000000000","updated":"2021-03-17 16:35:04.000000000","submitted":"2021-03-17 16:28:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778262-ephemeral-heat","meta_rev_id":"b7282c4d54aace0933aabaef4265b67a4938394f","_number":778262,"virtual_id_number":778262,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-17 16:27:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-02 23:16:29.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-04 09:55:43.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":1,"date":"2021-03-17 12:33:16.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 23:16:29.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 23:36:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-04 09:55:43.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-03-17 12:33:16.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"}],"messages":[{"id":"9252cde30328e7feeb10612c2fe6e41c9a28883b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-03-02 20:36:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b025dc239de80c1e06025d8b4cf7f0abc46a70c9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-02 23:16:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8b4c04be55734f6687854b886fb3c7ed : SUCCESS in 2h 37m 38s","accounts_in_message":[],"_revision_number":1},{"id":"1c4f3ad6a4bf65be0d0da6a7522b04c8cfc8364d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 23:36:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/362140f2958a45ab8b25afe6770e7804 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d55702e748f941e88c40b611f237e6a5 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad48bb22eabf40d997e4e27767e50590 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47012d5cd5974d30bd45a4dab2c56c13 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50725a11fffc4393ad56bb63c48d6d01 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c3ac2c901f4b11be5dbbacde67a531 : SUCCESS in 4m 15s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/823a3ecedc474f21bbf11cd27f26e197 : SUCCESS in 2h 50m 05s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5da9445f5c844b0e82e1de7fb1ea004b : SUCCESS in 1h 42m 11s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/904a63ce65a34ad88898f466c9c46bf8 : SUCCESS in 1h 04m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/61eb4d4131fa4499bc948f6a80f97bd0 : SUCCESS in 1h 20m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/300794d3ab2749c9bd87a5991ba218bc : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/77683099cbac4ac7b7f806ffae65a9ed : SUCCESS in 4m 16s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/8037f9c4dcb74382b4f75f8bf03469ee : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"efbde11576803ddfef724bbf2625c633d2f0a243","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-03-04 09:55:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e8e0ec1a5d761d1b884eabc396b658a8d266d69d","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-03-17 12:33:16.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0193eb31958758e4aedb9f4a13671dc8271d058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 12:33:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"44bd158d3e9240603447925598569d2a80198f64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 16:27:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aacb2e0aa4e4f86a13bb0c1bfd21283 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3709f779550c4619bb500a3d19e53e7d : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e64d200e9f64e41950a241413a090be : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4839a7e014433e9a7c959efd3d1316 : SUCCESS in 5m 08s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/43c5c872a8ea4cdb8822011da5edbb20 : SUCCESS in 3h 49m 56s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1fbf3c3bcce14bc4a3b811be313d62ef : SUCCESS in 2h 25m 24s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8985d4e08da24d51a45ee4f3c453d5e6 : SUCCESS in 1h 14m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d40a7b71a047426d99f2a5169e32d791 : SUCCESS in 1h 35m 34s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/10ae5a1730a24ff698d0bdf91a120513 : SUCCESS in 1h 57m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/047ac60a61cd4ef489358398ca680384 : SUCCESS in 5m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6d7c1bbe07a14fd8b7bf625ce1b0bdf0 : SUCCESS in 5m 00s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/74b4308aa4fd4820aa9f55a80d748134 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d9aa9bff2d93b5fa3d65e4ae883dcf19e50a3890","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 16:28:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b7282c4d54aace0933aabaef4265b67a4938394f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 16:35:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22a8e43e70e042e7bdeb14f92b42bc47 : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd751b6ff875f102de32c18aa5072642d3ffd9dc","revisions":{"cd751b6ff875f102de32c18aa5072642d3ffd9dc":{"kind":"REWORK","_number":1,"created":"2021-03-02 20:36:55.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/62/778262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/62/778262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/62/778262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/62/778262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/62/778262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/62/778262/1"}}},"commit":{"parents":[{"commit":"636fca4f29a18a9b9cd43a0d8334267605102d4e","subject":"Merge \"Remove layout override and wire up buildimage template instead\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/636fca4f29a18a9b9cd43a0d8334267605102d4e"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-03-02 20:34:42.000000000","tz":-300},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-03-02 20:36:43.000000000","tz":-300},"subject":"Don\u0027t require heat client for generate_passwords","message":"Don\u0027t require heat client for generate_passwords\n\nTo support using ephemeral heat, don\u0027t require a heat client in the\ngenerate_passwords function. Reading the existing values from the\nenvironment is now not required when using ephemeral heat.\n\nSigned-off-by: James Slagle \u003cjslagle@redhat.com\u003e\nChange-Id: Id47d6ff94edfb607eb3611aa4cd306c3de450fb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cd751b6ff875f102de32c18aa5072642d3ffd9dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cd751b6ff875f102de32c18aa5072642d3ffd9dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
