)]}'
{"id":"openstack%2Ftripleo-common~779171","triplet_id":"openstack%2Ftripleo-common~stable%2Fussuri~I40813ca2ffd6bf2b2b808e72607822f1f2fdbe7f","project":"openstack/tripleo-common","branch":"stable/ussuri","topic":"c8stream-test","hashtags":[],"change_id":"I40813ca2ffd6bf2b2b808e72607822f1f2fdbe7f","subject":"DNM centos8 stream tests - ussuri","status":"ABANDONED","created":"2021-03-08 08:05:55.000000000","updated":"2021-04-27 10:19:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6209d7c02e78167041d1d611642f9b94b641d177","_number":779171,"virtual_id_number":779171,"owner":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-08 12:07:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-08 10:40:06.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 10:40:06.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 12:07:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d397d13e6ed19ea49cfef52d48d9f0390ab562d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-03-08 08:05:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e47ab11114344da7bbdc57ae60ef4a0a52a087f1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-03-08 08:06:19.000000000","message":"Topic ussuri8st removed","accounts_in_message":[],"_revision_number":1},{"id":"8c93d5585a1c7db7523028e282ff96ab1be2bc12","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-03-08 08:06:34.000000000","message":"Topic set to c8stream-test","accounts_in_message":[],"_revision_number":1},{"id":"58d8649313021db5f81d3c429122cb8abb9df5c2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-08 10:40:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a829ad7ef80040c18f563d53b3240dd2 : SUCCESS in 2h 29m 06s","accounts_in_message":[],"_revision_number":1},{"id":"aaa51e8060c97f223e2f0576e11749f909c469ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 12:07:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dab3e051adcc4881a4dad2c2d75851bf : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/611b443135804c8a9aaaef39fec5bb03 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42f7dadab707482499fd870d00242cb7 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ed6a3d72b3ba476ea9d2892c3348f568 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4106e22339a4cdb8fc34a1847a3929e : SUCCESS in 4m 38s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/81f7a4c4612b490ab7f47f0ab22143fe : SUCCESS in 3h 55m 35s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e8546d58542b4acd8cec9e9d97c7a6ce : SUCCESS in 2h 07m 14s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d2732090fec44005b37bc0a3a6aa8a6f : SUCCESS in 1h 22m 40s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/cef2b23484614234b3a43417d4c7e8dc : SUCCESS in 1h 02m 03s (non-voting)\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/d4a0f3d5161142ec9a1f858cf9c5c716 : SUCCESS in 24m 40s (non-voting)\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/321575c77c0a4be0b69c9d3853ba253a : SUCCESS in 47m 12s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/941c6594a223464e9e49d24a2a4bc9d2 : SUCCESS in 1h 29m 16s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/4dc5da7c5ac144beb89ddfc15ae6172a : SUCCESS in 3h 53m 51s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/373f4efa0fdb49d385f85295e0f6c8e6 : SUCCESS in 2h 21m 23s\n- tripleo-ci-centos-8-undercloud-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/b479e3a3d4be4d8c87cd39be0e6d826e : SUCCESS in 2h 25m 12s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/1854ba0c0f3f4f008e8a259f0f41a1a6 : SUCCESS in 2h 21m 35s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/6f1f2f35503d4afe85ae2a4896e8175f : SUCCESS in 2h 06m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/26f5ef7629684db9a9346c4fef3d8617 : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6209d7c02e78167041d1d611642f9b94b641d177","tag":"autogenerated:gerrit:abandon","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-04-27 10:19:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a10e9ebc9f83b70b89e73b54a98009719b64e7ca","revisions":{"a10e9ebc9f83b70b89e73b54a98009719b64e7ca":{"kind":"REWORK","_number":1,"created":"2021-03-08 08:05:55.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/71/779171/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/71/779171/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/779171/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/779171/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/71/779171/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/71/779171/1"}}},"commit":{"parents":[{"commit":"573d6b5de6be6043b650f1513415936a2151860a","subject":"Merge \"Remove layout override and wire up buildimage template instead\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/573d6b5de6be6043b650f1513415936a2151860a"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2021-03-08 08:05:39.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2021-03-08 08:05:39.000000000","tz":120},"subject":"DNM centos8 stream tests - ussuri","message":"DNM centos8 stream tests - ussuri\n\nChange-Id: I40813ca2ffd6bf2b2b808e72607822f1f2fdbe7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a10e9ebc9f83b70b89e73b54a98009719b64e7ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a10e9ebc9f83b70b89e73b54a98009719b64e7ca"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
