)]}'
{"id":"openstack%2Ftripleo-common~779201","triplet_id":"openstack%2Ftripleo-common~stable%2Fvictoria~I95efa61361882875505e5e359dece23a5a67bfc9","project":"openstack/tripleo-common","branch":"stable/victoria","topic":"tripleo-ci-reduce","hashtags":[],"change_id":"I95efa61361882875505e5e359dece23a5a67bfc9","subject":"Removes unneeded zuul layout overrides - templates for consistency","status":"MERGED","created":"2021-03-08 10:38:42.000000000","updated":"2021-03-11 16:05:39.000000000","submitted":"2021-03-11 15:54:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779201-tripleo-ci-reduce","meta_rev_id":"62e60e015f72ef4e010af5ba70d5a1139d46f304","_number":779201,"virtual_id_number":779201,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-11 15:54:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-03-08 13:16:26.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":2,"date":"2021-03-09 09:20:46.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2021-03-11 11:51:06.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2021-03-11 11:51:06.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 13:16:26.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 15:18:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-09 09:20:46.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-03-11 04:36:18.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"}],"messages":[{"id":"26f766ce402f46168f4a1671047a4b022fbf9a1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-08 10:38:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d29d9210f9de4a3f0ef29701e5a2f4b4ff649597","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-03-08 10:57:51.000000000","message":"Topic set to tripleo-ci-reduce","accounts_in_message":[],"_revision_number":1},{"id":"3bdff156cd614202efc5a26e9c25374c7b14c575","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-08 13:16:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6ae57601c89d4599907946fdde0416f0 : FAILURE in 2h 36m 21s","accounts_in_message":[],"_revision_number":1},{"id":"aaa847f70b0baf1aceb7d8fab50edd654930aeb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 15:18:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adc1ecce34a149d4b923efdfd71afd6a : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cba8c14892849a7beecdb84879f97f2 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ba223d5259f45e0acc7623e24dfb0db : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f49b271dd1ad40aa98edb85cd5f1c529 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/711211f02eec48ada3900c6621040562 : SUCCESS in 5m 27s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/010cb352ace146588eedad7f19b1565c : SUCCESS in 3h 37m 30s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c0937e5edffd4fc7b6fbf845f14283bd : SUCCESS in 1h 53m 20s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/95dfceb1a8ec498ca97561c77aeaa67e : SUCCESS in 1h 09m 41s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d8044bf8fcc2492390fb39a67d4a8b22 : SUCCESS in 1h 41m 42s\n- tripleo-ci-centos-8-scenario002-standalone https://zuul.opendev.org/t/openstack/build/ed263e21f23b408ba5a54e2d4fa3614a : SUCCESS in 1h 28m 31s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/8c3b102e629f4b19bab89f10447851dd : SUCCESS in 1h 11m 46s\n- tripleo-ci-centos-8-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ddedff0f870e4aa4a11563d0bdc562f1 : SUCCESS in 1h 39m 04s\n- tripleo-ci-centos-8-scenario007-standalone https://zuul.opendev.org/t/openstack/build/d6e3e9e77dc14a5c8bf08a17b9efddd9 : SUCCESS in 1h 04m 44s\n- tripleo-ci-centos-8-scenario010-standalone https://zuul.opendev.org/t/openstack/build/17287961bf7c4e6cb3d54e3992e3bee8 : SUCCESS in 1h 40m 12s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/31e478a413524d6d8f63e161da7dd7f5 : SUCCESS in 1h 29m 07s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/dc94078f3e634ce1a5f8328105893a51 : SUCCESS in 4h 33m 23s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/b4092cb731e24e1ab2899410dfb4486c : SUCCESS in 2h 26m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5258ca0d9cdd46b0a9c5873c6348a162 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8ec03c3d4ad46849fa99df10935c224 : SUCCESS in 5m 11s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/0525bf940e844b92ba43e04aff755ec7 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2970ee3983eca948fa6dcbe375e6c585e31976bb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-03-09 09:20:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10030e7558f99f5825eb13c33ddfba0955ab2d89","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-03-11 11:51:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a00c6145b0d6c7a89d3bcfebb3b0f7c4c0160f68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 11:51:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3650dcc7e6f8fd4cb45f2cd605ff81dfd9d5804a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 15:54:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/681d439308a74d9d92d455008a01661c : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/193f091851f34b56adf59cb1f3dbbb6f : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39457810704b4624b626661bdd66aeeb : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ed5648c5d424e8e822fa39b9898b41e : SUCCESS in 7m 15s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c0deeff0fff1496ab971a1f0ee2055ed : SUCCESS in 4h 00m 43s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b78b87c31bd04a668c3e78c6cce39ea5 : SUCCESS in 1h 32m 13s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2ea0322a6fb54bf0ad335ddc2a57f570 : SUCCESS in 3h 58m 43s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/72e732aa46e343a9aa0b0146dff915d5 : SUCCESS in 2h 37m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a50733a262dd44f09c903e8a9f97f598 : SUCCESS in 5m 22s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/bc1f6a46f0ba4b64a0d60c5a4ff1411c : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"cef37b4db134d92c024466d14476178535f50d74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 15:54:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"62e60e015f72ef4e010af5ba70d5a1139d46f304","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 16:05:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b7b321ce8d49d198f10508967ec22e : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cf2715c91cbe696a64f80c8f7994f7d6d14d7e6e","revisions":{"cf2715c91cbe696a64f80c8f7994f7d6d14d7e6e":{"kind":"REWORK","_number":1,"created":"2021-03-08 10:38:42.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/01/779201/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/01/779201/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/01/779201/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/01/779201/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/01/779201/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/01/779201/1"}}},"commit":{"parents":[{"commit":"366b10b0feefed98691812b7cf07bc4c21a9283f","subject":"Merge \"Remove layout override and wire up buildimage template instead\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/366b10b0feefed98691812b7cf07bc4c21a9283f"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-03-05 12:26:29.000000000","tz":120},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2021-03-08 10:36:01.000000000","tz":120},"subject":"Removes unneeded zuul layout overrides - templates for consistency","message":"Removes unneeded zuul layout overrides - templates for consistency\n\nAs part of [1] this removes the unneeded zuul layout overrides.\nAll the jobs removed here are already included with the relevant\ntemplates. The files: matches that were used for the scenario\nstandalones are moved to the template layout via the depends-on.\n\nThe build-containers-ubi8 job is not needed as it is exercised\nby the content providers and we have removed this job from other\ntripleo layouts.\n\n[1] https://review.opendev.org/q/topic:tripleo-ci-reduce\nDepends-On: I633bc01bc1a8cd8d1216e503eaa9a43aef9a7fa0\n\nChange-Id: I95efa61361882875505e5e359dece23a5a67bfc9\n(cherry picked from commit 22e5366de13a342ca6f65173ce069513fa80e1f5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cf2715c91cbe696a64f80c8f7994f7d6d14d7e6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/cf2715c91cbe696a64f80c8f7994f7d6d14d7e6e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
