)]}'
{"id":"openstack%2Ftripleo-common~780889","triplet_id":"openstack%2Ftripleo-common~stable%2Fussuri~Ica1b5c0a165cc06fac668513114eef2b4ba73f56","project":"openstack/tripleo-common","branch":"stable/ussuri","topic":"mistral-ansible-log-stable/ussuri","hashtags":[],"change_id":"Ica1b5c0a165cc06fac668513114eef2b4ba73f56","subject":"[Victoria and bellow] Ensure rights on the ansible.log file","status":"MERGED","created":"2021-03-17 08:59:37.000000000","updated":"2021-03-18 22:51:51.000000000","submitted":"2021-03-18 22:50:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780889-mistral-ansible-log-stable/ussuri","cherry_pick_of_change":779438,"cherry_pick_of_patch_set":1,"meta_rev_id":"7eec39072b7bef60e850308d114be2368730c1b5","_number":780889,"virtual_id_number":780889,"owner":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 22:49:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-17 11:46:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-03-18 15:45:52.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2021-03-18 15:45:52.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-17 11:46:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-17 13:03:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 13:25:32.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2021-03-18 13:25:32.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-03-18 15:45:52.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"1b3b4d6c372c89ce06dec3e45dc2feea1f695fa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-03-17 08:59:37.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"1c13565946ef518347dde4de6fdf5d0400644e7c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-17 11:46:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/365834dff66e4deb9302bac2ab7e3a1b : SUCCESS in 2h 43m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f2eef04120702927d601e5abadd9f75f6bdab6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 13:03:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e849f29befe42d8924f5d45391c4166 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d1cfe0df47470c806ed31aef79eea8 : SUCCESS in 9m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b25a4af993664d64ae34286737abab70 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b4e21932bb144062adc1d45e3a323eca : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a851f47e9142e8b7624d01beb7ec7e : SUCCESS in 4m 51s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c2e56aa2f46e444ca82c03ce8b79cb7a : SUCCESS in 3h 50m 24s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1dff551689a54aec8af7e77700537bd7 : SUCCESS in 1h 52m 58s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b32c1c79923741b5b050cf812f169d55 : SUCCESS in 1h 07m 27s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/bf9c0e2306894d3aa40e15ab7478d9b0 : SUCCESS in 1h 44m 14s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/65a5e3eb280f4102ae72250727da758f : SUCCESS in 1h 25m 33s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/71180526a48d42a2a41af9f5b267a894 : SUCCESS in 3h 57m 20s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/d8a88600bd704341b83bfa7efc2203e6 : SUCCESS in 2h 34m 25s\n- tripleo-ci-centos-8-undercloud-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/937b645455e8404f8aebfcfb38cafc67 : SUCCESS in 2h 24m 54s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/4109d35528064e09b916f9ca723e62bd : SUCCESS in 2h 09m 56s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/864907d2264746b6868b575b21344ac8 : SUCCESS in 1h 51m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/db4c200013864abdab6e56b93e881106 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ad81c7042604bd9b6978d95f12134cfff328c47f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-18 15:45:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c587976ec4e06b8b9717fff3f691e2bbc20beea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 15:47:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8e972b44ac3c86e07cb1eeba3b3e113c68568794","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:49:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13f69ddda35a4cc38cd1f9b62b40a01e : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e4497dc530546da936fd8f90840dc6c : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17c4fa70dd664a0e844c4c066b918981 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b85d78c33a44bd8a9f84fe8960711e3 : SUCCESS in 4m 53s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b4eea2143de04c5299cb9835a9340205 : SUCCESS in 4h 42m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a000f4a0a6914a9b8e414de131af3154 : SUCCESS in 1h 44m 27s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/40a5a4dfb57d421991d2dbe84ddc0d17 : SUCCESS in 1h 07m 39s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/73468614e8004ea69dbc5851c1b8a708 : SUCCESS in 1h 34m 02s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d19e0d9387bd41c2aa6eabedf778a256 : SUCCESS in 1h 22m 26s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/9f2e5dc98bcb473e8561093764ec5eec : SUCCESS in 3h 51m 29s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/de2a210fa8ff45f9861c4b7efb09fb80 : SUCCESS in 2h 11m 24s\n- tripleo-ci-centos-8-undercloud-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/c68404d770954a40a709b478c0f0812d : SUCCESS in 2h 00m 30s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/f3099e5b808942fca881a8a3e63f13f7 : SUCCESS in 2h 03m 30s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/5b10cef069c3452bb51fbd2364b0e82e : SUCCESS in 1h 59m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dd5f24332ce541ebaee7d1bd37ba2255 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"02374eab319c959a69979e0895834544671997de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:50:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7eec39072b7bef60e850308d114be2368730c1b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:51:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/762572d3e5c14e118e120fe96979b284 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdd5c0c09161c69975c0b62a02cbf256295c0c48","revisions":{"fdd5c0c09161c69975c0b62a02cbf256295c0c48":{"kind":"REWORK","_number":1,"created":"2021-03-17 08:59:37.000000000","uploader":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"ref":"refs/changes/89/780889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/89/780889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/89/780889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/89/780889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/89/780889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/89/780889/1"}}},"commit":{"parents":[{"commit":"118af263cf69b633f456c9b1292e21cef00726e4","subject":"[stable-only] Fix legacy password handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/118af263cf69b633f456c9b1292e21cef00726e4"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-03-08 13:35:55.000000000","tz":60},"committer":{"name":"Cedric Jeanneret","email":"cjeanner@redhat.com","date":"2021-03-17 08:59:37.000000000","tz":0},"subject":"[Victoria and bellow] Ensure rights on the ansible.log file","message":"[Victoria and bellow] Ensure rights on the ansible.log file\n\nWhen mistral starts the Ansible action, a log file is created runtime.\nBut nothing takes care of its access right, leading to potential data\nleaks to unprivileged users (default mode is 0644).\n\nThis patch creates the logfile beforehands, and sets the needed rights\non it.\n\nChange-Id: Ica1b5c0a165cc06fac668513114eef2b4ba73f56\nCloses-Bug: #1918138\n(cherry picked from commit d485407159ea6cb2c7abf6d5788d85147f433996)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fdd5c0c09161c69975c0b62a02cbf256295c0c48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fdd5c0c09161c69975c0b62a02cbf256295c0c48"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
