)]}'
{"id":"openstack%2Ftripleo-common~782385","triplet_id":"openstack%2Ftripleo-common~master~If5ec5d4fa0d4d835561b5963d244e5172fe82cc8","project":"openstack/tripleo-common","branch":"master","topic":"ceph_dashboard_containers","hashtags":[],"change_id":"If5ec5d4fa0d4d835561b5963d244e5172fe82cc8","subject":"Use quay.ceph.io to pull the Ceph Dashboard related images","status":"MERGED","created":"2021-03-23 09:37:37.000000000","updated":"2021-03-25 01:57:45.000000000","submitted":"2021-03-25 01:52:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"782385-ceph_dashboard_containers","meta_rev_id":"f89db613d24ac6ffddbce3686a6c65ceeff7bf90","_number":782385,"virtual_id_number":782385,"owner":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":15205,"name":"Guillaume Abrioux","email":"gabrioux@redhat.com"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 01:52:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":1,"date":"2021-03-24 13:28:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":15205,"name":"Guillaume Abrioux","email":"gabrioux@redhat.com"},{"value":2,"date":"2021-03-24 10:06:16.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-24 10:39:29.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":15205,"name":"Guillaume Abrioux","email":"gabrioux@redhat.com"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-24 10:39:29.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":22318,"name":"Chandan Kumar","email":"chandan.kumar@imaginea.com"}],"reviewers":{"REVIEWER":[{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":15205,"name":"Guillaume Abrioux","email":"gabrioux@redhat.com"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22318,"name":"Chandan Kumar","email":"chandan.kumar@imaginea.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-23 09:39:46.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-03-23 09:39:46.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":15205,"name":"Guillaume Abrioux","email":"gabrioux@redhat.com"},"state":"REVIEWER"},{"updated":"2021-03-23 09:39:46.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-03-23 09:39:46.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":22318,"name":"Chandan Kumar","email":"chandan.kumar@imaginea.com"},"state":"REVIEWER"},{"updated":"2021-03-23 09:39:46.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2021-03-23 09:39:46.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2021-03-23 09:39:46.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2021-03-23 09:39:46.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-23 09:39:46.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-23 12:11:33.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"CC"},{"updated":"2021-03-23 12:43:19.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"CC"},{"updated":"2021-03-23 12:43:19.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"CC"},{"updated":"2021-03-23 12:43:19.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"CC"},{"updated":"2021-03-23 12:43:19.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"CC"},{"updated":"2021-03-24 10:06:16.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-03-24 10:30:07.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"}],"messages":[{"id":"c998c77797b2416f60ed48d94780fea0a4b31aa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-03-23 09:37:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"07e63fe488b4a1648400a6bceb3d80bde402a033","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-03-23 09:37:53.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d809f1c1fd94ffd23ab5cd89aa6cd858a216d852","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-03-23 09:39:46.000000000","message":"Patch Set 1:\n\nWIP since we\u0027re still mirroring the ceph dashboard containers on quay.ceph.io.","accounts_in_message":[],"_revision_number":1},{"id":"2bd858fdb7d2bc15f3cf3152d7d787f510f5e98f","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-03-23 11:19:39.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e WIP since we\u0027re still mirroring the ceph dashboard containers on quay.ceph.io.\n\nTesting it via https://review.opendev.org/c/openstack/tripleo-heat-templates/+/778915 (scenario004)","accounts_in_message":[],"_revision_number":1},{"id":"75d23f08c69a982212ddfd286b25d19cad00b56b","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-03-23 12:11:33.000000000","message":"Patch Set 1:\n\nhttps://9c09d6d31505666aa8c8-1c29e72385050ff895c097b41af27e3b.ssl.cf2.rackcdn.com/778915/7/check/tripleo-ci-centos-8-content-provider/9e3240a/logs/undercloud/home/zuul/containers-prepare-parameters.yaml contents corresponds to this change, although unsure about other ceph*-like namespaces that remain pointing quay.io","accounts_in_message":[],"_revision_number":1},{"id":"fc1683367b6e8ad2a93a1a90c2e0f60862f5ad56","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-03-23 12:16:59.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e https://9c09d6d31505666aa8c8-1c29e72385050ff895c097b41af27e3b.ssl.cf2.rackcdn.com/778915/7/check/tripleo-ci-centos-8-content-provider/9e3240a/logs/undercloud/home/zuul/containers-prepare-parameters.yaml contents corresponds to this change, although unsure about other ceph*-like namespaces that remain pointing quay.io\n\nYou should check PS 8 which has a depends-on this review.","accounts_in_message":[],"_revision_number":1},{"id":"abe0ba4aa0cc7f157fac18724e015a483e39c785","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-03-23 12:18:36.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e https://9c09d6d31505666aa8c8-1c29e72385050ff895c097b41af27e3b.ssl.cf2.rackcdn.com/778915/7/check/tripleo-ci-centos-8-content-provider/9e3240a/logs/undercloud/home/zuul/containers-prepare-parameters.yaml contents corresponds to this change, although unsure about other ceph*-like namespaces that remain pointing quay.io\n\u003e \n\u003e You should check PS 8 which has a depends-on this review.\n\n.. and it\u0027s still on CI","accounts_in_message":[],"_revision_number":1},{"id":"89d09ac3e000dcc7488c2441c3376ea43a56497a","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2021-03-23 12:19:40.000000000","message":"Patch Set 1: Code-Review+1\n\nSeems like a move in the right direction. Let\u0027s see how it does in CI.","accounts_in_message":[],"_revision_number":1},{"id":"b98d3daf94afb8401104a88e7294deef1bf9cf83","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-23 12:19:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/99e27071765d4f6f8ff5a92fe8cfc028 : SUCCESS in 2h 41m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a3c992efdd6108b552f19723ef0de9685591a5db","author":{"_account_id":15205,"name":"Guillaume Abrioux","email":"gabrioux@redhat.com"},"date":"2021-03-23 12:55:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0fb5241979f5a1071a05c613a79f6e7b05ef52f4","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-03-23 12:56:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3b30db7df6bb04ae29dc0c4b887c717c32b4907","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 13:12:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d3e6bbad94a4cfca47cb27ec79d37d3 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e45e8bdaf094e0ba8257226a7618bff : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb196391822f42adb9e1c5f3e7ecad67 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f274028991b24d2399755f97471356f0 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89bf3f9cddba4a91893bc8dd49098cf5 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/851076a291fe4eb485f525fa74e18fa3 : SUCCESS in 5m 39s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3bc9c773a69548d69715da40eab78245 : SUCCESS in 3h 09m 48s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b295fa7690af4eabae08fa8affad638b : SUCCESS in 2h 01m 07s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d60675096fe54d5ca9f632933a760af7 : SUCCESS in 1h 23m 46s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/693f82390b4f42c8b70fa78eb120387e : SUCCESS in 1h 31m 13s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/17dc46f3f26b4ec9afd4b54bb2a5f768 : SUCCESS in 1h 48m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7a172fffad3e43d288d44313d79804e6 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6e1d325baf84229b444027447c5cbfb : SUCCESS in 5m 30s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/da5f92a7b62f4479a3b545468c56cc7e : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9b0bfd2ae76a6b19bdc6f212c7b570a7a18ccdd7","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-03-23 13:31:11.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nAs per [1] sounds like it worked as expected and scenario004 (the one containing the ceph dashboard bits) worked properly [2].\nOnce the CI is stable, the review(s) [3] are merged and we the bug [4] is solved, I propose to merge and backport this change.\n\n[1] https://c147281a6742ad573643-b429f7b1939d4fc1f35718a4403a0d7a.ssl.cf2.rackcdn.com/778915/8/check/tripleo-ci-centos-8-content-provider/b8f21f2/logs/undercloud/home/zuul/containers-prepare-parameters.yaml\n\n[2] https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_ead/778915/8/check/tripleo-ci-centos-8-scenario004-standalone/ead874e/logs/undercloud/home/zuul/standalone-ansible-433s81x9/ceph-ansible/ceph_ansible_command.log\n\n[3] https://review.opendev.org/q/e0b0d859efb179c6868020b3fc10c2155ed96820\n\n[4] https://bugs.launchpad.net/tripleo/+bug/1920873","accounts_in_message":[],"_revision_number":1},{"id":"95e2803bd746cec4424ac5dea2617986703d931a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-03-23 14:34:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"57dd8b0a11bf63fd88af5d4daf20e72098c1aeca","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-23 17:38:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/793974764949404faee1028be6488dce : SUCCESS in 3h 01m 39s","accounts_in_message":[],"_revision_number":2},{"id":"b18529b4ad8f862e298c989503d27d38973cfa9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 18:47:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e173752f245c4178ba7b644d15768e84 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53fb2eb5a96f444883201ad19ddd824e : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/393783f24b3148b6949e78ab906e72f3 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf19ccb3b9b8427e87d49814db080553 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cba3b1683d74ddea4579dbf0a87f974 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa896dd7aff4bcab4f6a78e288c063b : SUCCESS in 5m 20s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/497f1a25a1ae4feeb273dad169f95b77 : SUCCESS in 3h 59m 57s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2eb3ef1c3a5f422ab53d52590be04def : SUCCESS in 1h 55m 23s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6c69141960da491e8815206b2ec2505b : SUCCESS in 1h 11m 05s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/19412c1b0468408f9a15dd3e826c27f8 : SUCCESS in 1h 31m 16s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ff12ffb9d89141e2922a99f89c7ed9dd : SUCCESS in 2h 33m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/76a542ca7452489fabe88bf07524ffa0 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5b77f38aa8d48a6a0e3a359d459b5c7 : SUCCESS in 6m 25s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/0f2041fe934c4a9bbc36ca8557bab8ff : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"d2a210277c2a3bc02591709d802fab064451043f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-03-24 09:57:20.000000000","message":"Patch Set 2:\n\n(4 comments)\n\nNice job fmount, we can also add\n\nCloses-Bug: #1920873 and remove WIP from patch","accounts_in_message":[],"_revision_number":2},{"id":"ac7f24fbd6afd732c58d8addf81f10752e275d76","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-03-24 10:03:47.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"afa24e04ac5aba0d443b6f9c56d19ffbffda48e8","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2021-03-24 10:05:00.000000000","message":"Patch Set 3:\n\nReady to review/merge this one","accounts_in_message":[],"_revision_number":3},{"id":"fd1cb67db0f5896ccc1f717580802066c0af781e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-03-24 10:06:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9c95380b0927b4ff5e9c7cdffd37cfa013d44dd2","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2021-03-24 10:39:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"44a725c6eb68045959a7d6c4ea8a2d1e75d4573c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-03-24 13:28:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/cce10b4ee15a424fa84614594da4ef8c : SUCCESS in 3h 16m 01s","accounts_in_message":[],"_revision_number":3},{"id":"4cd214d9f366d7d22d8816a86a91bbbac865b40e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 13:49:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20323f341ff340748cbe0ede68279f38 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b071c60a074042eebf58b9cdb15e6f54 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c59706e95104fdfa2ee8056b39b8449 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5400077a123e4139aa4a39407bacb440 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5012c0f1f14439f890045e5a5fc87bf : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b2d2275abe146428ee2b8dd02f23fbf : SUCCESS in 6m 14s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d48233564d7540ebbd8ca51d58ea6083 : SUCCESS in 3h 28m 26s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/596ccf1a642a4b31b6043427fa57d845 : SUCCESS in 1h 58m 13s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/deb5ad57720c4f55a517d418b13a8632 : SUCCESS in 1h 26m 16s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/633fcf943c5a442b9d27807a8a4c0765 : SUCCESS in 1h 34m 02s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5cd2132b45ca4b74b73b081e84561d8f : SUCCESS in 1h 53m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6383cbad4a104934996dac80706cb6d4 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94d91240857c494281a0a5cbbd8c8cd1 : SUCCESS in 5m 30s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/09baadf225b74d1f94d9ae7e365f9183 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":3},{"id":"beaa70c1c093cca76d84cad874ae0ec460e62570","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 13:49:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20323f341ff340748cbe0ede68279f38 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b071c60a074042eebf58b9cdb15e6f54 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c59706e95104fdfa2ee8056b39b8449 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5400077a123e4139aa4a39407bacb440 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5012c0f1f14439f890045e5a5fc87bf : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b2d2275abe146428ee2b8dd02f23fbf : SUCCESS in 6m 14s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d48233564d7540ebbd8ca51d58ea6083 : SUCCESS in 3h 28m 26s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/596ccf1a642a4b31b6043427fa57d845 : SUCCESS in 1h 58m 13s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/deb5ad57720c4f55a517d418b13a8632 : SUCCESS in 1h 26m 16s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/633fcf943c5a442b9d27807a8a4c0765 : SUCCESS in 1h 34m 02s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5cd2132b45ca4b74b73b081e84561d8f : SUCCESS in 1h 53m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6383cbad4a104934996dac80706cb6d4 : SUCCESS in 5m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/94d91240857c494281a0a5cbbd8c8cd1 : SUCCESS in 5m 30s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/09baadf225b74d1f94d9ae7e365f9183 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":3},{"id":"c904aa7967978ae2b8f4386939d5be3ef07d3354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 14:00:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e3d6ac10c4d3f4c728419abdf41e8159baceb975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 01:52:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71a068d51de842278afa9c616bddb39b : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20d71abcaa8e4b029e90c2cc357eac14 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f8a96c5ab2c4093b12d64f2ab5f9dfa : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fbc88c200c7498b80fc6b6d6ca95778 : SUCCESS in 4m 35s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/091776f712784b438269910d88885a71 : SUCCESS in 3h 25m 19s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ac581f0694014f8d9f3f533e8d4f5ec1 : SUCCESS in 1h 51m 48s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b7fe2d42875047bc80dd181006d727fb : SUCCESS in 1h 30m 05s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ac9215bf3a964a758688b152d0bc623b : SUCCESS in 1h 29m 48s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/34af3a36b71e489a8cf3d75a6fb333b4 : SUCCESS in 1h 50m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9a8d66f25bb04e609ceb3b3b863eaa7e : SUCCESS in 4m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9551a23175f14e9cb4e1423a360c24c6 : SUCCESS in 4m 35s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/da497ded49c24235898628a1fb3b857a : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":3},{"id":"f15e1265f5d88fe5f3c1233bba1ca5b938cc6ac7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 01:52:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f89db613d24ac6ffddbce3686a6c65ceeff7bf90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 01:57:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a3ecfe1a3a4fccacf69368be9dbe56 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"72c7d4ac71d33c8803120767108c6bf2ffeb2edc","revisions":{"5cda453a78a3cee1493ba0b1419e770990928e07":{"kind":"REWORK","_number":1,"created":"2021-03-23 09:37:37.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/85/782385/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/85/782385/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/85/782385/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/85/782385/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/85/782385/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/85/782385/1"}}},"commit":{"parents":[{"commit":"a9846a90aa7e867b07d88ed8a79ee53142585df9","subject":"Merge \"Add passwords_env parameter to generate_passwords\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a9846a90aa7e867b07d88ed8a79ee53142585df9"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2021-03-23 09:34:55.000000000","tz":60},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2021-03-23 09:37:35.000000000","tz":60},"subject":"Use quay.ceph.io to pull the Ceph Dashboard related images","message":"Use quay.ceph.io to pull the Ceph Dashboard related images\n\nThe purpose of this change is just to switch to quay.ceph.io\nand pull from that location the Ceph Dashboard related images.\nquay.ceph.io is maintained by the Ceph organization and can\nbe trusted to find the tags that we need.\n\nChange-Id: If5ec5d4fa0d4d835561b5963d244e5172fe82cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5cda453a78a3cee1493ba0b1419e770990928e07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5cda453a78a3cee1493ba0b1419e770990928e07"}]},"branch":"refs/heads/master"},"73f7301d9148afd8a331f51a9196b57ce1a15b0d":{"kind":"REWORK","_number":2,"created":"2021-03-23 14:34:38.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/85/782385/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/85/782385/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/85/782385/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/85/782385/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/85/782385/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/85/782385/2"}}},"commit":{"parents":[{"commit":"e0b0d859efb179c6868020b3fc10c2155ed96820","subject":"Switch to tripleo namespace grafana images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e0b0d859efb179c6868020b3fc10c2155ed96820"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2021-03-23 09:34:55.000000000","tz":60},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2021-03-23 14:34:24.000000000","tz":60},"subject":"Use quay.ceph.io to pull the Ceph Dashboard related images","message":"Use quay.ceph.io to pull the Ceph Dashboard related images\n\nThe purpose of this change is just to switch to quay.ceph.io\nand pull from that location the Ceph Dashboard related images.\nquay.ceph.io is maintained by the Ceph organization and can\nbe trusted to find the tags that we need.\n\nChange-Id: If5ec5d4fa0d4d835561b5963d244e5172fe82cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/73f7301d9148afd8a331f51a9196b57ce1a15b0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/73f7301d9148afd8a331f51a9196b57ce1a15b0d"}]},"branch":"refs/heads/master"},"72c7d4ac71d33c8803120767108c6bf2ffeb2edc":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-24 10:03:47.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/85/782385/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/85/782385/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/85/782385/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/85/782385/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/85/782385/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/85/782385/3"}}},"commit":{"parents":[{"commit":"e0b0d859efb179c6868020b3fc10c2155ed96820","subject":"Switch to tripleo namespace grafana images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e0b0d859efb179c6868020b3fc10c2155ed96820"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2021-03-23 09:34:55.000000000","tz":60},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2021-03-24 10:03:18.000000000","tz":60},"subject":"Use quay.ceph.io to pull the Ceph Dashboard related images","message":"Use quay.ceph.io to pull the Ceph Dashboard related images\n\nThe purpose of this change is just to switch to quay.ceph.io\nand pull from that location the Ceph Dashboard related images.\nquay.ceph.io is maintained by the Ceph organization and can\nbe trusted to find the tags that we need.\n\nCloses-Bug: #1920873\nChange-Id: If5ec5d4fa0d4d835561b5963d244e5172fe82cc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/72c7d4ac71d33c8803120767108c6bf2ffeb2edc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/72c7d4ac71d33c8803120767108c6bf2ffeb2edc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
