)]}'
{"id":"openstack%2Ftripleo-common~784584","triplet_id":"openstack%2Ftripleo-common~stable%2Fvictoria~I236db070b5d929cb51d34c6eefd47866e07de6a7","project":"openstack/tripleo-common","branch":"stable/victoria","topic":"add_ceph_ansible-stable/victoria","hashtags":[],"change_id":"I236db070b5d929cb51d34c6eefd47866e07de6a7","subject":"Add ceph-ansible to tripleoclient container","status":"MERGED","created":"2021-04-04 17:50:37.000000000","updated":"2021-04-13 15:54:55.000000000","submitted":"2021-04-13 15:53:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784584-add_ceph_ansible-stable/victoria","cherry_pick_of_change":784354,"cherry_pick_of_patch_set":1,"meta_rev_id":"086a9f9711a2c309dbc7944d05f11f1e17a12ecc","_number":784584,"virtual_id_number":784584,"owner":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-13 15:52:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-09 10:02:42.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-04-12 18:54:40.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-04-06 16:47:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2021-04-12 18:54:40.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-04 20:45:00.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-04 21:34:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 15:21:31.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"state":"CC"},{"updated":"2021-04-06 16:47:41.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"state":"REVIEWER"},{"updated":"2021-04-12 18:54:40.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"df166267889f391de007f3dd3535a85e5dd27396","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-04 17:50:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c53cf4f2acd262d193e4492c583d814047a1828a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-04 20:45:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/071dfa947fa144e89d228f726d9ccec5 : SUCCESS in 2h 53m 16s","accounts_in_message":[],"_revision_number":1},{"id":"7bb4b9081dd7110a515627a819f494f4d5cfe84c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-04 21:34:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29da3bd97710433ab52a73670ea6d38a : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca3cf466f84c4fc7863790190bb06cc8 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a5ad8a30eb24960a52102c6a87c5a02 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4555dee475da425cbbbc82ef9b1bcd94 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1ea403f7b154fa8bad317479f4f61ef : SUCCESS in 4m 38s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/763d51a3d7da4d69bb154437e00fc677 : SUCCESS in 3h 33m 45s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6c327d17cdbe4e06859e350eaeb46c92 : FAILURE in 1h 49m 34s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9df53f2e1e0b4d7caa326cc39aa3c137 : SUCCESS in 1h 15m 28s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/76bbc9349907485392cbc4c02cfc242a : SUCCESS in 1h 33m 48s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/2d9adf6b0ffe441d8c756599588c9776 : SUCCESS in 3h 42m 13s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/0cb2719e29724a18b2b48acda70d7978 : SUCCESS in 2h 14m 34s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/91c40318e4444c20b1d1005fd60a772a : SUCCESS in 1h 30m 11s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/b4ad01653ac14d109818ff8bd41c255f : SUCCESS in 2h 03m 26s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/b22f71a288074a6284fdec7dbb3fe9f6 : SUCCESS in 1h 54m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f3bb9f1b50744adb94f0177ea2825f3f : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/40b6e8fdfe5d4c51a1e9675c18ac2295 : SUCCESS in 4m 49s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/3a4b9fe3df59436d9b42a9fcb2c03ad4 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1f861eef649c02627caec264067cc5d2568d4ade","tag":"compute.df.review","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-04-06 15:21:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"02211e99739d9e9306ebb09e308de02a4516eed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 16:27:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51379f4e1a2449179fb5fd8f742ec055 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/963028dab0104ab19ffe3bff749b743e : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7115b48f5c454b48b627819d89f7468b : POST_FAILURE in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c6a7e1b4ea94f26b1f71fe6f60273fd : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995455d12af4457ca9792df8888a26bc : SUCCESS in 5m 12s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/d60a41fbe4b64fed9647d6bbe5e38cf0 : SUCCESS in 1h 04m 22s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b159dc1bf43d4f13b737cf8414e28dc6 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/12c9d61bad304bae8798c7c3f5844269 : SUCCESS in 5m 52s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/a47d035b22444f808703474f44706918 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1c72ab522abf13974dbf06e3ab041b92b536668a","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-04-06 16:47:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5199debd006810b612be84a743f6f6d3031033d1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-06 16:58:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/96741419158245bb9643e1eacbbd594c : FAILURE in 1h 30m 01s","accounts_in_message":[],"_revision_number":1},{"id":"36ce4b589591ec22c1ee620f7f21f5abdc29943f","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-04-09 07:20:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c02f1bd1eb45e6c2542a860d29a7efafc292dab9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-09 10:02:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e13570394e534676beca551eb9fab3dc : SUCCESS in 2h 40m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b6187102bd7d43b5bea3b44393e8b5c9f0007b6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 11:07:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c82c468c683f446fa6924231e66613d8 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e77e649841574354b709a65cb5098eb8 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21f76b325f654bed9df0a8bdf8669cb3 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c130c4274204fbd97303906f31a5584 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c6feaf708da441980ae2d75cf2fc0a4 : SUCCESS in 4m 10s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d32119f15a804033894a218dae2383a6 : SUCCESS in 3h 35m 31s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/d87ccc3d0b2b456882ab5bb5310dde7a : SUCCESS in 1h 30m 09s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e5904e13d3e14d1f85a91ee1f204ddba : SUCCESS in 1h 01m 25s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f8b99af937544fbf96bf5d46b30dc8ab : SUCCESS in 1h 45m 00s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/c8bc6fa45c7d4feab445b0ba9a967b8a : SUCCESS in 3h 41m 49s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/e595040a2a5f43a5b3da2e9f39397bad : SUCCESS in 2h 12m 59s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/80992c122be442a79dba4ad0dfe1cfea : SUCCESS in 1h 30m 08s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/bfb6c0da75184c2695eaaf285c0c7baf : SUCCESS in 2h 11m 02s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/b7fe1b480b264a949158fea851f34a54 : SUCCESS in 2h 08m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/69b39af034bc42998901e0c4ec9709ca : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7ee27a4c85d45ac9808288ffa4ef73e : SUCCESS in 5m 28s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/8038f87f4abe42eaa612349ecfac02f0 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"688597818e370d11f3729d33abb5a5587ca9c8cc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-12 18:54:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b54abd079a1798d02b710171a903bedbd106b1ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 06:26:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fee79bfe188ec1d68dcfb692ec0cb3a17289c512","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 15:52:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe55faa600014ae293be076c9e15deea : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ea4a80dcf9f4b948803d60d5bc279fe : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c8f8d84877b456e82da9d5a6ac80489 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/566f53e1e6fb4998afdf760d60533450 : SUCCESS in 5m 17s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/497cac44cbc44feba58520ada2591e38 : SUCCESS in 3h 17m 38s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a741cf613e81449c8662b34cd48e0e97 : SUCCESS in 2h 07m 57s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8fe3078c525746e7b5d69469ba10b867 : SUCCESS in 1h 09m 39s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/93d6a39642ec48b59101fd492975b9f3 : SUCCESS in 1h 12m 39s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/bec9b760cadf43d6878087bb723b82bb : SUCCESS in 3h 17m 24s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/9adf0836177b4fc5b1872f841a05c5fd : SUCCESS in 2h 11m 03s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/921b23b7c0184309bda4651bc163795f : SUCCESS in 1h 18m 38s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/6687279a029b41e3842124e245cd26ca : SUCCESS in 1h 44m 15s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/5c14155176184569b64241bcc59bfe50 : SUCCESS in 1h 37m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8a558e68072f4c5c8b8978614e9db4b3 : SUCCESS in 6m 04s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/1b262e31a99749dbbb1bfd13e955128a : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":1},{"id":"382cbff8ddb2e990c055896335cfaab6e0d2a22f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 15:53:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"086a9f9711a2c309dbc7944d05f11f1e17a12ecc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 15:54:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f176c6c7928749049b6854466fa0d850 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c33a2c1f75d965625acd8d79128dfa2a85564a32","revisions":{"c33a2c1f75d965625acd8d79128dfa2a85564a32":{"kind":"REWORK","_number":1,"created":"2021-04-04 17:50:37.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/84/784584/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/84/784584/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/84/784584/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/84/784584/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/84/784584/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/84/784584/1"}}},"commit":{"parents":[{"commit":"e5575a781813cf2a0af0092761a2c437cd06f28d","subject":"exclude redhat-release from yum update in openstack-base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e5575a781813cf2a0af0092761a2c437cd06f28d"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-01 09:43:16.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-04-04 17:50:37.000000000","tz":0},"subject":"Add ceph-ansible to tripleoclient container","message":"Add ceph-ansible to tripleoclient container\n\nChange-Id: I236db070b5d929cb51d34c6eefd47866e07de6a7\n(cherry picked from commit 6a5277b4f6378de8f37d433550d53edbcf7a8f36)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c33a2c1f75d965625acd8d79128dfa2a85564a32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c33a2c1f75d965625acd8d79128dfa2a85564a32"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
