)]}'
{"id":"openstack%2Ftripleo-common~788884","triplet_id":"openstack%2Ftripleo-common~master~Id07b3f7c63cac14e40ac231ede6dd543387a4ee1","project":"openstack/tripleo-common","branch":"master","topic":"bp/whole-disk-default","hashtags":[],"change_id":"Id07b3f7c63cac14e40ac231ede6dd543387a4ee1","subject":"Delete python2 based images","status":"MERGED","created":"2021-04-30 04:37:35.000000000","updated":"2021-05-29 10:09:30.000000000","submitted":"2021-05-29 10:08:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"788884-bp/whole-disk-default","meta_rev_id":"d9694e4ac5e1fbdad142ba5c4fdc2c1eaf0a96a1","_number":788884,"virtual_id_number":788884,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-29 10:08:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-04-30 06:31:53.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2021-05-29 06:46:15.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2021-05-03 14:28:03.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-05-04 13:21:43.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-04 08:51:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":1,"date":"2021-05-29 06:46:15.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 06:31:53.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-30 08:22:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-03 14:28:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-05-04 08:51:07.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-05-04 13:21:43.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2021-05-29 06:46:15.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"621884da67b2e630d30e6c1ac6030beb044b0e06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-04-30 04:37:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73bedb10ccb3e91bea24c9c9680a4b02ccb76933","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-04-30 06:31:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/501ca1e371c44524a082eda6145ba943 : FAILURE in 1h 52m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b024c33e512bbdae6855b72e1c065ebef25cbc2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 08:22:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4312fa48791a43d48f14319157a077ed : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09e69d481bbb4e2e8eb64bebb70bb88b : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d75368e33474a84a74026934667df8c : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a21388f444584b91953f51897489721e : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7791dfe9d2cb437eb802602c2d2b3d67 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c74706027e4a9ab74b29a23cc8a25b : SUCCESS in 4m 01s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1f9596c3589f4e0dbf5255c197c27499 : SUCCESS in 3h 27m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/950511abba6f4fd8954e89c4d7865aec : SUCCESS in 1h 55m 51s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ed5822cdfa8b48099d167c6b220f2694 : SUCCESS in 1h 08m 14s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/dc2b5928bfd94d7da706fc044e984dd4 : SUCCESS in 41m 30s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/b1c9405fa7ac4b2f9e8ce5348dc8bf37 : SUCCESS in 45m 28s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/272748f1af3a4aefa3dc2fa20a7e25fb : SUCCESS in 36m 21s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/18d4d971ef8e44838354efe6684cbc3c : SUCCESS in 1h 40m 07s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d8106db5e6cd45eea1bfd01a2e3adae9 : SUCCESS in 2h 10m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cefe587bde5e48679fed5b81761abbaf : SUCCESS in 3m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/248002926c944b6db13aa61f46c31e6c : SUCCESS in 9m 36s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/b397e69caeb9433b86c58fb3f31dba7d : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"40b8ee22199997094879bc38dee21c9e8ccd5cf0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-03 14:28:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b59384cdf45419e6c473b90cc22d93bdf2f6115d","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-05-04 08:51:07.000000000","message":"Patch Set 1: Code-Review+1\n\ngoodbye old python :).","accounts_in_message":[],"_revision_number":1},{"id":"069b4f0e75cacefaec571c52c08b238f1cb378fb","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-05-04 13:21:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"55dd519ffbfb212eac63f8a06f3bda6cd8b4c7b6","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-29 06:46:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"38d54aa996f80e56cf00015ff673beeb762a8305","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-29 06:46:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"acf22da39fffbc97142c0f28b1ee7d5c0c0082c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-29 10:08:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d639cdc49f404e139b43c6db294b5d2f : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f897ca2bab9549b3a9242430a99c9d5e : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bf1d33c80ef4540b0738caaedb7784e : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7212718dc9247508a2f671b9174fa3c : SUCCESS in 4m 55s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c6e654d9b43544308cd742e6134ebe64 : SUCCESS in 3h 21m 36s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/edac8bedadf1454197174e0badbb7a1d : SUCCESS in 1h 53m 00s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3cd51fe437664df480a8679afa1d424c : SUCCESS in 1h 02m 00s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/37d3108230554bfb8fbac9911f86ec9a : SUCCESS in 1h 07m 28s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/787780d69dab40b4a79d0ce03ff636f4 : SUCCESS in 23m 59s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/70f5bf79c85d471f9d43f5985b9b6d6e : SUCCESS in 39m 29s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4a937cc6d98f4b5ab2b5d700e98dcd99 : SUCCESS in 1h 27m 03s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cc09ef68498a4c87885d8e55267b3577 : SUCCESS in 1h 34m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/821a7fcd40224cc4ab80e006690e905b : SUCCESS in 4m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7571ae5e630a4befb7013e2e847c5485 : SUCCESS in 4m 42s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/46cff62d4afa4e3cb3dfc9a2e55956a6 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"26e2f547505925c1f95f08e0de503b1880aefe94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-29 10:08:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d9694e4ac5e1fbdad142ba5c4fdc2c1eaf0a96a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-29 10:09:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/896352ab0b6c48eea9f7a397815ec084 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bf4158ebd51381b3ccedcbff98e24fc51835b75","revisions":{"4bf4158ebd51381b3ccedcbff98e24fc51835b75":{"kind":"REWORK","_number":1,"created":"2021-04-30 04:37:35.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/84/788884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/84/788884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/84/788884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/84/788884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/84/788884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/84/788884/1"}}},"commit":{"parents":[{"commit":"3b23df6d146698caf2b459768f077439cbbf1ad4","subject":"Merge \"Add interface-names to overcloud-hardened-images-uefi-rhel8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3b23df6d146698caf2b459768f077439cbbf1ad4"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-04-30 04:34:28.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-04-30 04:37:24.000000000","tz":720},"subject":"Delete python2 based images","message":"Delete python2 based images\n\nThe py2/py3 transition has been complete for multiple releases, so\nthere is no need to keep these files in master. py2 stable jobs can\ncontinue to use these files in the stable branches of tripleo-common.\n\nBlueprint: whole-disk-default\nChange-Id: Id07b3f7c63cac14e40ac231ede6dd543387a4ee1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4bf4158ebd51381b3ccedcbff98e24fc51835b75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4bf4158ebd51381b3ccedcbff98e24fc51835b75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
