)]}'
{"id":"openstack%2Ftripleo-common~790070","triplet_id":"openstack%2Ftripleo-common~master~I36cb49007942d95f401eda1719a6eac327479503","project":"openstack/tripleo-common","branch":"master","topic":"I36cb49007942d95f401eda1719a6eac327479503","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-08-11 08:05:18.000000000","reason":"\u003cGERRIT_ACCOUNT_8449\u003e replied on the change","reason_account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},"15895":{"account":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"last_update":"2022-08-16 22:23:44.000000000","reason":"Change was submitted"},"7144":{"account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"last_update":"2022-08-09 21:46:03.000000000","reason":"\u003cGERRIT_ACCOUNT_7144\u003e replied on the change","reason_account":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}},"30073":{"account":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"last_update":"2022-08-16 22:23:44.000000000","reason":"Change was submitted"},"6926":{"account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"last_update":"2022-08-12 13:59:58.000000000","reason":"\u003cGERRIT_ACCOUNT_6926\u003e replied on the change","reason_account":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}},"28223":{"account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"last_update":"2022-08-10 11:33:41.000000000","reason":"\u003cGERRIT_ACCOUNT_28223\u003e replied on the change","reason_account":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}}},"hashtags":[],"change_id":"I36cb49007942d95f401eda1719a6eac327479503","subject":"Remove sysfsutils from nova container","status":"MERGED","created":"2021-05-06 13:33:59.000000000","updated":"2022-08-16 22:24:47.000000000","submitted":"2022-08-16 22:23:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"790070-I36cb49007942d95f401eda1719a6eac327479503","meta_rev_id":"7ba912660026dc977e1a43ff0b87dc4077856e16","_number":790070,"virtual_id_number":790070,"owner":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-16 22:23:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-12 16:20:40.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"recommended":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":1,"date":"2022-08-11 08:05:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2022-08-12 13:59:58.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":2,"date":"2022-08-10 11:33:41.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":1,"date":"2022-08-16 12:05:55.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"}],"CC":[{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 16:42:16.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-06 16:55:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-10 13:11:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-05-10 14:21:19.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-05-10 14:27:53.000000000","updated_by":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2021-05-10 14:30:22.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2022-06-23 21:33:30.000000000","updated_by":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"reviewer":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"state":"CC"},{"updated":"2022-07-01 12:55:49.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"CC"},{"updated":"2022-07-15 14:23:59.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2022-07-15 14:24:39.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"},{"updated":"2022-07-15 14:24:39.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2022-08-01 07:41:26.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"CC"},{"updated":"2022-08-11 08:05:18.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"}],"messages":[{"id":"0bb61a5a459bc7bf67dbc411cd0336aeb75dae31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2021-05-06 13:33:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"279b0708e05e5b538c3304339ff0186d37507219","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-06 16:42:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1f0bb177718f40cf99ccd110c73a4606 : SUCCESS in 3h 05m 59s","accounts_in_message":[],"_revision_number":1},{"id":"02dc7a696e140f06db272b870613db8034e32cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 16:55:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b777c97c884840439a45317409fd7b0f : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6600cc50a4c643c19d384b0b66b36801 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa1c67b6d4a44b5c874dabbfe927daf6 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb8cf2e01ebc4b469d6b703d667122a2 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1549c08b2f2d428a88914f81601c2490 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66192003258c43ec90e724b018956b70 : SUCCESS in 4m 30s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/610f8b1bb81945ab9678d48c3e691c1d : SUCCESS in 3h 08m 52s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5246dc6629ee49b5a387d449656a54ab : SUCCESS in 1h 47m 07s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6f66137851464550906781d689066fa6 : SUCCESS in 1h 08m 33s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3ee8ded938774b7ebc5aafe717b48c39 : SUCCESS in 1h 26m 30s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/313fdcb27a2a41228f7775c6005b0533 : SUCCESS in 1h 49m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/365ad6e6a3c743788387547e2c29b181 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2fa8676fb32f4a9d9c6ecdf7382583b4 : SUCCESS in 5m 16s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/9a0d3fa3a89e464eb9ebee0304151554 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":1},{"id":"35e39ed2bdadc095ad467f106eb1e3d7d12ff7fb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-10 13:11:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"06148da6e01220de908c5294671881e04ad8c32c","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-05-10 14:21:19.000000000","message":"Patch Set 1: Code-Review+1\n\nSounds legit, but would love some Nova reviewers.","accounts_in_message":[],"_revision_number":1},{"id":"4187a907dfd9a63164a2bcd5445fd31c995ba27f","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-05-10 14:27:53.000000000","message":"Patch Set 1: Code-Review+1\n\nSane from Nova\u0027s POV","accounts_in_message":[],"_revision_number":1},{"id":"f6de287d14e2a29e16820f834cf3af1f4206f892","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-05-10 14:30:22.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM, AFAIK this dependency was moved over to os-brick and appears to have been removed since with os-brick instead interacting directly with sysfs.","accounts_in_message":[],"_revision_number":1},{"id":"3a3570d3a0ba52588eb0dc2c6eabe46e9dc497d1","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-05-10 14:31:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSince we got ACKs from Nova - let\u0027s go! Thanks Artom, Lee!","accounts_in_message":[],"_revision_number":1},{"id":"b0781f7ec5ffff17af4d13a1e5807882bcad02aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 14:32:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5d21c6ef6511c0641bae5f51d126cfb187f37a4e","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-05-10 14:33:52.000000000","message":"Patch Set 1: -Code-Review\n\n\u003e Patch Set 1: Code-Review+1\n\u003e \n\u003e LGTM, AFAIK this dependency was moved over to os-brick and appears to have been removed since with os-brick instead interacting directly with sysfs.\n\nApologies I borked the grep looking for systool in os-brick, it\u0027s still there:\n\nhttps://github.com/openstack/os-brick/blob/1b2e2295421615847d86508dcd487ec51fa45f25/os_brick/initiator/linuxfc.py#L162-L164","accounts_in_message":[],"_revision_number":1},{"id":"3750ad4bb7308655e2c4e0b319249691b65fc4ad","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-05-10 14:37:28.000000000","message":"Removed Workflow+1 by Cedric Jeanneret \u003ccjeanner@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"05e1b795773cd07377db2d6101cfc25774e0ec7f","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-05-10 14:37:29.000000000","message":"Removed Code-Review+2 by Cedric Jeanneret \u003ccjeanner@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"b2c758ac1702960f688a079e1e93738690593978","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-05-10 14:40:54.000000000","message":"Patch Set 1: Code-Review-2\n\nget it out of gate","accounts_in_message":[],"_revision_number":1},{"id":"c993e7191d0e32a7000bf1bb14397f0a74d39388","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-10 14:42:40.000000000","message":"Abandoned\n\nabandoning to kick out of the gate. will restore","accounts_in_message":[],"_revision_number":1},{"id":"b03a5bea686bed4a2a71f91eca5c0e8aa59324e2","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-10 14:42:44.000000000","message":"Removed Code-Review+2 by Alex Schultz \u003caschultz@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"4f0cedac7b974e0234a2655613c5f2e9b84245cf","tag":"autogenerated:gerrit:restore","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-05-10 14:42:47.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"6f2f8e3a126c72ea292c2298943134116cbe1304","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-05-10 14:43:18.000000000","message":"Removed Code-Review-2 by Cedric Jeanneret \u003ccjeanner@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"18326208fc8e3b6a58b9acb3de76b68a3c8d5e95","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-05-10 14:47:19.000000000","message":"Patch Set 1: Code-Review-1\n\n\u003e Patch Set 1: -Code-Review\n\u003e \n\u003e \u003e Patch Set 1: Code-Review+1\n\u003e \u003e \n\u003e \u003e LGTM, AFAIK this dependency was moved over to os-brick and appears to have been removed since with os-brick instead interacting directly with sysfs.\n\u003e \n\u003e Apologies I borked the grep looking for systool in os-brick, it\u0027s still there:\n\u003e \n\u003e https://github.com/openstack/os-brick/blob/1b2e2295421615847d86508dcd487ec51fa45f25/os_brick/initiator/linuxfc.py#L162-L164\n\nSo to be clear we should only go ahead with this removal once os-brick has been updated to drop this direct dependency on sysfsutils and the systool binary it provides. I\u0027m going to also follow up and ensure this is reflected in both the openstack-nova and os-brick RPM spec files.","accounts_in_message":[],"_revision_number":1},{"id":"faa1f9b7eda965818ae6aee5c7c932a350755226","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-05-10 16:09:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/337bfbe8184643dfa522f17e8bbd2790 : RETRY_LIMIT in 22m 02s","accounts_in_message":[],"_revision_number":1},{"id":"bad6f57124d8abadfa9a31938a4e9a2ce200068f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 16:18:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a65d376ed7f4224a311299790aa0530 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c0c63c554f640f5b42e0e2c6f2025a5 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4fe02b5d66204884bfdb2b8a00d54eec : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5861ce0886564394a86385b0b49f9610 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d6d4a87c88e47b48f506aad0186f653 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fac812ace2246269d7c86ef6e1b4e50 : SUCCESS in 5m 48s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/88204bcd2512452ea569912301c7e6c0 : FAILURE in 1h 26m 07s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d9cbb94ce9224590827f2c3163b4b4c8 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fcf5ae4304074708ab77da9cece7fc32 : SUCCESS in 5m 36s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/0562e94c8e1c4f1eb96c9edf57c1cd94 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4b95d80d01d3ceb346a8100fa6f43abe30913ca2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-09-14 15:41:18.000000000","message":"Abandoned\n\nAbandoning this patch per the TripleO Patch Abandonment guidelines\n(https://specs.openstack.org/openstack/tripleo-specs/specs/policy/patch-abandonment.html).\nIf you wish to have this restored and cannot do so yourself, please reach out\nvia #tripleo on OFTC or the OpenStack Dev mailing list.","accounts_in_message":[],"_revision_number":1},{"id":"fd2ad9af8c379161d6c338d697df8dd6faa12d22","author":{"_account_id":16282,"name":"YaZug","email":"jschluet@redhat.com","username":"yazug"},"date":"2022-06-23 21:33:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c4c5fae0970eb888a2e1f1e144b3f82c8b1d1a88","tag":"autogenerated:gerrit:restore","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-01 12:54:26.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"779075a6b9864b24d38c0af65d90da54bb2bdac0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-01 12:54:41.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging opendev.org/openstack/tripleo-common for 790070,1","accounts_in_message":[],"_revision_number":1},{"id":"6bac3dc3f9194ddde1fba201845a6caaf37a109e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 12:54:50.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/tripleo-common for 790070,1","accounts_in_message":[],"_revision_number":1},{"id":"024b061840d578940c11e3f3e46388c4a1306c9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-01 12:55:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bb20c1f0ff989212870d9d0ab84d4e8f9fb887fb","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-01 12:55:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e58b2556e21c4b93994346935c3c7986fd25ee6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 15:31:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44f8a1012e8943819bfb6fa9aa10182d : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bca5a5f5c18470a813b6c09b0fc77e9 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdeae2f6e2dd4005ace4be3c494bba0d : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e22e2ed029674c52a9afb5b60d826f65 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e3e289cb62f4da79a57f50ef284f556 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fb47af8e04b4a62acbf4f7c1247c1c0 : SUCCESS in 3m 55s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ad49fc0e55a04d37b525ccf09888687f : SUCCESS in 2h 26m 46s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/c84a6f4a3d054f588c4c888edb05bf13 : SUCCESS in 1h 20m 52s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/63a8d9d231c24bdca81c0a0c0e4a8620 : SUCCESS in 1h 24m 28s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/597712f704304b7bbf1058f96b63f3f1 : SUCCESS in 37m 24s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/064b7bb58159474daf34647b4286db9b : SUCCESS in 1h 11m 30s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/b6c068c334a5430da74bf138ee1300a3 : SUCCESS in 2h 05m 00s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2158cba04c2c47129e34299b21e01f7c : SUCCESS in 1h 16m 49s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/56c3e9d10835479189187e54d2717204 : SUCCESS in 55m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8d406399ae604a33832388211107b29f : SUCCESS in 4m 15s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/a7c91a7e6ea549ee9a0e18ced2d4b075 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":2},{"id":"adc6f72b773abbc6e88460b5876d01e0cd8c332e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-01 16:30:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a9f2e1113f444e1cad11dc5fdcc35470 : FAILURE in 2h 45m 50s","accounts_in_message":[],"_revision_number":2},{"id":"713913def85495607da063c7ae6fe4b8e51fca46","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-15 14:23:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9cf822941ed6f89700a4c1b46ec38590167c98d3","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-08-01 01:06:36.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bfb5a7a783ce912b1e0792d081f512c2242c33dc","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-08-01 07:41:26.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a61329beec7e7f1eeb1e2891581cb64bfc252342","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-08-03 12:52:36.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f4db59fb2c576c8e6d7d13f0226d65af7da7ea2","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-08-03 12:56:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1efa5340e25e4700ae62ff4b27ae34e888d18d45","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-08-09 21:46:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bd91c22a02ce9a72c67b09b71d4407979eb64271","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-08-09 21:47:21.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"090bf06a31e306493aa66c10727d182b22770e0e","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-08-10 07:47:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"adadd3172ed52e1d9a79b85e9ebe651bb48b2f06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2022-08-10 11:01:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d65f93b658218f5c863c3e5f2f5715d1320b00f4","author":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"date":"2022-08-10 11:04:01.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0a96420f8d2f8a51f9c48b3822bf0469018fa058","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-08-10 11:33:41.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"df6ba87d2ce42af7d7ab29d248f644c17ae9b6d1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-10 11:56:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0124338409c2488697462a8f4c40d90d : FAILURE in 52m 45s","accounts_in_message":[],"_revision_number":3},{"id":"7d646f59eaebf9ce56981156596da91071393e3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 14:04:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e257fbdf4d01452dadbba43884fee9f8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c1eb027b7724e89af0d13305007a815 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1967da714234a4a8be28c3c3593f3e4 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75a11e57d12546c4835b5aa500aeeb4e : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33096974af324e6d858a1bd5375e3754 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/488316321ba745aebe12c954927d4cb3 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eae900b0af7d4ebbb950368cdbb8888e : SUCCESS in 4m 37s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/10e8374e5f4e419d8aeecfe959d2c72c : SUCCESS in 2h 37m 13s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/4527cf202b2e4af6baaf231ec3a91541 : SUCCESS in 1h 33m 50s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/003970bce9324983b9925814b10af43f : SUCCESS in 1h 29m 20s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8019b909185d4cad9cd0f23aeed9664e : SUCCESS in 41m 30s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/0eee38a5f434439cbe6286c502369f53 : SUCCESS in 1h 18m 10s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/2351c8d1ffd34ca8a92f843ca29838da : FAILURE in 1h 03m 29s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/5a807c60b16b41a99fd86937999b731d : SUCCESS in 2h 21m 31s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ebac938d203448c88a384a6304f1e3c7 : SUCCESS in 1h 26m 56s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/7307ef8a7626402198ae6fc698c7a731 : FAILURE in 55m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/119f3856aeb14dcf8fce928ca715856e : SUCCESS in 5m 05s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/31b32d3580ae4d80bf1049034e480019 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":3},{"id":"2353a3921ccc62de08dc879beeae7596e84867f5","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-08-11 08:05:18.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8a59beb3b91665716c46e0677a46240daced6352","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-08-12 13:59:58.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9433ba3ffd18fe410996a8e9ff57684cf0354a0c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-12 16:20:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3991bcec84bb4c7891bc3568c4f9fb3b : SUCCESS in 2h 15m 23s","accounts_in_message":[],"_revision_number":3},{"id":"12a6d7dec27ce419f406c0d516ad3eaba271b802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-12 17:17:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ac7b4f3080f426596c4d482171faf78\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30853f6a90e34bda9d6d2d76d51e5008 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38bb1e6d8aa64f24b6435e9cb1db8dd7 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a72a0c41816c462393ab7981f7b428c3 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93d6a8a1a50f40ca8b5ec11f7c3dcc3f : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01037e1dd2d342c1b7d0c6622b1d448c : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/717e429c8c5843fc8dfa7bb6ad67aad0 : SUCCESS in 4m 08s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c24528b78a3947199a4d44ccb7ce7db4 : SUCCESS in 3h 03m 05s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/3426c43a410142b592172866f2ed2b69 : SUCCESS in 1h 45m 46s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/39f4dbc0bd364fcfa727ee018e80274d : SUCCESS in 1h 40m 27s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8c67ff1c79de4dbfb2f65708d4f45f2f : SUCCESS in 53m 29s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/0f1cec9e84bc46c6af6eea664c4ab450 : SUCCESS in 1h 23m 19s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/bf0d48a0c86a4acb8423657d1abb23bf : FAILURE in 1h 07m 09s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/50b9e957703f4203b6d2b8280a4a95f1 : SUCCESS in 2h 48m 35s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/10c3e0b6fcbc4b26a9ff2fe30b87ed43 : SUCCESS in 1h 33m 06s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/980a10cb1fb94688915d142ad218c072 : SUCCESS in 1h 21m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aaec6cdc8598415daacfe064d586226e : SUCCESS in 3m 22s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/d410caacb9484efeb8c4c9141546e632 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":3},{"id":"a80f79e79e43ee121a429fc45810f80800582801","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-08-16 12:05:55.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7d34e8b3a2b6a95c88a67eb5e96cdffab9aaf4ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 12:06:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4925088fac892c77631042cd8bb8e2a665bcf95c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 14:31:22.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbb73c94a30d4a349a28e83433da9246\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3b819bda2954ce4b931fe80ac55e81b : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18f9436c509a453c8a10826b32691e9f : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01fbe5a9605549838c2e9993c13ee6d1 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8128b2f050364ce5b9727fafd434e21e : SUCCESS in 4m 30s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/54d7eb853cc049d3abf7d28f0dd603f1 : SUCCESS in 2h 24m 09s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/cf6f1b50948a43b7866dc15336c0a813 : FAILURE in 1h 08m 22s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/dddb94f561504481b1a2c54be0bdc2e4 : SUCCESS in 48m 50s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/79b49f6b24174c23aff1f991e8425b0c : FAILURE in 37m 18s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/566c101535fb4b43b11ae41e8320ec8e : SUCCESS in 2h 11m 05s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/f2043d6091604c5f90e352891b35e23e : SUCCESS in 1h 06m 55s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3843c9f86765463f9bd181b7775529ca : SUCCESS in 1h 21m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cc72575473d842ad99c56939f74cb3eb : SUCCESS in 4m 24s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/337e821963d04bb0be54fc24d2f924fd : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"880de0f042ea8908946ac76887145158c6fa0e2d","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-08-16 14:48:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5ac93e33a82d537a8700343b162391a7b4a5d47d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-08-16 17:13:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d9321a13c07b4d568222896eef02e8c1 : SUCCESS in 2h 22m 18s","accounts_in_message":[],"_revision_number":3},{"id":"c50c4f85d6c1265fa9d30b60f463f3b26eb2f2f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 17:43:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9937e04f589c474a8d2987f6f26f3f10\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/618c769e2feb4aea82d5c94edadb0ccc : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bcb5cf59cc349d4b0081c2b012db3c8 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c89ddb128dc548dab48bec8f95d99b57 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55f089484262417f851a60916931a1a4 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0027506ed0b4bc39afbbfd6c3daf498 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e94561b10f1493da585384f9bd14389 : SUCCESS in 4m 16s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/fe5ca41c5a274fe2b47f1abd09af6403 : SUCCESS in 2h 42m 41s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f558a668f35145088eec1a7f4cbb1e63 : SUCCESS in 1h 38m 45s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a596eae07e324749bd7767e84349e7ef : SUCCESS in 1h 27m 46s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/4c30167c6a3544bbac48c40c6059ab91 : SUCCESS in 39m 44s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/c679c7ab84b249d5b90c4567504b92cd : SUCCESS in 1h 15m 23s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/52d2afab1d5e4a549c42cc730b9f7924 : SUCCESS in 1h 29m 38s (non-voting)\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/78a032f40ad1407b87b128196fccf9a7 : SUCCESS in 2h 22m 04s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0eaf4fdc158d4d28911fa5775a4ca8a2 : SUCCESS in 1h 26m 39s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/6dffb7d80f044e5ca7836d578b000012 : SUCCESS in 1h 11m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8cc8fd2e527f42429d67680bef2c6a8f : SUCCESS in 4m 07s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/e3ade88f451d46938594f399905d0db3 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":3},{"id":"ed7b3abe2f0048909cf3a98ef7853031d5bc4d22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 17:44:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5af0f8153da0a207a8fa88fa4a66ac591ecf6b84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 22:23:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/240d0734482d4a008f740a597ebb2751\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ef341ec7049436d9730805c2e444484 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c67fb818fc324551bfc2b5e1bb1baa89 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d17191e4358843a68baf0d9513515bcd : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb8934898bc34fbeadeda740c387226f : SUCCESS in 5m 10s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/169362621ef54b5da378b610af70669e : SUCCESS in 2h 44m 17s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/1ca885bdc8be4f35bd893b05773ff313 : SUCCESS in 1h 36m 37s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/51bb0502fba14ef68e4635f4377f9f77 : SUCCESS in 41m 30s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/6cc02011822b40a9b1cefa16fa69e6a0 : SUCCESS in 1h 18m 30s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d726ac324c87417f88725f61b4eea539 : SUCCESS in 2h 15m 34s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/8cfd7f6458084bec9966fb264065594e : SUCCESS in 1h 07m 14s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a85ed7f1e650447298a512d6759d9b68 : SUCCESS in 1h 27m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/95ee93574708413fa50bdf06a00f0372 : SUCCESS in 4m 51s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/1119af2e28054a628fdbedd1126ca6e1 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":3},{"id":"19fa19635c29fc4a9703795bec14bdbe4b18829d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 22:23:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"7ba912660026dc977e1a43ff0b87dc4077856e16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-16 22:24:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6407bf730cc403a939848cb26d46e23\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ac610738f9e4092ac64a262e9ea8b5d : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"def937575a0f46a9195bf01246ba3383bd41e138","revisions":{"68bbedae86a7c676d7e553cfd074620858dea832":{"kind":"REWORK","_number":1,"created":"2021-05-06 13:33:59.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/70/790070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/70/790070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/70/790070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/70/790070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/70/790070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/70/790070/1"}}},"commit":{"parents":[{"commit":"f7cfed9e5c81cb320d6a5577332c9ad822fda29d","subject":"Add rhosp-release rpm in rhel images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f7cfed9e5c81cb320d6a5577332c9ad822fda29d"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2021-05-06 13:32:18.000000000","tz":120},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2021-05-06 13:32:18.000000000","tz":120},"subject":"Remove sysfsutils from nova container","message":"Remove sysfsutils from nova container\n\nsysfsutils was an optional depency of nova for\nfiber channel storage integration.\nsysfsutils and systool shoudl nolonger be used in nova directly\nas a result this depency shoudl nolonger be required.\n\nsysfsutils is going away in centos 9 which will be the basis of\nso we should remove it from the package list in the\nnova-compute container.\n\ndirect usage of sysfsutils/systool was removed a while ago but it is\nstill used by fiber channel in os-brick, but os-brick does require\nthe package correctly.\n\nChange-Id: I36cb49007942d95f401eda1719a6eac327479503\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/68bbedae86a7c676d7e553cfd074620858dea832"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/68bbedae86a7c676d7e553cfd074620858dea832"}]},"branch":"refs/heads/master"},"9229811de0f938d5835583d95599cf1ea329224d":{"kind":"REWORK","_number":2,"created":"2022-07-01 12:55:25.000000000","uploader":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"ref":"refs/changes/70/790070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/70/790070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/70/790070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/70/790070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/70/790070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/70/790070/2"}}},"commit":{"parents":[{"commit":"8a5555d7fc0e91cc531134bbd55fb15816ec3fcd","subject":"Merge \"Remove overcloud-agent element from overcloud images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8a5555d7fc0e91cc531134bbd55fb15816ec3fcd"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2021-05-06 13:32:18.000000000","tz":120},"committer":{"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","date":"2022-07-01 12:55:12.000000000","tz":120},"subject":"Remove sysfsutils from nova container","message":"Remove sysfsutils from nova container\n\nsysfsutils was an optional depency of nova for\nfiber channel storage integration.\nsysfsutils and systool shoudl nolonger be used in nova directly\nas a result this depency shoudl nolonger be required.\n\nsysfsutils is going away in centos 9 which will be the basis of\nso we should remove it from the package list in the\nnova-compute container.\n\ndirect usage of sysfsutils/systool was removed a while ago but it is\nstill used by fiber channel in os-brick, but os-brick does require\nthe package correctly.\n\nChange-Id: I36cb49007942d95f401eda1719a6eac327479503\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9229811de0f938d5835583d95599cf1ea329224d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9229811de0f938d5835583d95599cf1ea329224d"}]},"branch":"refs/heads/master"},"def937575a0f46a9195bf01246ba3383bd41e138":{"kind":"REWORK","_number":3,"created":"2022-08-10 11:01:18.000000000","uploader":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"ref":"refs/changes/70/790070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/70/790070/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/70/790070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/70/790070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/70/790070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/70/790070/3"}}},"commit":{"parents":[{"commit":"8a5555d7fc0e91cc531134bbd55fb15816ec3fcd","subject":"Merge \"Remove overcloud-agent element from overcloud images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8a5555d7fc0e91cc531134bbd55fb15816ec3fcd"}]}],"author":{"name":"apetrich","email":"apetrich@redhat.com","date":"2021-05-06 13:32:18.000000000","tz":120},"committer":{"name":"apetrich","email":"apetrich@redhat.com","date":"2022-08-10 10:59:27.000000000","tz":120},"subject":"Remove sysfsutils from nova container","message":"Remove sysfsutils from nova container\n\nsysfsutils was an optional depency of nova for\nfiber channel storage integration.\nsysfsutils and systool shoudl nolonger be used in nova directly\nas a result this depency shoudl nolonger be required.\n\nsysfsutils is going away in centos 9 which will be the basis OS\nfor new versions so we should remove it from the package list in the\nnova-compute container.\n\ndirect usage of sysfsutils/systool was removed a while ago but it is\nstill used by fiber channel in os-brick, but os-brick does require\nthe package correctly.\n\nChange-Id: I36cb49007942d95f401eda1719a6eac327479503\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/def937575a0f46a9195bf01246ba3383bd41e138"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/def937575a0f46a9195bf01246ba3383bd41e138"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}}]}],"submit_requirements":[]}
