)]}'
{"id":"openstack%2Ftripleo-common~790077","triplet_id":"openstack%2Ftripleo-common~master~Iae6df8f2eebc2ce9ae47d59b69215b47c15ff4c3","project":"openstack/tripleo-common","branch":"master","topic":"docs/validationsfix","hashtags":[],"change_id":"Iae6df8f2eebc2ce9ae47d59b69215b47c15ff4c3","subject":"Removed mentions of mistral from the doc","status":"MERGED","created":"2021-05-06 14:23:19.000000000","updated":"2021-05-25 15:17:41.000000000","submitted":"2021-05-25 15:15:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"790077-docs/validationsfix","meta_rev_id":"cd46be8409bd419177c2c7fcb188c39113218e65","_number":790077,"virtual_id_number":790077,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-25 15:15:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2021-05-25 10:58:51.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-25 08:48:08.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2021-05-25 10:58:51.000000000","_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-06 14:25:47.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2021-05-06 14:25:47.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-05-06 14:25:47.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-05-06 14:25:47.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-05-06 14:25:47.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2021-05-06 14:25:47.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2021-05-06 14:25:47.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2021-05-06 15:15:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-07 07:32:31.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REMOVED"},{"updated":"2021-05-10 07:41:29.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":15895,"name":"Adriano Petrich","email":"apetrich@redhat.com","username":"apetrich"},"state":"REVIEWER"},{"updated":"2021-05-10 07:41:29.000000000","updated_by":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"}],"messages":[{"id":"c61c27a0ffc56ea69c27f8a02e019bbc394f5029","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-06 14:23:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"779d168ddda16060a5172737ce52c2861aa14c62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-06 14:26:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f3c6e749202316d015f62fa42b7614021ab1847","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-06 15:15:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5ab36cc81f14475083da3ec66baa494a : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6531505511a54f47aae38799577d206e : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99570d2981854390819c9e91a712eb20 : SUCCESS in 6m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8276bbc645e44ae5817df2dda5f0e5f0 : SUCCESS in 7m 00s","accounts_in_message":[],"_revision_number":2},{"id":"8c6098158cd1a10cc56e43a9a4f8549646062dc1","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-05-07 06:59:21.000000000","message":"Patch Set 2: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1ebf491041bc2287abeb9d94cb70534cfdb8ddd7","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-07 07:21:27.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c23b3b4ec44e7dd223fb7488e0c28d784005e271","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2021-05-07 07:32:31.000000000","message":"Removed reviewer Andreas Jaeger.","accounts_in_message":[],"_revision_number":2},{"id":"7b7eacc17239f59a6db03c446a4ea1bc35dd0ef1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-07 07:42:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"accb3628c039e7ccb3d4e23b75da6332f29da0c1","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-07 07:43:27.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"780f62bd0aa2e1fb533be8d2dd3640a932c62e35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 07:47:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e6cb70923bec45658c54f33e7e306580 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28cc0a3a17854548986f879bcfedf2d5 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ecd7c57ced44dcaca9362b14fcd13b : SUCCESS in 4m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4966576ebe414c61babbe5812762c319 : FAILURE in 4m 14s","accounts_in_message":[],"_revision_number":3},{"id":"c9056bb45ec202ed93b1f5fd93d8ffb3396eb304","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-07 07:57:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8b8546ac561f5f04dc3204c28523a63c2ab2b1c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 08:07:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/21fca78154a44f0c9e5f2d40d933d097 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/989c5b382d8d4b6ca0a17ae2db2565df : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6d14a6150c240c7a500a22fbbde2315 : SUCCESS in 4m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e0a074cd2cfc4b4198cde8e2ae716f5c : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":4},{"id":"ce8b24cb368c7173837181da1731c7ebe62740f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-10 11:29:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"dd3890cb3d05fb907624b88d5da5c93f7f7a346a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 11:44:57.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/01424b627a36425bbe7e71c6bafbfdd8 : SUCCESS in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77cf25b704c747b7899c29e6b761e2a4 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c00b5a4529a04e2597f776b6ee6f7749 : SUCCESS in 4m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5a7bca5ea104591b9a9cbfcbaad87da : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":5},{"id":"a14ed34e8a74644232baa092d7aadd6adf4e9496","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-10 11:50:12.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"83bfeb75017b12966c257ca13b14b490a0e0eef7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-10 13:57:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1b6848effea47546b0ef77f05142d8c028344bb2","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-10 14:07:20.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"23d878b3c1d20332326295005d2f3d42e3fd8a21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-10 14:08:06.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2b1799e23a384e25a433d2db95133245 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4af7e19211d4d47a35f57f0336ca44a : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae4b239cec4b47839f38ea5e1f233ca1 : SUCCESS in 4m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/99e59756453044fe8ca7b47d1b9aab19 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":6},{"id":"efdf250ef8bb266f96ad17c2ccb6befeb9a6146f","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-11 03:16:54.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"b003378dcb3c674420c88518d7b3b4cd6e9221a5","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-05-11 05:02:26.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)\n\nIt would also be great to change:\n\n- in the setup.cfg (s/workflows/deployment):\n\n   ...\n   summary \u003d A common library for TripleO deployment.\n   ...\n\n- And removing doc/source/_exts/workbooks.py which is useless now.","accounts_in_message":[],"_revision_number":6},{"id":"f88e884ee23fd451d1f0f2ffaa349a7db61311cf","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-05-11 05:43:03.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e614d4760a45b003621cba8ddd2b6f53dc872911","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-11 05:48:31.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"941b464b57b8f4920583108377d3188953f67964","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-11 05:50:14.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e848523b7b44e231c76e2b89fd7ca2dcb743d8ac","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-05-11 05:53:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"983d994f9cdd83486a32c548f3072cb21bf48ca6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-11 05:56:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/78c23fb31b884002b5723c303e5c37c7 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5005a9b83bd24b71986dfc542db7782d : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f42c4fc9e6848a4ad7d13a88898c116 : SUCCESS in 5m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/424e49656a1b43e4932147ce56472ec6 : SUCCESS in 5m 11s","accounts_in_message":[],"_revision_number":8},{"id":"b84c85108da951239259d4cc9445838c7a64ca28","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-05-25 08:48:08.000000000","message":"Patch Set 8: Code-Review+2\n\ngoodbye Mistral :).","accounts_in_message":[],"_revision_number":8},{"id":"3e3c8c5859b9e505d231fcc973f369a52afd85ae","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-05-25 10:58:51.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"23f9dcc4ee00f720e2ff265536aa7219ce0617da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 10:59:10.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"a0fc9a9018195b880ee8c2c3d7f48bbeec423c70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 15:15:17.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/03539eb07115445994ce8b928aa444ad : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/431753c05dbb4b83aed9591d54aab31c : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab7ffaee22544b968f608b9c033f5e6c : SUCCESS in 4m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/310489dbb075485aaa28d7d2f57f5bd9 : SUCCESS in 4m 22s","accounts_in_message":[],"_revision_number":8},{"id":"82947c11bd45bb163a5c9b95dd8c6c722238d43e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 15:15:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"cd46be8409bd419177c2c7fcb188c39113218e65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 15:17:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e671bdee5f7448ccb0492c1e1a01c1ba : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"5ab784739f2c7d40a8c3e0d5fe7cb01c28ab716f","revisions":{"9e5c2ef624a5188b66b71a1d94cfd7ac7a549c68":{"kind":"REWORK","_number":1,"created":"2021-05-06 14:23:19.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/790077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/77/790077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/1"}}},"commit":{"parents":[{"commit":"3b23df6d146698caf2b459768f077439cbbf1ad4","subject":"Merge \"Add interface-names to overcloud-hardened-images-uefi-rhel8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3b23df6d146698caf2b459768f077439cbbf1ad4"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-06 14:19:12.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-06 14:23:08.000000000","tz":120},"subject":"Docs updated with current state of VF/Mistral","message":"Docs updated with current state of VF/Mistral\n\nDocs now remind reader that Mistral is long\ndeprecated, and removed.\n\nWarnings point to the tripleo-validations docs\nand small primer on use of VF is included in readme.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Iae6df8f2eebc2ce9ae47d59b69215b47c15ff4c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9e5c2ef624a5188b66b71a1d94cfd7ac7a549c68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9e5c2ef624a5188b66b71a1d94cfd7ac7a549c68"}]},"branch":"refs/heads/master"},"8c0e500022e942dba594ed2059e7970180157150":{"kind":"REWORK","_number":2,"created":"2021-05-06 14:26:33.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/790077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/77/790077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/2"}}},"commit":{"parents":[{"commit":"3b23df6d146698caf2b459768f077439cbbf1ad4","subject":"Merge \"Add interface-names to overcloud-hardened-images-uefi-rhel8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3b23df6d146698caf2b459768f077439cbbf1ad4"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-06 14:19:12.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-06 14:26:24.000000000","tz":120},"subject":"Docs updated with current state of VF/Mistral","message":"Docs updated with current state of VF/Mistral\n\nDocs now remind reader that Mistral is long\ndeprecated, and removed.\n\nWarnings point to the tripleo-validations docs\nand small primer on use of VF is included in readme.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Iae6df8f2eebc2ce9ae47d59b69215b47c15ff4c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8c0e500022e942dba594ed2059e7970180157150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8c0e500022e942dba594ed2059e7970180157150"}]},"branch":"refs/heads/master"},"644cbec802ab52d3bdc842712187eb2f75b1bbf7":{"kind":"REWORK","_number":3,"created":"2021-05-07 07:42:01.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/790077/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/77/790077/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/3"}}},"commit":{"parents":[{"commit":"3b23df6d146698caf2b459768f077439cbbf1ad4","subject":"Merge \"Add interface-names to overcloud-hardened-images-uefi-rhel8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3b23df6d146698caf2b459768f077439cbbf1ad4"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-06 14:19:12.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-07 07:32:06.000000000","tz":120},"subject":"Docs updated with current state of VF/Mistral","message":"Docs updated with current state of VF/Mistral\n\nDocs now remind reader that Mistral is long\ndeprecated, and removed.\n\nWarnings point to the tripleo-validations docs\nand small primer on use of VF is included in readme.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Iae6df8f2eebc2ce9ae47d59b69215b47c15ff4c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/644cbec802ab52d3bdc842712187eb2f75b1bbf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/644cbec802ab52d3bdc842712187eb2f75b1bbf7"}]},"branch":"refs/heads/master"},"a63ac116aa55b0c0fbe7b22b41aa2b7a5c3098c4":{"kind":"REWORK","_number":4,"created":"2021-05-07 07:57:13.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/790077/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/77/790077/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/4"}}},"commit":{"parents":[{"commit":"3b23df6d146698caf2b459768f077439cbbf1ad4","subject":"Merge \"Add interface-names to overcloud-hardened-images-uefi-rhel8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3b23df6d146698caf2b459768f077439cbbf1ad4"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-06 14:19:12.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-07 07:57:04.000000000","tz":120},"subject":"Docs updated with current state of VF/Mistral","message":"Docs updated with current state of VF/Mistral\n\nDocs now remind reader that Mistral is long\ndeprecated, and removed.\n\nWarnings point to the tripleo-validations docs\nand small primer on use of VF is included in readme.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Iae6df8f2eebc2ce9ae47d59b69215b47c15ff4c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a63ac116aa55b0c0fbe7b22b41aa2b7a5c3098c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a63ac116aa55b0c0fbe7b22b41aa2b7a5c3098c4"}]},"branch":"refs/heads/master"},"35cd2ce396580e052b540bb5c43588ec24b95ce9":{"kind":"REWORK","_number":5,"created":"2021-05-10 11:29:37.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/790077/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/77/790077/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/5"}}},"commit":{"parents":[{"commit":"3b23df6d146698caf2b459768f077439cbbf1ad4","subject":"Merge \"Add interface-names to overcloud-hardened-images-uefi-rhel8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3b23df6d146698caf2b459768f077439cbbf1ad4"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-06 14:19:12.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-10 11:27:49.000000000","tz":120},"subject":"Removed mentions of mistral from the doc","message":"Removed mentions of mistral from the doc\n\nAlso removed mentions of VF and validations in general,\nas they are not relevant to this repo.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Iae6df8f2eebc2ce9ae47d59b69215b47c15ff4c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/35cd2ce396580e052b540bb5c43588ec24b95ce9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/35cd2ce396580e052b540bb5c43588ec24b95ce9"}]},"branch":"refs/heads/master"},"a29d3bfd9f13d8656b9988914050b6866b779348":{"kind":"REWORK","_number":6,"created":"2021-05-10 13:57:11.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/790077/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/77/790077/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/6"}}},"commit":{"parents":[{"commit":"3b23df6d146698caf2b459768f077439cbbf1ad4","subject":"Merge \"Add interface-names to overcloud-hardened-images-uefi-rhel8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3b23df6d146698caf2b459768f077439cbbf1ad4"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-06 14:19:12.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-10 13:57:03.000000000","tz":120},"subject":"Removed mentions of mistral from the doc","message":"Removed mentions of mistral from the doc\n\nAlso removed mentions of VF and validations in general,\nas they are not relevant to this repo.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Iae6df8f2eebc2ce9ae47d59b69215b47c15ff4c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a29d3bfd9f13d8656b9988914050b6866b779348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a29d3bfd9f13d8656b9988914050b6866b779348"}]},"branch":"refs/heads/master"},"89e230a6d2ff6fd7ef701fb091a119b84e920b40":{"kind":"REWORK","_number":7,"created":"2021-05-11 05:48:31.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/790077/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/77/790077/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/7"}}},"commit":{"parents":[{"commit":"3b23df6d146698caf2b459768f077439cbbf1ad4","subject":"Merge \"Add interface-names to overcloud-hardened-images-uefi-rhel8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3b23df6d146698caf2b459768f077439cbbf1ad4"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-06 14:19:12.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-11 05:48:22.000000000","tz":120},"subject":"Removed mentions of mistral from the doc","message":"Removed mentions of mistral from the doc\n\nAlso removed mentions of VF and validations in general,\nas they are not relevant to this repo.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Iae6df8f2eebc2ce9ae47d59b69215b47c15ff4c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/89e230a6d2ff6fd7ef701fb091a119b84e920b40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/89e230a6d2ff6fd7ef701fb091a119b84e920b40"}]},"branch":"refs/heads/master"},"5ab784739f2c7d40a8c3e0d5fe7cb01c28ab716f":{"kind":"REWORK","_number":8,"created":"2021-05-11 05:50:14.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/77/790077/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/77/790077/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/77/790077/8"}}},"commit":{"parents":[{"commit":"3b23df6d146698caf2b459768f077439cbbf1ad4","subject":"Merge \"Add interface-names to overcloud-hardened-images-uefi-rhel8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/3b23df6d146698caf2b459768f077439cbbf1ad4"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-06 14:19:12.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2021-05-11 05:50:08.000000000","tz":120},"subject":"Removed mentions of mistral from the doc","message":"Removed mentions of mistral from the doc\n\nAlso removed mentions of VF and validations in general,\nas they are not relevant to this repo.\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Iae6df8f2eebc2ce9ae47d59b69215b47c15ff4c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5ab784739f2c7d40a8c3e0d5fe7cb01c28ab716f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5ab784739f2c7d40a8c3e0d5fe7cb01c28ab716f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
