)]}'
{"id":"openstack%2Ftripleo-common~794736","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~Ia0baf8501884d9ed2324938c9acbb30d1687cf27","project":"openstack/tripleo-common","branch":"stable/train","topic":"bug/1929881-stable/train","hashtags":[],"change_id":"Ia0baf8501884d9ed2324938c9acbb30d1687cf27","subject":"Fix config parsing in memcached healthcheck","status":"MERGED","created":"2021-06-04 08:21:45.000000000","updated":"2021-06-09 21:56:56.000000000","submitted":"2021-06-09 21:55:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"794736-bug/1929881-stable/train","meta_rev_id":"f29428a3573a878a5d9e7aeff09b62f12a004d46","_number":794736,"virtual_id_number":794736,"owner":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 21:55:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-04 12:36:51.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":2,"date":"2021-06-09 08:24:58.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-09 09:20:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"date":"2021-06-08 14:48:46.000000000","_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":1,"date":"2021-06-09 08:24:58.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-04 12:36:51.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 17:39:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 08:24:58.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2021-06-09 09:20:01.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"28382337a223597d719e63f964f7a3d1afe82acb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2021-06-04 08:21:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df6c4026776d3125e5edc9a3d2b98043a5570443","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2021-06-04 08:21:59.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1b422c682e6e77784af6b9a11fbe3b2027166512","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-04 12:36:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c1a3dc8d7b164fdfb45b1442a390966e : SUCCESS in 3h 44m 54s","accounts_in_message":[],"_revision_number":1},{"id":"bb27da02a5397678bbfa33dddaa028eaf920b57d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:39:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b4bc52a1e084f15b1999ea9c1d51657 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3da238bcb534998842f004959e61d03 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5f12432859e84f79a472f29a82eb296e : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2da8802fcbf04dfc918e1f33c5c6f147 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4a22cbc769474705b2b60394ccf13a38 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b04f93f5684f42af91c03ffd1520109c : SUCCESS in 5m 11s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a685a8b709f0490282949fb219fd8144 : SUCCESS in 3h 30m 20s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/3b86f80eef3f436ca752f0d703a4e87a : SUCCESS in 3h 39m 07s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/aad2113be52249ca80526d7211021548 : SUCCESS in 2h 11m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/77d64b86384a476e9e18b3271ddae796 : SUCCESS in 2h 08m 07s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/57ece26429ee4263ba544084c29cd56b : SUCCESS in 1h 32m 03s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/9b65cdd5b849495c8ed370e82a0ae31a : SUCCESS in 55m 48s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/35523aaa3d6e48d7a4fefcf54c2cb9a6 : SUCCESS in 1h 18m 14s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/a85b2b1bb3694644892cf440b2a1858f : FAILURE in 43m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/37bbc25dc31f42b786f98cdb162e069f : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4c7f8dc44f33ae927004b041b9c0b22ddc823953","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2021-06-08 14:47:07.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"c873592ead357122730583037adeeefb4f453e09","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2021-06-08 14:48:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"edb7829b673598821648f2b5ac98a02e62f633b2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-08 20:43:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2b53f783cca943b297277a384417cf04 : SUCCESS in 3h 35m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bc8f2685859d14d9ea0f486c9bac7098c3f05049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 20:56:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49cdc849a74f4bfb8380c24c13903270 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5701cc8948464302bea88323b8d66267 : SUCCESS in 7m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c69ce3461c9642c0a7d4cdbcdcac1eea : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10068785e515497bbde7ce2801e65622 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7829f8a0edce46ba9c81b6517868f69d : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17bfc02f59ec4f65b3c8e71441a6c686 : SUCCESS in 6m 15s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ca20f4cbe82f4e43831728f161a1543b : SUCCESS in 3h 11m 35s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/a365261c21fe450395f401828fac1c54 : SUCCESS in 4h 26m 40s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/60f5607707f347a29be1f817457b4038 : SUCCESS in 2h 03m 59s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/8b6c3130cff649ec878b9650aa321c96 : SUCCESS in 2h 33m 12s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0a4a408df3a24d45b28144dd53f929ab : SUCCESS in 1h 30m 34s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/bafe422751e846c08995412ac1e21d80 : SUCCESS in 1h 34m 23s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/80d364431e9241048e6d4b30054fd7c3 : SUCCESS in 1h 09m 39s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/fa3f72c8b4744fbdb396067df287512d : SUCCESS in 1h 56m 12s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/50c22a2f728d44afaac118cf00fec8ee : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"147583d58d8c45f56d8470d9b786a5116dee9d12","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-06-09 08:24:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"16158e56f12fe07a40984fedf5c43385fe9f83b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 08:26:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3bf1f7f3c07dcdab111475d0b65d6b54b3e9d407","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-09 09:20:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8ae1cf2f472b2d0ef6c307e0b790ea8896a9a26b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 21:55:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cb83123f08e4d4986196719cd85a850 : SUCCESS in 6m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2521747c6ae64e8c93ee6d0ef813fc8f : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d1a5b895589441ca2200df9f05b5413 : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f87d31fe01654bf5b5a248b4646caee2 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28aa890c23924d9aa71a91a4cff4af99 : SUCCESS in 5m 54s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2e16cefe6f6b488190d805aa832cf278 : SUCCESS in 3h 19m 20s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/f6fc299a7dd547f280a9ab356c63bf4f : SUCCESS in 3h 32m 35s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7c9ea374d4724eb885a0c387cfd68431 : SUCCESS in 2h 04m 16s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/3d819c8bee5246ea8e4980d0a2b2a39b : SUCCESS in 2h 06m 22s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c9fc87f22b4e4ac3bd01e37ad92f37a6 : SUCCESS in 1h 29m 52s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/c460565774c74b12b4f4cb732321685a : SUCCESS in 54m 45s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9774865163104162acc72ca3356c8bd7 : SUCCESS in 1h 11m 40s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/63cb15799d4f46e385da08bfb8464fe6 : SUCCESS in 1h 57m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f4e4a291cef54f7e93907f12c882566d : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":1},{"id":"60f2e138f444e836285690629d3bebf2e2e4af91","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 21:55:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f29428a3573a878a5d9e7aeff09b62f12a004d46","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 21:56:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a344e878f42840b6b9011725f9ed3318 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdd2fb3b388548fc20d1f7e1e0578466448b35ca","revisions":{"fdd2fb3b388548fc20d1f7e1e0578466448b35ca":{"kind":"REWORK","_number":1,"created":"2021-06-04 08:21:45.000000000","uploader":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"ref":"refs/changes/36/794736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/36/794736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/36/794736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/36/794736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/36/794736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/36/794736/1"}}},"commit":{"parents":[{"commit":"7d9b08621b0246b3bd67736a1c32f362c41191e2","subject":"Merge \"Fix site-packages in horizon\u0027s extend_start.sh\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7d9b08621b0246b3bd67736a1c32f362c41191e2"}]}],"author":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2021-05-27 22:25:25.000000000","tz":120},"committer":{"name":"Damien Ciabrini","email":"dciabrin@redhat.com","date":"2021-06-04 08:21:32.000000000","tz":120},"subject":"Fix config parsing in memcached healthcheck","message":"Fix config parsing in memcached healthcheck\n\nWhen memcached is configured to listen on multiple IP, or\nis configured with TLS-e, make sure the healthcheck parses\nthe right IP to connect to and its associated port.\n\nChange-Id: Ia0baf8501884d9ed2324938c9acbb30d1687cf27\nCloses-Bug: #1929881\n(cherry picked from commit 3f84495b8f8dc0a20148dfa8eaaa22d750ada39d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fdd2fb3b388548fc20d1f7e1e0578466448b35ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fdd2fb3b388548fc20d1f7e1e0578466448b35ca"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
