)]}'
{"id":"openstack%2Ftripleo-common~794814","triplet_id":"openstack%2Ftripleo-common~stable%2Fqueens~I688de151c654a19003397adacf07ae67f66001a7","project":"openstack/tripleo-common","branch":"stable/queens","hashtags":[],"change_id":"I688de151c654a19003397adacf07ae67f66001a7","subject":"[TRAIN ONLY] Handle empty role Specific Parameter","status":"ABANDONED","created":"2021-06-04 16:43:59.000000000","updated":"2021-07-08 15:18:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":758021,"cherry_pick_of_patch_set":3,"meta_rev_id":"c3141ece002b8e945358c79f6382326483935392","_number":794814,"virtual_id_number":794814,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16347,"name":"Purandhar Mannidi","display_name":"Sai","email":"sai438@gmail.com","username":"mpsairam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-04 19:38:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-06-04 18:31:03.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"all":[{"value":1,"date":"2021-06-08 15:54:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2021-06-04 18:08:54.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-06-29 00:41:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":16347,"name":"Purandhar Mannidi","display_name":"Sai","email":"sai438@gmail.com","username":"mpsairam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16347,"name":"Purandhar Mannidi","display_name":"Sai","email":"sai438@gmail.com","username":"mpsairam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16347,"name":"Purandhar Mannidi","display_name":"Sai","email":"sai438@gmail.com","username":"mpsairam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-04 16:43:59.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":16347,"name":"Purandhar Mannidi","display_name":"Sai","email":"sai438@gmail.com","username":"mpsairam"},"state":"REVIEWER"},{"updated":"2021-06-04 18:08:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-04 18:31:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 19:38:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-08 15:54:04.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"}],"messages":[{"id":"b4658d423dd56b44c7a0832ced536660245bc789","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-06-04 16:43:59.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"86f39b2911fc6497e0c4a65dd955f99f203bb28c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-04 18:08:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"35d3cc308f75240b863564b950f2572d5790c07a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-04 18:31:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e9328f42e8f8488caa7070b8d39e5aa9 : FAILURE in 1h 34m 52s","accounts_in_message":[],"_revision_number":1},{"id":"95fc3d20a9aeca624c2f1a170247a7a4d29ff76e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 19:38:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/eb199b333c1b48aeaf53fde6a5603bae : SUCCESS in 2h 20m 31s\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/49d29496f3914887b215acaba9bab44f : SUCCESS in 1h 09m 59s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a1f0eae1cd7c42ddbed54452bfd812b6 : FAILURE in 53m 29s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/31ce70564acd4184ae182aeeb5a42c37 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f86c18eded194bbe97bf72061f5e0841 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e854ea1b76774962be5606fc03211f4c : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/001e2d881852421fa1b22c8c215cc7dc : SUCCESS in 5m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1561c826816245698badbc2fbbc9d5c5 : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"052bc9ba5de24dce388d6764b891aa0cc6b07904","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-08 15:54:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"efb56894b1976eba2286a74888eb46efa4dbf198","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-06-10 13:20:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"305661ddfcc37f3186d06e63b0b0dffc889e850c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 15:35:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/2b5ea492128f462dabc50d294a53e78c : SUCCESS in 2h 13m 59s\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/e17bedcc5f0140d4aa8a6c4d88eb3320 : SUCCESS in 1h 04m 18s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/880b1969b154401f94a3acbc07cbfb31 : FAILURE in 56m 05s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/10dc888aade9496fb7b7103dee513e04 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96dd934b64da49e6a9878d9ea6886925 : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ea32b1d59172487bb9fa74915a75150b : SUCCESS in 5m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/61adb2d771f54b59967d7a99af3f1252 : SUCCESS in 5m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e07a8422f5f4472817a39aae11ef8e0 : SUCCESS in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28ba83f4ef46ff4cda12a7c0accac9ed16790b14","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-10 16:17:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ce996a1c8be041d9a3f7cdf4ac262811 : FAILURE in 1h 45m 31s","accounts_in_message":[],"_revision_number":1},{"id":"1c68310488133e161e04775df739bd132e0dc632","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-06-28 21:46:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e35e0ff15c750049ebec90d483b36da3043e28e5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-28 23:27:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/139c49655601453ead7c1df000ea13b7 : FAILURE in 1h 39m 33s","accounts_in_message":[],"_revision_number":1},{"id":"819955e28c3075c4ab8dcb1b17984f4c7ae35359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 00:02:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/7c458d253d4e4566b76e12c9d09e16f6 : SUCCESS in 2h 13m 41s\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/32b00f52fbbe4a368cdcf8f7a184ed83 : FAILURE in 1h 02m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/1d411622a3a94d65b30c7ba4869e7166 : FAILURE in 1h 00m 17s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4b4475b41b0a4d449dca87059cfbeae9 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ea694e683a34a2ebdbfb8e7e2c11222 : SUCCESS in 6m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b0a995ae4bc047488ff89660ba0b903b : SUCCESS in 6m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ddadd8ba2588417995467d110aec38ab : SUCCESS in 6m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f18c27ece1194019a6a22c23d4b37a8c : SUCCESS in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7a02a10e0cc7e13020c725589f050c119ee01d53","author":{"_account_id":16347,"name":"Purandhar Mannidi","display_name":"Sai","email":"sai438@gmail.com","username":"mpsairam"},"date":"2021-06-29 00:41:17.000000000","message":"Patch Set 1: Code-Review+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9d131b9e37cb16e4ba5bde5aa0ad8649f6ed062a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-29 02:45:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f10861455d5542cebd14860fd50ff422 : FAILURE in 1h 31m 59s","accounts_in_message":[],"_revision_number":1},{"id":"926fac9a70ddf8c2f42f07e2560348904b232e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 02:58:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/4855cc6e36a344e7957c1fca4cf77185 : SUCCESS in 2h 15m 33s\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/83456e0493a34a98809fae80c8d575cc : FAILURE in 56m 32s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/a8c7751ab79e4b759e82bd50f0b2c886 : FAILURE in 46m 46s\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/13310d67570d49418ba0fdf62c14fbb9 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52df7880271748178e0777500e399019 : SUCCESS in 5m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a80433f8149949dda11f3c3c25a44ba3 : SUCCESS in 5m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/24a895d6d47045b887878cc518e7cafd : SUCCESS in 6m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/061ccad61cb045b0b02363ee473821d5 : SUCCESS in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff4d496a9222a5fbda9bea2473d25c127f64a40b","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-07-08 14:22:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e6b60f9c8912ed06b36464efe347bcb006a503cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 14:51:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/4ee1f99b9b7f4a328cc479df39dbee7b : FAILURE in 17m 11s\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/13ee4090255641a4a60ef1405e4d69a7 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36df2a8661c64ecfbdc28ee1eef40709 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8bdf678e3ed74661b3d4bad3e5ab2bad : SUCCESS in 4m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b43602be562e4982aeaaf64a84d37e9d : SUCCESS in 4m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5efcfa93ff6542068de61137a007a780 : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3141ece002b8e945358c79f6382326483935392","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-07-08 15:18:15.000000000","message":"Abandoned\n\nbranch is going EOL see http://lists.openstack.org/pipermail/openstack-discuss/2021-July/023537.html","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a827e5aabcebf4140a72f9114f36d1eeb832609","revisions":{"4a827e5aabcebf4140a72f9114f36d1eeb832609":{"kind":"REWORK","_number":1,"created":"2021-06-04 16:43:59.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/14/794814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/14/794814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/794814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/794814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/14/794814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/14/794814/1"}}},"commit":{"parents":[{"commit":"8afc5d23cdf6532c4ec57689c0e1ff82e079748c","subject":"Ensure tripleo ansible inventory file update is atomic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8afc5d23cdf6532c4ec57689c0e1ff82e079748c"}]}],"author":{"name":"Purandhar Sairam Mannidi","email":"pmannidi@redhat.com","date":"2020-10-14 07:37:12.000000000","tz":600},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2021-06-04 16:43:59.000000000","tz":0},"subject":"[TRAIN ONLY] Handle empty role Specific Parameter","message":"[TRAIN ONLY] Handle empty role Specific Parameter\n\nHaving empty role specific parameters will fail the deployment as\nmistral workflow YAQL expression fails.\n\nChange-Id: I688de151c654a19003397adacf07ae67f66001a7\nCloses-Bug: #1900810\n(cherry picked from commit 5a86b8c117a7c55d240331ef91e87846c78aa9da)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4a827e5aabcebf4140a72f9114f36d1eeb832609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4a827e5aabcebf4140a72f9114f36d1eeb832609"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
