)]}'
{"id":"openstack%2Ftripleo-common~795033","triplet_id":"openstack%2Ftripleo-common~stable%2Fwallaby~Ib96e2c0dd789be2be6bdaf66682b985cb864baf8","project":"openstack/tripleo-common","branch":"stable/wallaby","topic":"remove_tripleo-validations_legacy_dir-stable/wallaby","hashtags":[],"change_id":"Ib96e2c0dd789be2be6bdaf66682b985cb864baf8","subject":"Remove references to tripleo-validations legacy directory","status":"MERGED","created":"2021-06-07 06:52:44.000000000","updated":"2021-06-11 08:52:35.000000000","submitted":"2021-06-11 08:49:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795033-remove_tripleo-validations_legacy_dir-stable/wallaby","cherry_pick_of_change":794145,"cherry_pick_of_patch_set":1,"meta_rev_id":"81669561efc54c8e4a89b3ae11d6bceb310b2b11","_number":795033,"virtual_id_number":795033,"owner":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-11 08:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-09 08:01:46.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-06-07 13:48:29.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-09 09:31:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2021-06-07 07:05:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2021-06-10 16:28:12.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-07 06:53:22.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2021-06-07 06:53:22.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-07 06:53:22.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-06-07 06:53:22.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"state":"REVIEWER"},{"updated":"2021-06-07 06:53:22.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":27427,"name":"David Peacock","email":"dpeacock@redhat.com","username":"davidjpeacock"},"state":"REVIEWER"},{"updated":"2021-06-07 06:53:22.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":16515,"name":"mbu","email":"mat.bultel@gmail.com","username":"matbu"},"state":"REVIEWER"},{"updated":"2021-06-07 10:12:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-07 10:32:11.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-09 09:31:10.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"f1c4669ef0b8648a9a4bcac6d1b8e68d2dd3a76b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-07 06:52:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d1b173537b488ac22e65edd36fe1ac731faf08be","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2021-06-07 07:05:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2de29da5953ea6bb02ec1dbc167339a8d2d1bc73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 10:12:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/579e661530af4594aa3ef796c75f6bc3 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a8992ffa2f542cd90a7a4082c959662 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65b1faa693bf4e3387238f894adecbde : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/305a78d7d0774133ad5b44bb64a36733 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a8f67d7939b4c6bb7c5935f9e5969dd : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1bd70aebfb04911804105a4eb5e7c0f : SUCCESS in 4m 34s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/21cdd83d60a940ae8c69c6418b2b0e2c : SUCCESS in 2h 58m 37s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/2afff63502fd447f82f639447cb6cb70 : SUCCESS in 1h 36m 06s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4a64dab21e1c4a6e9bf146fffd00095c : SUCCESS in 1h 05m 09s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/611aa4a971de4d009b4cdde57387d347 : SUCCESS in 1h 16m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8d3189fafbb84dee8b455a13e7698805 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f70866d39a8a4c9e813ff07b8f2566c5 : SUCCESS in 4m 41s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/5eb155123a344a9a95383f98c0d3d2e5 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f56b66af923f419414e382eb9daf05f39ab63abc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-07 10:32:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e28a39104e504e29b8903fc67a99e9d7 : FAILURE in 3h 10m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f8b08e4c93be1a38ed43f58852139e80cdee6345","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-07 13:48:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2b5048c09d00442a169e7ba27dea8e9f5946ad85","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2021-06-09 04:48:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"79282acb6ca0ce03218e3b1deedd46d30989181c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-09 08:01:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/bee4d1e1b24e40c5bc97e8365f2ce55b : SUCCESS in 3h 11m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bff45427e76a1ae92635ca2f41d17c5228bd60b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 08:13:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b02154d4a19447a895bb0d75b0c900b3 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9cfe7a0e58a4bd2bade479bf6984bd2 : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a7dd67ca74b4c3e813a41b1382f9258 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30f9f34c0a9c487ba50b9ffe53879b75 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5eee14c776aa4f5a81113689ab50d1b7 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/007314aa2d1542dbbb892b66842e0005 : SUCCESS in 4m 22s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e2f0cb8466b443bc89aaac719ed08118 : SUCCESS in 3h 07m 24s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ffe12b08ebbc4cb3a18566cc4b2420da : SUCCESS in 1h 41m 05s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9eb27a75ebcc418e9432fe8335767534 : SUCCESS in 1h 06m 07s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f82ef4b2561d486389b6a7415722aca6 : SUCCESS in 1h 22m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/98406cc6b3864edf99f5b15ccfaef696 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4fd358dd45c04891aa2e48039222c943 : SUCCESS in 4m 43s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/6d057de544cf4391a86e8be5d9be61f5 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d7155b9f63bac1a386cf71eebf15bbe7c29f4f87","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-09 09:31:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"157f82b81d4d9af82d03bc931f8517d031a079a9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-10 16:28:12.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b6d688fead2e3fa88c995907cdccb9d864a18dd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 21:54:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"792a6ee7835002d0982b06a6bcab4c12c6ed7fa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 02:11:29.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd66d26504e52d4c0b6f63a91197d31d3f18bc70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 08:49:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3c183a4cf5242da921fe01f1cf779fe : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0590480700c458ab88a254411e0a7a1 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/daab5be3ce7b47c3815417e894cffa23 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cfbaf82dc2d444eaab51390a8c89a4c : SUCCESS in 4m 42s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1df29aff17084924b41243eb336083eb : SUCCESS in 2h 57m 48s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5ad1a494d6db4c628054e2cfb0158b16 : SUCCESS in 1h 37m 11s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3c89298e3d4c44e1bcbe25c39b22c321 : SUCCESS in 1h 03m 46s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c61076328c5d405ebc057622cd7f318e : SUCCESS in 1h 24m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/669093d62a61426a9745c582ef640ccb : SUCCESS in 4m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6c28928b7fa14730983b90b7d8958352 : SUCCESS in 4m 39s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/f3a2d0793e8f45dd979a0812fae6ee81 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"efa0ec6fec882b8ee74531aaf65a5d0b61456b6b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 08:49:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"81669561efc54c8e4a89b3ae11d6bceb310b2b11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 08:52:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27572f898534b2b8d73c41b6addbf7c : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ba767305d47687f7fca334ab720f61873179e77","revisions":{"7ba767305d47687f7fca334ab720f61873179e77":{"kind":"REWORK","_number":1,"created":"2021-06-07 06:52:44.000000000","uploader":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"ref":"refs/changes/33/795033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/33/795033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/795033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/795033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/795033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/33/795033/1"}}},"commit":{"parents":[{"commit":"d39a3282bdf17cc849e1e281b6c780919ad0ee61","subject":"Merge \"Remove xinetd from the mariadb container\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d39a3282bdf17cc849e1e281b6c780919ad0ee61"}]}],"author":{"name":"Gael Chamoulaud (Strider)","email":"gchamoul@redhat.com","date":"2021-06-02 07:13:24.000000000","tz":120},"committer":{"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","date":"2021-06-07 06:52:44.000000000","tz":0},"subject":"Remove references to tripleo-validations legacy directory","message":"Remove references to tripleo-validations legacy directory\n\n\u0027/usr/share/openstack-tripleo-validations\u0027 doesn\u0027t exist anymore since\nTrain. Tripleo-validations installs its bits in \u0027/usr/share/ansible/\u0027.\nIt is now quite safe to remove the references to this legacy directory.\n\n**Note**\nThis patch will be applied to master and stable/wallaby only. This logic\nwill be kept as is for stable/train like this directory may still be\npresent during an upgrade.\n\nChange-Id: Ib96e2c0dd789be2be6bdaf66682b985cb864baf8\nSigned-off-by: Gael Chamoulaud (Strider) \u003cgchamoul@redhat.com\u003e\n(cherry picked from commit 755cc6095de5894b802e2a9ed709abf82e55d1dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7ba767305d47687f7fca334ab720f61873179e77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7ba767305d47687f7fca334ab720f61873179e77"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
