)]}'
{"id":"openstack%2Ftripleo-common~796516","triplet_id":"openstack%2Ftripleo-common~stable%2Fwallaby~I7db4adaf4509cf45545eb16e777e764a2b19df3b","project":"openstack/tripleo-common","branch":"stable/wallaby","topic":"bug/1931995","hashtags":[],"change_id":"I7db4adaf4509cf45545eb16e777e764a2b19df3b","subject":"[Wallaby-only] Update default container namespace for wallaby","status":"MERGED","created":"2021-06-15 15:38:54.000000000","updated":"2021-06-16 07:44:13.000000000","submitted":"2021-06-16 07:41:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796516-bug/1931995","meta_rev_id":"e6140d725c1396d1f47eba7cb7666d1f0661c847","_number":796516,"virtual_id_number":796516,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 07:41:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-15 18:56:37.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":2,"date":"2021-06-15 15:59:30.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2021-06-15 20:08:16.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2021-06-15 20:08:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2021-06-15 20:08:16.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 15:47:02.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2021-06-15 15:50:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 18:56:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 20:08:16.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-06-15 20:08:25.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"}],"messages":[{"id":"cdd1fab9d4dba3458f860ba4bdcb30bf84f3c71d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-15 15:38:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a58efb8cedde8168854da79a89beb8ab1fb2e56a","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-06-15 15:47:02.000000000","message":"Patch Set 1: Code-Review+2\n\ntripleo container image prepare sets -\u003e namespace: quay.io/tripleowallaby with this patch.\n\nThat seems correct.","accounts_in_message":[],"_revision_number":1},{"id":"b55d896cd38df7b2677ec46c44cd1aa9f039fdd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 15:50:44.000000000","message":"Patch Set 1: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f88c4158755e467e8ce537e375c983b8 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e770e563171942ada0adfc6caf95ad4c : FAILURE in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c37755c0d40b4de8bcc042917bdce324 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc2612a6bafe460a8975c14cb0d3970e : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8ac6872a01245bfa5d9979c005f4bcd : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c74549d654bf4a35be3b982857e42a34 : SUCCESS in 4m 58s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0014a45f8ad3427eb9fa9cfb145991ba : FAILURE in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7f7d712f40eb40269bb905a238ea0a55 : SUCCESS in 5m 16s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/e9c97d584ced486189ff315cb4750979 : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"daa5a8b34e33eb977099aa22bb8072e168f9811f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-15 15:54:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bcb05256ba80eb21261b1ecb56c21b7878dfef23","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-06-15 15:59:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2617c298dc610f85b5e87837352c183514cb8b49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 18:56:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e9cc80d134c4f768717bd84da862a74 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3dd50eaf0d45fcbcac1e558e5d64f1 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b28feaf253c64342b44c1d446d87f654 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc5ea6279add43b99ca12e13e1669702 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/588b420f2a954a83a6158514c9acee51 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46ea60cb8c2a4d91902c8cee017096ef : SUCCESS in 6m 29s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/596621dcc7b84a2ba724ef2bec6d9435 : SUCCESS in 2h 42m 32s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c559a76d7d4f4f0986772a415860fd08 : SUCCESS in 1h 31m 02s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f70b3584f49545b7aeabe008ca66fb1f : SUCCESS in 1h 12m 57s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/d231e7c6f1094600a29f3fcd3c3b0c5a : SUCCESS in 1h 10m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/59d31123e70a42a589215a8c113144ac : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1c7b1a09fe94c05aea444e11fae48a2 : SUCCESS in 6m 46s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/374751c2012f4ee8aa412d999c0a695c : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":2},{"id":"4fdc84d5a2a33496bc5f02ed4d473f6ef38f8d71","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-15 18:56:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d3db7dcd46c4424dadb20610d234b884 : SUCCESS in 2h 59m 48s","accounts_in_message":[],"_revision_number":2},{"id":"7c49d6b8e1eaf42a54ab8a30677aff7dc165e8ef","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2021-06-15 20:08:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6d3588fb206cc065158861e9254068e8bf513d8f","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-06-15 20:08:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"c1f65b8383ec9cac1ba10eefe51975aabed9db02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 01:42:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0c24f96d40ff47d0a66ff66845da60048ba6e549","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 07:41:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f54614605894420597f21f27d6545f4e : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b93c569f0f4d468fbe8eb0ee4ffb74ea : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54f006df60c84a6f859e53a1b307319c : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2a91e537a694756b34c521a020c0fad : SUCCESS in 5m 04s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/24b14c719b0c4841a1d9791dbe79e0a8 : SUCCESS in 3h 07m 05s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e06d796b1c1e4f1d83a5592fd739358a : SUCCESS in 1h 36m 44s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fa83925eb93b49c0b042cf61a96fdf55 : SUCCESS in 1h 03m 16s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0f96a818564845aba05cd621a48eee16 : SUCCESS in 1h 14m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f343d39710c40748838e7dc1b387f26 : SUCCESS in 5m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/881ffb7232fd4e54812b2aabace3c273 : SUCCESS in 5m 07s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/bd37e8f80d4a47689c73aff3b9cf8709 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":2},{"id":"b2848166fe1a5a1aac06ef12a7e64fbe04d6d582","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 07:41:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e6140d725c1396d1f47eba7cb7666d1f0661c847","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 07:44:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/babf7cc590254c2083dbc237a2377bdf : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"31468f279470778d16a73b92a644a52fe4e26598","revisions":{"fcc2210762748b2d593b1b0f94f01267010ed9e8":{"kind":"REWORK","_number":1,"created":"2021-06-15 15:38:54.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/16/796516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/16/796516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/16/796516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/16/796516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/16/796516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/16/796516/1"}}},"commit":{"parents":[{"commit":"6041b235117cb8d1b046709a74de8a3cc50c8de6","subject":"Merge \"Move rabbitmq healthcheck to something more modern\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6041b235117cb8d1b046709a74de8a3cc50c8de6"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-15 15:38:04.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-15 15:38:04.000000000","tz":-360},"subject":"[Wallaby-only] Update default container namespace for wallaby","message":"[Wallaby-only] Update default container namespace for wallaby\n\nAfter cutting a stable branch, we need to patch it with the correct\ncontainer information for the release.\n\nChange-Id: I7db4adaf4509cf45545eb16e777e764a2b19df3b\nCloses-Bug: 1931995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fcc2210762748b2d593b1b0f94f01267010ed9e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fcc2210762748b2d593b1b0f94f01267010ed9e8"}]},"branch":"refs/heads/stable/wallaby"},"31468f279470778d16a73b92a644a52fe4e26598":{"kind":"REWORK","_number":2,"created":"2021-06-15 15:54:48.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/16/796516/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/16/796516/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/16/796516/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/16/796516/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/16/796516/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/16/796516/2"}}},"commit":{"parents":[{"commit":"6041b235117cb8d1b046709a74de8a3cc50c8de6","subject":"Merge \"Move rabbitmq healthcheck to something more modern\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6041b235117cb8d1b046709a74de8a3cc50c8de6"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-15 15:38:04.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-15 15:54:44.000000000","tz":-360},"subject":"[Wallaby-only] Update default container namespace for wallaby","message":"[Wallaby-only] Update default container namespace for wallaby\n\nAfter cutting a stable branch, we need to patch it with the correct\ncontainer information for the release.\n\nChange-Id: I7db4adaf4509cf45545eb16e777e764a2b19df3b\nCloses-Bug: 1931995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/31468f279470778d16a73b92a644a52fe4e26598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/31468f279470778d16a73b92a644a52fe4e26598"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
