)]}'
{"id":"openstack%2Ftripleo-common~798324","triplet_id":"openstack%2Ftripleo-common~stable%2Fwallaby~I9b6cdad78f834c097226bd1681fa2e13791c33aa","project":"openstack/tripleo-common","branch":"stable/wallaby","topic":"bug/1933850-stable/wallaby","hashtags":[],"change_id":"I9b6cdad78f834c097226bd1681fa2e13791c33aa","subject":"Rename hieradata_files result var","status":"MERGED","created":"2021-06-29 00:26:35.000000000","updated":"2021-06-29 08:58:31.000000000","submitted":"2021-06-29 08:46:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798324-bug/1933850-stable/wallaby","cherry_pick_of_change":798343,"cherry_pick_of_patch_set":1,"meta_rev_id":"0eaaee8ccb58d712d3b6ded9f9c1babf51c38188","_number":798324,"virtual_id_number":798324,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-29 08:46:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-06-29 03:31:28.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"recommended":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2021-06-29 08:52:19.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2021-06-29 03:57:31.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-29 05:46:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2021-06-29 01:07:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":1,"date":"2021-06-29 03:57:31.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 00:26:35.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-29 01:07:43.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"CC"},{"updated":"2021-06-29 01:07:50.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2021-06-29 03:20:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 03:31:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 03:57:31.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2021-06-29 05:46:10.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2021-06-29 08:52:19.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"}],"messages":[{"id":"a39fbf9eca60ef4f4f6c9deff1e671b062853908","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-29 00:26:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0eb52a070046d45fbeb097edfc631c4fd92cdfc8","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-06-29 01:07:43.000000000","message":"Patch Set 1:\n\ntested it in testpatch[1] tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-with-pre against master patch - this works and job is green[2] with this patch\n\n\n[1] https://review.rdoproject.org/r/c/testproject/+/31954\n[2] https://logserver.rdoproject.org/54/31954/6/check/tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-with-pre/0a02ca1/","accounts_in_message":[],"_revision_number":1},{"id":"1e626842ab00da31173e136791fa0db5750b88bd","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-06-29 01:07:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2c7c88017f82704ab3f7f52e26e6347ed28a01a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 03:20:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46e8c42026fb423f8268ac3eb9e80ef3 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e37175e79bcb4f9da086d23b1befc0b2 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/370e33af83a946e59647fe9ecef3d8d1 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66e124791d844933a0a013611c051c94 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71cd30384a044a90916e2e0c4e955611 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d085e970c8a4e93a10503bd78836b8f : SUCCESS in 5m 54s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/feff60e96d044c1793571d2fa6d9faaa : SUCCESS in 2h 42m 43s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/63ed366ccb9c4654ae86d6c1b37ba2c0 : SUCCESS in 1h 38m 55s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5804b9e6484d4b0a877f3b66b70b7e17 : SUCCESS in 58m 12s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e37f7063771c43aa941e069d03b56604 : SUCCESS in 1h 15m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/40daecc9a58041bebdf78d26cc66d402 : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f86850809f24914beb43f4673a5a78c : SUCCESS in 6m 03s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/87a691a369b04cf3a4119052463578b4 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9543cf1d693196e555fd50d1aadfad08fdadf1c5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-29 03:31:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/57775a1d91b14732a61d3206a1e73c1b : FAILURE in 2h 19m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b4347297d6473600199841f70b8f9563f766d07d","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2021-06-29 03:57:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bc488166515f8121f2d398620548fb8fd871ed47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 03:57:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1822ed9807e9cf641bf7e04f8e06ac7c4f479fde","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-06-29 05:46:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2c19c842e558f4d32829b22318fa664d85655a9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 08:46:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be25dfca88e24294acb55b7a3fc7eb4b : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/955ee6a8769b4aefbc8a5411d7010e9b : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/caa2b017330842418f138857bb770762 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3e276a5dff449492d0b4c3559167f9 : SUCCESS in 6m 51s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f5ff59fa300448359b4c4ed80adaf112 : SUCCESS in 2h 50m 21s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/14501463ea6c498781c301e02a6a7d50 : SUCCESS in 1h 34m 24s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ede6022d0bf94ec5afffa28a9c0753be : SUCCESS in 1h 02m 09s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/91b199935e22492ca714877cdf284130 : SUCCESS in 1h 18m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7bd4b1dc03ad411796c6269c9e947b7b : SUCCESS in 7m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fc7aa6dec86845a9868cd4a2f34b9758 : SUCCESS in 6m 38s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/2693ac0af43a4ac39235bac3488bb412 : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e214303e5d899954c8dcce82f717ca68f33c9820","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 08:46:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4c804f45e1c8722a4e6b1025dcea1885f2699ad","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2021-06-29 08:52:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0eaaee8ccb58d712d3b6ded9f9c1babf51c38188","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 08:58:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c827d29276844222baf95e30c17ddebc : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6b493303955911c9aa1568e3e0dbaaff250b161","revisions":{"d6b493303955911c9aa1568e3e0dbaaff250b161":{"kind":"REWORK","_number":1,"created":"2021-06-29 00:26:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/24/798324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/24/798324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/24/798324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/24/798324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/24/798324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/24/798324/1"}}},"commit":{"parents":[{"commit":"277801289e85fcb4b3c00148597b8311fbeb4d22","subject":"Merge \"Refactor enabled_services for roles\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/277801289e85fcb4b3c00148597b8311fbeb4d22"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-28 15:18:11.000000000","tz":-360},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-29 00:26:35.000000000","tz":0},"subject":"Rename hieradata_files result var","message":"Rename hieradata_files result var\n\nThis var is used internally in tripleo-ansible tripleo_hieradata role so\nthe check result is overwriting the variable used to generate hiera.yaml\n\nCloses-Bug: #1933850\nChange-Id: I9b6cdad78f834c097226bd1681fa2e13791c33aa\n(cherry picked from commit 2118998220805bc34e99dc5875bc6c711939fa5c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d6b493303955911c9aa1568e3e0dbaaff250b161"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d6b493303955911c9aa1568e3e0dbaaff250b161"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
