)]}'
{"id":"openstack%2Ftripleo-common~798325","triplet_id":"openstack%2Ftripleo-common~stable%2Fvictoria~I9b6cdad78f834c097226bd1681fa2e13791c33aa","project":"openstack/tripleo-common","branch":"stable/victoria","topic":"bug/1933850-stable/wallaby-stable/victoria","hashtags":[],"change_id":"I9b6cdad78f834c097226bd1681fa2e13791c33aa","subject":"Rename hieradata_files result var","status":"MERGED","created":"2021-06-29 00:31:15.000000000","updated":"2021-07-02 01:10:38.000000000","submitted":"2021-07-02 01:09:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798325-bug/1933850-stable/wallaby-stable/victoria","cherry_pick_of_change":798324,"cherry_pick_of_patch_set":1,"meta_rev_id":"b6800076b92e4bd8b93601db9dc072cd75061e89","_number":798325,"virtual_id_number":798325,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-02 01:09:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-30 19:29:34.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"recommended":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"all":[{"value":1,"date":"2021-06-29 08:52:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":2,"date":"2021-06-29 13:05:59.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-29 09:43:30.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-30 11:49:03.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"CC":[{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 00:31:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-29 01:07:36.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"CC"},{"updated":"2021-06-29 04:41:37.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 04:41:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 08:52:49.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2021-06-29 09:43:30.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-06-29 13:05:59.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"}],"messages":[{"id":"7257f52ec2bda53f7a245e49d03a40a4667bcba6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-29 00:31:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"df4d6d0937f5828d9d358b3d493d82ae34ed191f","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-06-29 01:07:36.000000000","message":"Patch Set 1:\n\ntested it in testpatch[1] tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-with-pre against master patch - this works and job is green[2] with this patch\n\n\n[1] https://review.rdoproject.org/r/c/testproject/+/31954\n[2] https://logserver.rdoproject.org/54/31954/6/check/tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-with-pre/0a02ca1/","accounts_in_message":[],"_revision_number":1},{"id":"052675b23ea143338568e4241491c500ba65ba41","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-29 04:41:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4e91ef460ab94141991b297730dabfa0 : FAILURE in 3h 09m 16s","accounts_in_message":[],"_revision_number":1},{"id":"95d62a930f944682f042ec61d5c3f1bb860aacd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 04:41:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/592a3faae69d4aa4810a68377733f55a : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d671652eb84b47acfbb703e394e910 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2837a70f4f144bc3b0b467f875ca2b64 : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/519b0be7d31b44b3aaee1c7817939660 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e369e15ca174549ab5440a4ac237432 : SUCCESS in 6m 49s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d9fbc2f6ed10459999a2cb4033212783 : SUCCESS in 3h 58m 08s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6866a49b4ab74454bb531a610aa9e65c : SUCCESS in 2h 16m 42s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/49a04a1d45d9450a9ed30daf562d5ed9 : SUCCESS in 1h 23m 32s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/36d08bfc841d4bd5bcd413829b0df871 : SUCCESS in 1h 54m 54s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/a173634a7b0a4af8ab272c5f4d595b7b : SUCCESS in 4h 08m 32s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/5e63f3150bab48b6bd9a3bf147368959 : SUCCESS in 2h 33m 46s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/e5fd6b4c91494e2cbf5198b4a2f5dd00 : SUCCESS in 1h 44m 36s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/d786873f3efb4349be690c93c18280fe : SUCCESS in 2h 05m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/a587e2c63b7849d1b9d70f01bf0501b1 : SUCCESS in 2h 10m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5e2f0c0a179043f3b7c108e767613843 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b2c34e5c53f34408ae7ec1a58878e521 : SUCCESS in 7m 26s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/ab6beb8d62ef45848b6caae6d6b45cc3 : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"4f81fd932a6f8a2d6b2c0e439808fb050e553e25","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2021-06-29 08:52:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3c868c976cdbfbc917b2b4c8401339dc6f7e38c8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-29 09:29:31.000000000","message":"Patch Set 1:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":1},{"id":"c1a651e218f0b3a29a0128af067f007785a2dfa6","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-06-29 09:43:30.000000000","message":"Patch Set 1: Code-Review+2\n\nwaiting for rdo","accounts_in_message":[],"_revision_number":1},{"id":"1c2f57cf29af9e7090cbff0bc7e15cee5018a1dc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-29 10:39:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/None : NODE_FAILURE in 0s","accounts_in_message":[],"_revision_number":1},{"id":"225143a12286ccfede6270845b9e44fef1a57bb9","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2021-06-29 13:05:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1e266261c7bd697a27f3bb784a43c1118dbf6218","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-06-30 11:49:03.000000000","message":"Patch Set 1: Workflow+1\n\nnode failure isn\u0027t linked to the patch, and all backports are merged. Let\u0027s close this issue.","accounts_in_message":[],"_revision_number":1},{"id":"9d174f2d57ba5842313b1782015ba97a0acfa8a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 11:49:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69268e768e111ec48353325a43365f7a21b2b396","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 16:05:13.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a0a8316e8c44a87b653cd31ee7fcfeb : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/503664f97c0d4155a918974f3240784a : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3271df8820b4723874e19c2d36d2766 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c986f36c07d14fb48b6826199a7693c7 : SUCCESS in 6m 05s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a97cff8154ab47729e43ece0f1f40541 : SUCCESS in 4h 10m 35s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b3d25404e9c14b80ba863ad4adac7e31 : SUCCESS in 2h 19m 12s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8eb1fa422e0942e5b5abb4282ce20a91 : SUCCESS in 1h 39m 17s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2c814ddcade74b80aeb38dbc75b702d3 : SUCCESS in 1h 39m 16s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/c9ade42529b04e4a812344f2871fd522 : FAILURE in 42m 15s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/28347d7c0a9e4872a22ef4fb530adf0c : SUCCESS in 2h 19m 05s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/4496cd500cfd472da2c7ddbd4f7eba4e : SUCCESS in 2h 18m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dbc97a78aab640d7bca866a5867a28ba : SUCCESS in 6m 54s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/20e2acbfc36d413e82839992c3405a71 : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9c708703762aacd4ad422ef1b553be77eca1f5f4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-30 16:10:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9db42daa6f3174bb1277ed85442c82ac626d8bc2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-30 19:29:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/cdebeafdf4b442a6802e1051e9b7403b : SUCCESS in 2h 59m 53s","accounts_in_message":[],"_revision_number":1},{"id":"4986daa1b1c648633a7153beab13a947feffb6ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 20:28:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bd279d26ddd433dbf98dae375abadf6 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4af3045741614dcf922ceb21d84fd288 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53802198c66743be93647c85bde69082 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2e45c2aac444f2692874824ec0104b8 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f29d653970bc4342b2de5acb1e6d76af : SUCCESS in 5m 29s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3e1d123d2f5147069b9a0d894298856b : SUCCESS in 4h 03m 28s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/7f3a20125e0449a9a124bc5af20cebc1 : SUCCESS in 2h 21m 08s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/46418cda1be74cf68c855988f86b632b : SUCCESS in 1h 31m 46s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/227924d0210743789aee685ed223b117 : SUCCESS in 1h 43m 30s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/47bf7e22791148c8b37d041cff8a426b : SUCCESS in 3h 18m 10s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/16e011655a9d431da347853a1142c405 : FAILURE in 1h 06m 25s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/e6cffd4905ad4a9aa79d6d1193b74be5 : SUCCESS in 1h 33m 39s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/adfc2032e0da468989f179ef6a7f48e6 : SUCCESS in 2h 20m 17s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/88667343ed1f48feb606d57abcbb5ed6 : SUCCESS in 2h 18m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/12f2550d552b48cb9d57f3e3620bb7b3 : SUCCESS in 5m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/921dfb2fbbb5445196ab8d5a2c8fbb62 : SUCCESS in 5m 55s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/f0e32321210c489d8100a9532049818a : SUCCESS in 5m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f5ac029e5fb5f033996e58c4516eb96c14f133c3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-01 00:02:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f53dd0edb7dfbbd67c5cf2c2eb4f7910ac6965d6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-01 03:38:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/23e0c869ec2b44518a3e458e383c6c04 : SUCCESS in 2h 59m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d162203719c71df1cb9c656fcd5f009eacfec3ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 06:18:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d396754512d4bcbb62b28dafd99588a : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b4d8812ba74c4fa2c1479f053e80b5 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6931b5a96cc148d1b39fc2f01d6ed5b4 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8220be681b2641648f760c69b96ef5be : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ac38b41914848f79c5ced2154b53c92 : SUCCESS in 5m 03s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f36ad4de50394ccdba44d552269746bf : SUCCESS in 6h 05m 44s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0a2d1afc0524486498dd0b6f3c373640 : SUCCESS in 2h 31m 31s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bff91693b95b4e2f9afde03bdfcea0bd : SUCCESS in 1h 48m 27s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/eb255403c750469095d860ebb8668946 : SUCCESS in 1h 42m 53s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/60f932af7d034c55b1bf851797277b41 : SUCCESS in 5h 41m 27s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/21b967ba62784e0580a94ae62158c8a3 : SUCCESS in 2h 32m 30s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/d2ba3fd966f741f39b34c32230357513 : SUCCESS in 2h 04m 15s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/9385e27c23844474a7a1deaec26cdb50 : SUCCESS in 2h 13m 17s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/3c0acb75289f44f6ba33fbaff5434ab6 : SUCCESS in 1h 59m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/80e6e89d695d43ed8130d83ffd8d36f0 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3af8bbabe6404aceb060b48b69055891 : SUCCESS in 5m 13s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/552f88e6c3b44a07b981f760fdef934e : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":1},{"id":"8fd0159cf276bcde18fc49be828a4fdbc29f9120","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 06:19:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6b13581aa6189797017d8a711d84ff368ebc31c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 11:03:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a4a1357ffc34e258df8e2b7c056ce97 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b90799d290db47cb90b18e6b2f7d7b87 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d09432cdd574741bf036e533c353eea : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7894dcdc8de34883999da40e697d15b9 : SUCCESS in 5m 10s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d8cd62116a954d729f9bffb23e2a1d50 : SUCCESS in 3h 46m 20s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e574f71ba68141fd8a65a2765cca7a6b : SUCCESS in 1h 58m 28s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/02f9a51784654f9dbd588777ec685f53 : SUCCESS in 1h 20m 27s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/575dd755381444868b694b9b1999cc60 : SUCCESS in 1h 28m 59s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/222e27c371f64fd0bf8c4d4d2420216c : SUCCESS in 3h 47m 46s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/7ac94171bc4b434fb869fce2ee54e379 : SUCCESS in 2h 24m 58s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/f91eaf7a5ce745e595f0d65acf6ce0f9 : SUCCESS in 1h 37m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/18b0964d7b7f4c96a7fd9b45aed835d3 : FAILURE in 1h 47m 37s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/f0fa192286c64c1ea70c9f4f375dc25e : FAILURE in 1h 40m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a83441cb4bff49eebdc6fe207a48ee45 : SUCCESS in 5m 59s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/0bb67caf88a946629aa4aa4c64d4f712 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7172315d76b3e0597d696309ea3343c8a7e88a01","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-01 14:08:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e3f7f4243283f5926d8ca7027e46ce69048dd4b5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-01 17:55:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a8d611c08e6f4ee690be6043608688cf : SUCCESS in 3h 30m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d345517a6e1fc5096cb3b1c60ba342ae46a342e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 17:58:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce1e7c08fda24c9ea866cad6ad0c8092 : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b693858fe01e4a118af4524496a9287b : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7703a3c2e2234a1d87c75cb119d2924e : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56c165d9e65d452fb3d5dba8063ffdd6 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/713f6c1ab3854c9598d100b18bbdd8dc : SUCCESS in 4m 54s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7f3ee92403c048198c69b793e0c96ff9 : SUCCESS in 3h 42m 43s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/84da68ee3e704c2c88d9231a178ef5f7 : SUCCESS in 2h 06m 14s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/179cf7aeadd3417aa06027122e9b3ffd : SUCCESS in 1h 22m 07s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5be565f9b88a48999c8479f8bb935ca7 : SUCCESS in 1h 29m 59s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/32c827dfc1ec4cffb762cf2168b63c64 : SUCCESS in 3h 48m 07s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/54e61e71f3254c0081062190cf7c6a8c : SUCCESS in 2h 26m 28s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/11f6a62df7aa4058a590b559847ab2d2 : SUCCESS in 1h 36m 40s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/806ddc710e2146aeb0419d4e9d5c77b5 : SUCCESS in 2h 03m 58s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/63941084f44144708ddf0ce2e8960e0c : SUCCESS in 1h 57m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/324292dfc85d47b08e22ae98a8904b54 : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e327c1afae4441892c2931c1cdf4b68 : SUCCESS in 5m 11s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/c2aedf9ff69e46ad9d75300f77619339 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"e55e2cf84b3a288c40472a5ffaf585f94460fa0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 19:26:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4eb6bc67bfaef27b70d034c49a78f52e1c52f958","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 01:09:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/038bd94363aa4334b51bc637955424e5 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0269b8face0746a594c7f5504375afca : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8318e60d14b34ba1981836a74c79736c : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff75219906f34d14ae094a24ff56fc99 : SUCCESS in 5m 11s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a8d6e8c156754369b1ca4f58a0ee1ec6 : SUCCESS in 3h 30m 59s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4fe721bef5164835b2f2b6a10d4610b6 : SUCCESS in 2h 03m 50s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a49b9df1155c47d5a4354770a257de31 : SUCCESS in 1h 12m 07s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ec71be875f71436dbf920c1a9b6d7ee1 : SUCCESS in 1h 29m 47s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/afc7260fb05744d4bde2429acdd896ca : SUCCESS in 3h 27m 53s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/e7342a43ac6747e1be40283131ba25a4 : SUCCESS in 2h 18m 57s\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/f75ecd33701045e08fe5dd4b4e841160 : SUCCESS in 1h 34m 57s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/ab968741d3f74741b8020d940b976825 : SUCCESS in 2h 01m 58s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/b7d6a3f2332a444a9298c33d726e87e1 : SUCCESS in 1h 55m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b672449352064b10b9a72d687788a0f3 : SUCCESS in 5m 20s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/56947ad1cd8e46c6a8e87ba9bf81a29d : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0a3decf6e8a3d9edb9c65e3e6ad9205caf4a2867","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 01:09:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b6800076b92e4bd8b93601db9dc072cd75061e89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 01:10:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/124ef802dd8d40179ad5f19dc5130687 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7e69815c4ebaa84b2920f0603e437ec984cd83e","revisions":{"e7e69815c4ebaa84b2920f0603e437ec984cd83e":{"kind":"REWORK","_number":1,"created":"2021-06-29 00:31:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/798325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/25/798325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/25/798325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/25/798325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/25/798325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/25/798325/1"}}},"commit":{"parents":[{"commit":"f7dabb4625322cbe5b01cb89931faa56f72d0e8e","subject":"Move rabbitmq healthcheck to something more modern","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f7dabb4625322cbe5b01cb89931faa56f72d0e8e"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-28 15:18:11.000000000","tz":-360},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-29 00:31:15.000000000","tz":0},"subject":"Rename hieradata_files result var","message":"Rename hieradata_files result var\n\nThis var is used internally in tripleo-ansible tripleo_hieradata role so\nthe check result is overwriting the variable used to generate hiera.yaml\n\nCloses-Bug: #1933850\nChange-Id: I9b6cdad78f834c097226bd1681fa2e13791c33aa\n(cherry picked from commit 2118998220805bc34e99dc5875bc6c711939fa5c)\n(cherry picked from commit d6b493303955911c9aa1568e3e0dbaaff250b161)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e7e69815c4ebaa84b2920f0603e437ec984cd83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e7e69815c4ebaa84b2920f0603e437ec984cd83e"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
