)]}'
{"id":"openstack%2Ftripleo-common~798326","triplet_id":"openstack%2Ftripleo-common~stable%2Fussuri~I9b6cdad78f834c097226bd1681fa2e13791c33aa","project":"openstack/tripleo-common","branch":"stable/ussuri","topic":"bug/1933850-stable/wallaby-stable/victoria-stable/ussuri","hashtags":[],"change_id":"I9b6cdad78f834c097226bd1681fa2e13791c33aa","subject":"Rename hieradata_files result var","status":"MERGED","created":"2021-06-29 00:31:47.000000000","updated":"2021-06-29 14:24:31.000000000","submitted":"2021-06-29 14:22:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798326-bug/1933850-stable/wallaby-stable/victoria-stable/ussuri","cherry_pick_of_change":798325,"cherry_pick_of_patch_set":1,"meta_rev_id":"304766e245c7ed7548ebde047540e0e8f69e3e87","_number":798326,"virtual_id_number":798326,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-29 14:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-29 03:40:20.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"recommended":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"all":[{"value":1,"date":"2021-06-29 08:52:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-29 01:06:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"value":2,"date":"2021-06-29 09:43:19.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"value":1,"date":"2021-06-29 09:43:19.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"reviewers":{"REVIEWER":[{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-29 00:31:47.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-29 01:06:34.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"CC"},{"updated":"2021-06-29 01:06:43.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2021-06-29 03:40:20.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 04:00:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-29 08:52:53.000000000","updated_by":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"reviewer":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"state":"REVIEWER"},{"updated":"2021-06-29 09:43:19.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"}],"messages":[{"id":"1503abd661c588020055b517493bf041c5e39181","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-29 00:31:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"320dd32ed37ba2772c60fd1a51801ebc49ffdfeb","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-06-29 01:06:34.000000000","message":"Patch Set 1:\n\ntested it in testpatch[1] tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-with-pre against master patch - this works and job is green[2] with this patch\n\n\n[1] https://review.rdoproject.org/r/c/testproject/+/31954\n[2] https://logserver.rdoproject.org/54/31954/6/check/tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001-with-pre/0a02ca1/","accounts_in_message":[],"_revision_number":1},{"id":"25849fa9ee9ff5579702baacc391001b2f1e6c30","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2021-06-29 01:06:43.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"13fa3827f298cf77d52acc38b58c3d084d27f3d7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-06-29 03:40:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8753c5b145c34d75ab67d6b1400b1d42 : SUCCESS in 3h 04m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ff1c1f9e5ae287fca646a8435289b88a97ac4b51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 04:00:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51336b39e85e4c6ca93b7423977ff7ec : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58717790ff814fedb60d739ac2400913 : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/074b3a8d7b6c4c2fb86962b68d1bb6fb : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/af861f228be840289c60a000e5ecbd4b : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cbc29c67a77436dbe6d48cb12d782de : SUCCESS in 5m 35s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f15abd275ab142808ab8e84c84040649 : SUCCESS in 3h 20m 34s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/29c325d4aeda4ad4ab61f440f934da23 : SUCCESS in 2h 06m 51s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/436455169d6b4f52bb85ffbf6ced63e0 : SUCCESS in 1h 03m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/408cff2d69d6486686f1134cbc0e6e69 : SUCCESS in 1h 28m 01s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/06ee5f2b0c854eb7b2cf10adb8761ed9 : SUCCESS in 3h 18m 31s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/0b440267a5db4198b48e9c3cc7b65e76 : SUCCESS in 2h 08m 14s\n- tripleo-ci-centos-8-undercloud-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/58b54710fba442c7be4704c2ddd4b5e1 : SUCCESS in 1h 51m 02s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/c709ad9340374cfb826490a61a70200d : SUCCESS in 2h 14m 22s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/7ea054ec02d34d0a9e3794d5b3a9fb7e : SUCCESS in 1h 52m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e4777667e8f44b2d82b732e1a16ef201 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":1},{"id":"ed74ec2c9de679a55091184c86294a538d03806b","author":{"_account_id":6816,"name":"Jesse Pretorius","email":"jesse@odyssey4.me","username":"jesse-pretorius"},"date":"2021-06-29 08:52:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"145771a26a077e747af2535cf8a9a3d890e10d9f","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-06-29 09:43:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"68d2353afe0b236438865672de1785135afd73e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 09:43:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9b69b176380e39375ca0058149c9dfaa172df1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 14:22:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7891bf75d70f4f26bcb825163bea04a9 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09f22a5847304cfdb3cefeb59a397f93 : SUCCESS in 7m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e436ab764b544840b756aca90008a68a : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb8ad0ce80cd4393931ce8c340226432 : SUCCESS in 7m 24s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cd3f81e94b1c49139f9c608d12191d34 : SUCCESS in 4h 34m 25s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/802947fa0124410e8bd12cf39b56e79c : SUCCESS in 2h 19m 06s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f0bcedef336c44a58ac1465bb227653c : SUCCESS in 1h 16m 32s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2ad01152518d4c4a8cdcb2ca9fd96de2 : SUCCESS in 1h 40m 21s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/95f0c1c3469b44ec8e445c7fe8975b52 : SUCCESS in 4h 33m 26s\n- tripleo-ci-centos-8-standalone-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/96375aea1e7240cbb19d9587eb55a94a : SUCCESS in 3h 09m 12s\n- tripleo-ci-centos-8-undercloud-upgrade-ussuri https://zuul.opendev.org/t/openstack/build/2643a0ed988e43cda742385a3fe5c322 : SUCCESS in 2h 20m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-ussuri https://zuul.opendev.org/t/openstack/build/cbea9eb7455440678d0d2a0a71fec7c4 : SUCCESS in 2h 17m 51s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-ussuri https://zuul.opendev.org/t/openstack/build/0c65a6c9ae034a229f9f86df31f21b22 : SUCCESS in 2h 09m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9987e87236854a568dfe26eaffa9ee56 : SUCCESS in 8m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6ee9cd86037a519e887a035649a322abcb36cfdd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 14:22:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"304766e245c7ed7548ebde047540e0e8f69e3e87","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 14:24:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a000ac7c800349bebce5b3116065a1c4 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"66a5cc7e768ee9943103078fd7b0d70aae2f4f5d","revisions":{"66a5cc7e768ee9943103078fd7b0d70aae2f4f5d":{"kind":"REWORK","_number":1,"created":"2021-06-29 00:31:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/798326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/26/798326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/26/798326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/26/798326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/26/798326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/26/798326/1"}}},"commit":{"parents":[{"commit":"6ac7803b8a50409de8bad96773b2ed959d7df084","subject":"Move rabbitmq healthcheck to something more modern","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6ac7803b8a50409de8bad96773b2ed959d7df084"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2021-06-28 15:18:11.000000000","tz":-360},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-29 00:31:47.000000000","tz":0},"subject":"Rename hieradata_files result var","message":"Rename hieradata_files result var\n\nThis var is used internally in tripleo-ansible tripleo_hieradata role so\nthe check result is overwriting the variable used to generate hiera.yaml\n\nCloses-Bug: #1933850\nChange-Id: I9b6cdad78f834c097226bd1681fa2e13791c33aa\n(cherry picked from commit 2118998220805bc34e99dc5875bc6c711939fa5c)\n(cherry picked from commit d6b493303955911c9aa1568e3e0dbaaff250b161)\n(cherry picked from commit e7e69815c4ebaa84b2920f0603e437ec984cd83e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/66a5cc7e768ee9943103078fd7b0d70aae2f4f5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/66a5cc7e768ee9943103078fd7b0d70aae2f4f5d"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
