)]}'
{"id":"openstack%2Ftripleo-common~799593","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~I55905cb299ae9e35b605f0a9219177f809d6919e","project":"openstack/tripleo-common","branch":"stable/train","topic":"healthcheck/port/v6-stable/wallaby-stable/victoria-stable/train","hashtags":[],"change_id":"I55905cb299ae9e35b605f0a9219177f809d6919e","subject":"Add missing IPv6 support for healthcheck_port","status":"MERGED","created":"2021-07-06 16:35:47.000000000","updated":"2021-07-08 16:35:26.000000000","submitted":"2021-07-08 16:32:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799593-healthcheck/port/v6-stable/wallaby-stable/victoria-stable/train","cherry_pick_of_change":799592,"cherry_pick_of_patch_set":1,"meta_rev_id":"3a9f74e1e8d073c9e9921114f6c827ada6555e34","_number":799593,"virtual_id_number":799593,"owner":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-08 16:32:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-08 09:33:34.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-08 06:00:48.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-08 08:33:13.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-06 16:35:47.000000000","updated_by":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2021-07-06 20:20:45.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-06 20:56:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4cc7d7f3c9461bc395cf72524ee50b53e447293d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2021-07-06 16:35:47.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"bef386e5ef0a80a5f84db6db7a423499db891927","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-06 20:20:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/92bd06f13841483daa64b9edbaef488f : SUCCESS in 3h 42m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ff72c34827fa7c387c60a985162d7ee1746734db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 20:56:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63d48d3c9be046c7bf110d52e68afcda : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3ffe334617f4b1090623b8328cc1245 : SUCCESS in 5m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0d06226965f4b3eb3817d853f170ecb : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e52b8885e3944891bcc97ab00f4c412a : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/774726d0c81c41628b361e85d936229c : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/606272cc7f4140d89e90d0625d9e0d63 : SUCCESS in 4m 49s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9c4229cef816460b9f06f571325868e2 : SUCCESS in 3h 42m 50s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/b7e7cc26d9674d76b58c6dd541b6fcaa : SUCCESS in 4h 14m 59s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/06802c0d26ac4873a1d89765c86f7f83 : SUCCESS in 2h 11m 40s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/954cb5660f88407c87863117c85e0e69 : SUCCESS in 2h 25m 02s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1913459908c2487089a82b6bf4210bc5 : SUCCESS in 1h 34m 57s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/010ee4c2a16f4d5792364901684b1db6 : SUCCESS in 1h 04m 53s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3afad161b1744079a04135c6392bcb98 : SUCCESS in 1h 20m 00s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/7d9c9ec302ca429cac37473d6d9949cf : FAILURE in 46m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0d9765e47c4f4e1ebc4d400460ddfc70 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"9683a19692b388e3e5024a5d38a9ec726e126f3e","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2021-07-07 08:30:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"970225ba732fcf3462f94810cc15eea4344d9a89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2021-07-07 10:59:34.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"75fbd96bf6b86d3365c8e50c9a89092cece9c1d4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-07 12:05:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0b7a455e07a2454dbfe3c78a5721c833 : FAILURE in 56m 16s","accounts_in_message":[],"_revision_number":2},{"id":"d1f5d736cc95471fe60857b3089be7b1067ac2b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 14:34:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb595e2a4a0846edbdb13b40df2ef2fe : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7b16fe6f0b34f619ff19f59700a3d95 : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c84704adb94047469ab5e694da03e477 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/553710e9d4814b3faa91bf894c94a69f : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6601d1b031334ee0b626eabb02ff63db : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb2e57eb9c23407e83b480d3ac5ed619 : SUCCESS in 5m 00s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2d6ca0adac244581ae7a094630beaf9c : SUCCESS in 3h 08m 18s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/c3906b970f894b5ea5a3e8c44ba59dbb : SUCCESS in 3h 33m 52s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3cdc0ec6ab4a4842b639b7a8838d4dd6 : SUCCESS in 2h 05m 12s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/ce77c5122cca48b2bcd5aa8b5068c8e9 : SUCCESS in 2h 12m 54s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/71739ac4f7294600b95a95cc1355b02b : SUCCESS in 1h 25m 44s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/79667d3857a347db885e88470039e9d3 : SUCCESS in 45m 14s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6f1fb556cb754904a0bb09dc4597958b : SUCCESS in 1h 13m 23s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/9ad8ccba49cb49089909cb9523c5c238 : SUCCESS in 1h 54m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/695a08ed3dcc483f96f9287c3d64b954 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":2},{"id":"8b6717612e788fb943fe19103b8bd95a51037eb5","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-07-08 06:00:48.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"24f84dedcc3da3d5afb8caee747dba476455f2d9","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2021-07-08 08:33:13.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3be7fb0f10bcd7335045f7179d42f562e80cab2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 08:33:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"698d0fcd2513bb8796c1546a6b5712c8df87b946","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-08 09:33:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0758adc1afae4a898d9884dbc7ba4c39 : SUCCESS in 3h 20m 46s","accounts_in_message":[],"_revision_number":2},{"id":"6db85d3c9edc05cc176eefd8826e10e3ae01e0d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 10:06:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/213276fd2b69478698083b742c6dbf29 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7493084610e94dac8cd19afa9e7b503f : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/842b351060a84a8599779ee8b03d772b : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f3afd08911c45afa6bd5958c815c9dc : SUCCESS in 6m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6f79b467ebc947e9ba328067428b0c22 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180d18dd93d348f28d73cf5e2ffa6b28 : SUCCESS in 5m 43s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a94bccaa22774bb0a131ecf3ec8d2aea : SUCCESS in 3h 15m 23s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/681dfe66e3cb493ab73a8b9654bdce99 : SUCCESS in 2h 05m 12s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/3b93411414c24da9b5b745fcdb050b09 : SUCCESS in 2h 06m 23s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/bbc45d21d8b648ac8342e3a17550a7df : RETRY_LIMIT in 9m 01s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e47cad87d9be402e93fcf4009d4d4afa : SUCCESS in 1h 28m 06s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/391b89ba57b940fa8b9ec9dcf7c57db4 : SUCCESS in 1h 01m 23s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/47f150491156481988542aa4dbd3a75a : SUCCESS in 1h 18m 03s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/e00f01141ba040cc89c82a0a82f1df2b : SUCCESS in 1h 54m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fd139c3266874b8e98a314f9a6d5b6db : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":2},{"id":"97107189168609e05fc7ae04480b70d2b0144dd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 16:32:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcb2924718104645a8e9915d181950f8 : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b37bb8c8afc4d6b970cdc168f518e8d : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76f9cb4522b741a7a2d4c7d9fdaa3e7b : SUCCESS in 6m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/944e960c265249d9b5181aa99bd2d58c : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56918d6e4f974e1d966a144d791c175f : SUCCESS in 6m 16s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/70d0724778a24c18aacb1ce765140e59 : SUCCESS in 4h 10m 51s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/6520183707aa40eeb2793c48d45ecf5c : SUCCESS in 3h 36m 28s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/346ec1693da94fddb21aae36eeed8577 : SUCCESS in 2h 21m 57s\n- tripleo-ci-centos-7-containers-multinode https://zuul.opendev.org/t/openstack/build/c9026ee768614aa3a58ddf9adb7dc047 : SUCCESS in 2h 13m 23s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/98c5f2c72c084ff4bd375698ce67bf7b : SUCCESS in 1h 18m 45s\n- tripleo-build-containers-centos-7-buildah https://zuul.opendev.org/t/openstack/build/4cd58c38bac046b7a3c5685623743084 : SUCCESS in 52m 37s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/71a1aa8f854f4161b9a4e92ac934f9f0 : SUCCESS in 1h 09m 19s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-train https://zuul.opendev.org/t/openstack/build/efb0d13e9be643768ce6f2b382bcd4aa : SUCCESS in 1h 53m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0dafcf9b5a934757adeb7d3a162875ee : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":2},{"id":"c659de89eea7767a81e76e92ecf4ebba2ec2bca7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 16:32:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3a9f74e1e8d073c9e9921114f6c827ada6555e34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 16:35:26.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/764d33ea6d1e4800b77e456a8e00e2ee : FAILURE in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"248622aa5aa13dd9c498e976e368bdd5f5e4008a","revisions":{"e77180718fb1dadaa6d0661892c6f1cd4a2efa6f":{"kind":"REWORK","_number":1,"created":"2021-07-06 16:35:47.000000000","uploader":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"ref":"refs/changes/93/799593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/93/799593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/799593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/799593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/799593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/93/799593/1"}}},"commit":{"parents":[{"commit":"746a8d74ac022df30eaa3b17a4450b5d67ea3f82","subject":"directly use files instread of tcib_gather_files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/746a8d74ac022df30eaa3b17a4450b5d67ea3f82"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-07-06 11:09:40.000000000","tz":120},"committer":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2021-07-06 16:35:47.000000000","tz":0},"subject":"Add missing IPv6 support for healthcheck_port","message":"Add missing IPv6 support for healthcheck_port\n\nWith change I64776992a7e457781aa8ddaba359ef085d4cb77d, the IPv6 network\nwas overlooked, leading to issues in some cases with composable roles\nand IPv6 network.\n\nChange-Id: I55905cb299ae9e35b605f0a9219177f809d6919e\nCloses-Bug: #1934772\nResolves: rhbz#1979524\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e77180718fb1dadaa6d0661892c6f1cd4a2efa6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e77180718fb1dadaa6d0661892c6f1cd4a2efa6f"}]},"branch":"refs/heads/stable/train"},"248622aa5aa13dd9c498e976e368bdd5f5e4008a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-07 10:59:34.000000000","uploader":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"ref":"refs/changes/93/799593/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/93/799593/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/799593/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/799593/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/93/799593/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/93/799593/2"}}},"commit":{"parents":[{"commit":"746a8d74ac022df30eaa3b17a4450b5d67ea3f82","subject":"directly use files instread of tcib_gather_files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/746a8d74ac022df30eaa3b17a4450b5d67ea3f82"}]}],"author":{"name":"Cédric Jeanneret","email":"cjeanner@redhat.com","date":"2021-07-06 11:09:40.000000000","tz":120},"committer":{"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","date":"2021-07-07 10:59:23.000000000","tz":0},"subject":"Add missing IPv6 support for healthcheck_port","message":"Add missing IPv6 support for healthcheck_port\n\nWith change I64776992a7e457781aa8ddaba359ef085d4cb77d, the IPv6 network\nwas overlooked, leading to issues in some cases with composable roles\nand IPv6 network.\n\nChange-Id: I55905cb299ae9e35b605f0a9219177f809d6919e\nCloses-Bug: #1934772\nResolves: rhbz#1979524\n(cherry picked from commit 4ea75f40b7d199ead6bb294f73dc6a8c96722ec1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/248622aa5aa13dd9c498e976e368bdd5f5e4008a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/248622aa5aa13dd9c498e976e368bdd5f5e4008a"}]},"branch":"refs/heads/stable/train","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
