)]}'
{"id":"openstack%2Ftripleo-common~801396","triplet_id":"openstack%2Ftripleo-common~master~Ifb46702abe32d79ea097f02a684a2940887c0e23","project":"openstack/tripleo-common","branch":"master","hashtags":[],"change_id":"Ifb46702abe32d79ea097f02a684a2940887c0e23","subject":"Add apache to the designate api container","status":"MERGED","created":"2021-07-19 22:23:29.000000000","updated":"2021-07-24 04:02:29.000000000","submitted":"2021-07-24 03:57:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801396","meta_rev_id":"c2831c91c3969e000dc6b8a22d1e227ee61f9b3b","_number":801396,"virtual_id_number":801396,"owner":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-24 03:57:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-20 01:25:42.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-07-23 21:00:06.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-23 14:08:02.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2021-07-23 21:00:06.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-20 01:18:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-20 01:25:42.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-23 14:08:02.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2021-07-23 21:00:06.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e5f66e2c25f3af6209293968cba3ad2845bdecc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2021-07-19 22:23:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddddcec819017a06d651ee9e03171c8bd2018a59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 01:18:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66da48411a5a43b6afb72668513e5f9d : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1bb62a89044cb99281f667530a567a : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f03f0384c3e14f259e30fd166f3becb8 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d6bdd7fdccc4a76b7f1551784cacc8e : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1ff67a440964b938321118c160a8df9 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0a33c11c8a641edb87f1c748913615c : SUCCESS in 5m 09s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d1e63c176cae499097b8ba54b9d4a587 : SUCCESS in 2h 46m 16s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/42bfc86c72ec45fb93647152d5a2950f : SUCCESS in 1h 32m 25s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/81c6e2dad16d4356be17cac6f1af2544 : SUCCESS in 42m 08s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/e8d1928495b4408f83d57e43c27fff1a : SUCCESS in 57m 40s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/154e3fab1a464b198e6f7c6e57dfffd9 : SUCCESS in 1h 11m 01s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/f2b1089e18a54a92a2b29ba00707412d : SUCCESS in 2h 38m 38s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4637a3672a7d4e869c03b53816fddb0f : SUCCESS in 1h 24m 38s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/ecbb985ce2df480aaf44471125e9d994 : SUCCESS in 1h 18m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c1c113bddbb4e43a713f7a623bfb337 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/387b4ac56d3749b4814e3a063b60afdb : SUCCESS in 4m 41s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/6e6acf4cd14a4139b40e2a653f898321 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d2dd8c52024be682a51fa9ec23f24cbeaf3cba6a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-07-20 01:25:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/124b33ae12c44e6d961dae1e36144bcf : SUCCESS in 3h 01m 41s","accounts_in_message":[],"_revision_number":1},{"id":"ff272d89fb28f389705c716ddc86b8f82faf761f","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-07-23 14:08:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1eb8f652119a525622fa331982ba08b1b00d59fd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-23 21:00:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b213752fb6447bc1842f4f6c3be6ca23bbb9abf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 21:01:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"375bd849c2b91a9dd81ff2c67965dd4358eb32b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-24 03:57:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2721601940a4ab1bacc7d7c6ff02f98 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4a93bd820634effb6f7c54b781fc21b : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddbba698d3b74a49b59b9b0b5a21e277 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f177f864bc9481883ea0220814ac6c8 : SUCCESS in 4m 33s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/bd60237c23634c3aab31ce83f10de05a : SUCCESS in 3h 31m 09s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/891333ff3c7d46d498d1f314b5499a4b : SUCCESS in 1h 51m 14s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/73822da00b9e4a7090e164a9e8799708 : SUCCESS in 52m 44s\n- tripleo-ci-centos-8-scenario003-standalone https://zuul.opendev.org/t/openstack/build/8dab7a8664f74d07989fca20cb34e2cf : SUCCESS in 1h 20m 00s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/980d3dea9f8541aabe0b486a6dc5532a : SUCCESS in 1h 23m 31s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ac343c8bb0e648758587572a6b4158a0 : SUCCESS in 2h 37m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/19a1305a8dce41458d97e451d4da6010 : SUCCESS in 1h 36m 03s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/b35beae26597494cb14b1704bcf68cbc : SUCCESS in 59m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9bd9e2c6b45b47c59ced19a47f16036e : SUCCESS in 5m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e544e3fd3c3a4f35ac164c5b2ea3e942 : SUCCESS in 4m 57s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/b28d02f4911d4612a0870369eb9b8477 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"13a4bc09596b4866bb771b63816fe1a3e2f5f1da","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-24 03:57:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c2831c91c3969e000dc6b8a22d1e227ee61f9b3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-24 04:02:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d22ab1a8a5744658ce89cebd43f3321 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2984f46468e86aea42e1f05c631f42f45b454723","revisions":{"2984f46468e86aea42e1f05c631f42f45b454723":{"kind":"REWORK","_number":1,"created":"2021-07-19 22:23:29.000000000","uploader":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"ref":"refs/changes/96/801396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/96/801396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/96/801396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/96/801396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/96/801396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/96/801396/1"}}},"commit":{"parents":[{"commit":"86d0e65a2905f46eca05cbf4070f97feaa392463","subject":"Fix placement endpoint check when endpoints are not set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/86d0e65a2905f46eca05cbf4070f97feaa392463"}]}],"author":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-07-19 22:25:30.000000000","tz":-150},"committer":{"name":"Brent Eagles","email":"beagles@redhat.com","date":"2021-07-19 22:25:30.000000000","tz":-150},"subject":"Add apache to the designate api container","message":"Add apache to the designate api container\n\nThe designate API should run behind apache.\n\nChange-Id: Ifb46702abe32d79ea097f02a684a2940887c0e23\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2984f46468e86aea42e1f05c631f42f45b454723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2984f46468e86aea42e1f05c631f42f45b454723"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
