)]}'
{"id":"openstack%2Ftripleo-common~806521","triplet_id":"openstack%2Ftripleo-common~stable%2Fwallaby~Id8d99560ccd412014fe1d13faad19c2a114cd6e6","project":"openstack/tripleo-common","branch":"stable/wallaby","topic":"readd_tripleoclient-stable/wallaby","hashtags":[],"change_id":"Id8d99560ccd412014fe1d13faad19c2a114cd6e6","subject":"Re-add tripleoclient and tempest container","status":"MERGED","created":"2021-08-30 13:40:04.000000000","updated":"2021-09-08 09:52:31.000000000","submitted":"2021-09-08 09:50:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"806521-readd_tripleoclient-stable/wallaby","cherry_pick_of_change":801987,"cherry_pick_of_patch_set":3,"meta_rev_id":"4975215645ac25b9b02f9f34d04a924337c3ea00","_number":806521,"virtual_id_number":806521,"owner":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-08 09:50:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-31 09:14:55.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2021-09-08 06:46:30.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2021-09-07 18:46:49.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2021-09-08 05:02:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":1,"date":"2021-09-08 06:46:30.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-30 14:17:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-30 14:53:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-02 06:23:38.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2021-09-02 06:23:38.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-06 06:25:56.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-09-06 06:25:56.000000000","updated_by":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2021-09-08 05:02:21.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"}],"messages":[{"id":"b213db27efc81ceea41ec24f21053b3ac6689ae2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-08-30 13:40:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"33866e7c2ca3bfece23275213a95204f58127f7c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-30 14:17:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9941401a7daf4cc2b9793db943e41767 : FAILURE in 36m 36s","accounts_in_message":[],"_revision_number":1},{"id":"fb8d1a54a1d92d60db7f6fa73b64a62db9b1542b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 14:53:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab9087d1ccd74a889e46eea5f8ae56ac : FAILURE in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a226de1880a4337864d8cf3587fa599 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9e7639e413648a1980dea74f07f8b88 : FAILURE in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b661d40e71c4ed487609d867c97e965 : FAILURE in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4716fd168771458083289edb861c4aba : FAILURE in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde5024e80bd46ff9c20dfb4f737214b : SUCCESS in 6m 44s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/f57b93852378491da1d1817b06acc7c7 : SUCCESS in 1h 11m 18s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7ac4ca739171493fa11991ab7323802e : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98ea7a8a52e2448ba8e0f4726762ac22 : FAILURE in 7m 22s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/83e2d79a63e44dd1af1a7e8d476d37e4 : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":1},{"id":"13ad70d2d043d803033bd9490b85ac381eaf5d61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-08-30 15:48:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad3b1b885b930cbd24559ed0ca056a86e053547e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-30 18:17:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/00c0ba94f6434c939c75bc6b1a747c7d : FAILURE in 2h 26m 50s","accounts_in_message":[],"_revision_number":2},{"id":"3dbac12ab627d295977d325fa81dbf2b94bb03d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 19:02:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d161d1075f44059b4cb4e0590973c5a : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec18e9f8b459437dba0ba49c62225800 : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4bb76bc601545f685494ac98587a1fc : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8812ed23ed8d4801a5d4b5d895544500 : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e28f4dd5346d448eaa9b80e09788a83a : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a349ade0b8b442008774941e2101672e : SUCCESS in 5m 07s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6ce5c4668e514d61bb9f37cc8a6076e3 : SUCCESS in 2h 59m 27s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/0d8021cbab314d978a62017df6f9a290 : SUCCESS in 1h 37m 54s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9e0166fbafef4575b1c7fecc5c9d5df4 : SUCCESS in 1h 04m 10s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/66b25d36b54944729f1212a9c1d0605b : SUCCESS in 1h 18m 52s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/587231f0b4354753936afdc750d1fc6d : SUCCESS in 2h 47m 02s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-wallaby https://zuul.opendev.org/t/openstack/build/be2b4e2aa34347128756e0cc9ff771bc : SUCCESS in 1h 31m 24s\n- tripleo-ci-centos-8-undercloud-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/c93eaa6c3ebe481db533485439151780 : SUCCESS in 1h 14m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fb11330ebe66487b8004e8dd8b87b020 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/daa877a79d984fd4a33fbde0b7c67ed2 : SUCCESS in 6m 36s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/458c8ead71a14805b26fdf7635c99a50 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"9fd8028974be21730f3d49e1b6d64ef0704e06a0","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2021-08-31 06:01:16.000000000","message":"Patch Set 2:\n\ncheck-rdo","accounts_in_message":[],"_revision_number":2},{"id":"926bfdb7af73bef8db6f4bd10a823f1834e619cf","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-08-31 09:14:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/bff30213c6da491f8e1289d6ed151448 : SUCCESS in 3h 06m 35s","accounts_in_message":[],"_revision_number":2},{"id":"99fe450c7d52e3641d63e18dc7b3b5da385ee852","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2021-09-07 18:46:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f0e8a8bcb2fe69d73eed8ffa875899e8eac2680b","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-09-08 05:02:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8a11cbe73be59f44339a9fcf7e71db5bcbde136e","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2021-09-08 06:46:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4501ef01eb513cd38d4b981183eff0bf17ed4e43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 06:47:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c1f6979af1ffb53d58e6a619442721e11c102db1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 09:50:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/852a8600dbc341e5b32e45021664d85b : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88b4baf12baa4fe39763ca2a47ec123f : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2359cdac11c84cd9b8eb267914de9fce : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aba90b9876944dfa211e07ae792f0c1 : SUCCESS in 6m 59s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/956025141ad34bdbba9f0d9bc8e3ab5d : SUCCESS in 3h 02m 41s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/4d3b0f11b74340c2bd16235c1e6673a1 : SUCCESS in 1h 39m 42s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d2063472c2be43e493bfdcefa7ca5edb : SUCCESS in 1h 00m 19s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f5a9f8629408406387dfccbc5727eb79 : SUCCESS in 1h 20m 59s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/afbd5eaa42b4409a82e4eb741da2fd88 : SUCCESS in 2h 46m 55s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-wallaby https://zuul.opendev.org/t/openstack/build/43cc1333c3b94a65b6d2daba64c5c743 : SUCCESS in 1h 27m 48s\n- tripleo-ci-centos-8-undercloud-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/e37d7a29636a44aa99b8c3e484ab56b5 : SUCCESS in 1h 22m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2bf7683ad46d419fae6f40d57e76cf4a : SUCCESS in 7m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/724d1beec4c34dbbbd75543ad48d75fb : SUCCESS in 6m 52s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/b078a06c464a45efb34abce9ffe4b4f4 : SUCCESS in 8m 49s","accounts_in_message":[],"_revision_number":2},{"id":"99aedec506c2b1c38ee20d25b92e95eaac9ad732","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 09:50:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4975215645ac25b9b02f9f34d04a924337c3ea00","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 09:52:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70adf797255245b68eddf1cadb41ea67 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"189ca8cc8c9d996c578a59297d52c108898d7a92","revisions":{"b56d5974de2e9bf6d90ce3cd771f6b807599db76":{"kind":"REWORK","_number":1,"created":"2021-08-30 13:40:04.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/21/806521/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/21/806521/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/21/806521/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/21/806521/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/21/806521/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/21/806521/1"}}},"commit":{"parents":[{"commit":"7a51e01001bd7d538d3ab888e30ae904da2f3b39","subject":"Install Neutron BGPVPN in neutron-server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7a51e01001bd7d538d3ab888e30ae904da2f3b39"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-07-23 07:57:11.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-08-30 13:40:04.000000000","tz":0},"subject":"Re-add tripleoclient and tempest container","message":"Re-add tripleoclient and tempest container\n\nIn the osp-director-operator [1] effort which uses tripleo to deploy\nopenstack using openshift to manage the hardware/vm provisioning.\nTo do so we rely on the tripleoclient and tempest container for\ndeployment/testing. Lets add the containers back to be built.\n\n[1] https://github.com/openstack-k8s-operators/osp-director-operator\n\nChange-Id: Id8d99560ccd412014fe1d13faad19c2a114cd6e6\n(cherry picked from commit ede41284e02d4783e580d639c0ba650b91f98eee)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b56d5974de2e9bf6d90ce3cd771f6b807599db76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b56d5974de2e9bf6d90ce3cd771f6b807599db76"}]},"branch":"refs/heads/stable/wallaby"},"189ca8cc8c9d996c578a59297d52c108898d7a92":{"kind":"REWORK","_number":2,"created":"2021-08-30 15:48:06.000000000","uploader":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"ref":"refs/changes/21/806521/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/21/806521/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/21/806521/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/21/806521/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/21/806521/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/21/806521/2"}}},"commit":{"parents":[{"commit":"7a51e01001bd7d538d3ab888e30ae904da2f3b39","subject":"Install Neutron BGPVPN in neutron-server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7a51e01001bd7d538d3ab888e30ae904da2f3b39"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-07-23 07:57:11.000000000","tz":120},"committer":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-08-30 15:47:58.000000000","tz":120},"subject":"Re-add tripleoclient and tempest container","message":"Re-add tripleoclient and tempest container\n\nIn the osp-director-operator [1] effort which uses tripleo to deploy\nopenstack using openshift to manage the hardware/vm provisioning.\nTo do so we rely on the tripleoclient and tempest container for\ndeployment/testing. Lets add the containers back to be built.\n\n[1] https://github.com/openstack-k8s-operators/osp-director-operator\n\nChange-Id: Id8d99560ccd412014fe1d13faad19c2a114cd6e6\n(cherry picked from commit ede41284e02d4783e580d639c0ba650b91f98eee)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/189ca8cc8c9d996c578a59297d52c108898d7a92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/189ca8cc8c9d996c578a59297d52c108898d7a92"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
