)]}'
{"id":"openstack%2Ftripleo-common~808044","triplet_id":"openstack%2Ftripleo-common~master~I26174f6a450885fd65f8b1af5622d7d68c0aaf40","project":"openstack/tripleo-common","branch":"master","hashtags":["cs9"],"change_id":"I26174f6a450885fd65f8b1af5622d7d68c0aaf40","subject":"Switch to mariadb:10.5 module","status":"MERGED","created":"2021-09-09 09:59:40.000000000","updated":"2021-09-12 22:56:44.000000000","submitted":"2021-09-10 01:57:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808044","meta_rev_id":"3c16457e10ed2b668a1425556c18b896a4a1623c","_number":808044,"virtual_id_number":808044,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 01:57:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-09 18:24:15.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":2,"date":"2021-09-09 15:36:49.000000000","_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2021-09-09 16:22:19.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2021-09-09 20:45:12.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"}],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-09 12:37:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 12:45:36.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"},{"updated":"2021-09-09 12:45:36.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2021-09-09 14:52:45.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2021-09-09 14:52:45.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2021-09-09 15:36:49.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-09-09 16:22:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-09 18:24:15.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4294bda36cd606ec2a9299f9ea8c620ba5ffbf53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-09-09 09:59:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c5a386aa95f60c04afb78353ec332b6ca5930f1","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-09-09 10:00:02.000000000","message":"Hashtag added: csg","accounts_in_message":[],"_revision_number":1},{"id":"95f3abafe3d92f361f04bd2fd2ce98b90a4ff7fd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-09-09 10:00:03.000000000","message":"Topic c9s removed","accounts_in_message":[],"_revision_number":1},{"id":"c5145a6dabf790d403164baec71844a4fbe85f89","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-09-09 10:00:10.000000000","message":"Hashtag added: cs9","accounts_in_message":[],"_revision_number":1},{"id":"9b48fbc0c07dde3955cb249ddf84db45a400b6c1","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-09-09 10:00:12.000000000","message":"Hashtag removed: csg","accounts_in_message":[],"_revision_number":1},{"id":"21578cc6ddf5a95eb46245726cde549de0b0cd1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 12:37:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a1214e92fdf4dc2ac33651cef4c5613 : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa5b4fdd053e463184a85f8120d1bb79 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8f9aa2d71d34bff99abc0eeb8aae826 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/695906ba1e0a4041a3d523e15290870a : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8d9b2c6cbbd41e29081a8f445a6e5f4 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78d770a142d44635b4009a5e7ab10480 : SUCCESS in 5m 02s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9fb1788c59fd40978a0b43b35b0395cc : SUCCESS in 2h 07m 56s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/61fe76f36f4448e7bca1fee5d36851fb : FAILURE in 37m 27s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bcc7d5f774ee44fc96f50c49823ccac0 : FAILURE in 56m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5ca104f984474a58ade8fe44e9707d68 : FAILURE in 37m 21s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/72593b17b38546dc9b599bae17607d06 : SUCCESS in 2h 30m 14s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f81f3663f4924d0b8529704ffdb779ee : FAILURE in 40m 50s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/070c622b9e7e4dd1a6043ccc8ed481a9 : SUCCESS in 59m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8801159bb4014b42aeba810ecf0298eb : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe1852adb6a74b34be0ef37f76cc6736 : SUCCESS in 4m 43s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/736f2ab980424f8eb232bd828201579b : SUCCESS in 7m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d31c787a7042689868d9d7b7cfa120781edeef8d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-09-09 12:43:45.000000000","message":"Patch Set 1:\n\nhttps://0e01ca15cd4d54a0e684-4b29bd9ae378cf0db90ada93d04207ec.ssl.cf1.rackcdn.com/808044/1/check/tripleo-ci-centos-8-containers-multinode/61fe76f/logs/undercloud/home/zuul/undercloud_install.log\n```\nTASK | Create containers managed by Podman for /var/lib/tripleo-config/container-startup-config/step_2\n2021-09-09 11:58:57.718323 |                                      |    WARNING | ERROR: Can\u0027t run container mysql_bootstrap\nstderr: INFO:__main__:Loading config file at /var/lib/kolla/config_files/config.json\nINFO:__main__:Validating config file\nINFO:__main__:Kolla config strategy set to: COPY_ALWAYS\nINFO:__main__:Copying service configuration files\nINFO:__main__:Creating directory /etc/systemd/system/mariadb.service.d\nINFO:__main__:Copying /var/lib/kolla/config_files/src/etc/systemd/system/mariadb.service.d/90-limits.conf to /etc/systemd/system/mariadb.service.d/90-limits.conf\nINFO:__main__:Deleting /etc/my.cnf.d/galera.cnf\nINFO:__main__:Copying /var/lib/kolla/config_files/src/etc/my.cnf.d/galera.cnf to /etc/my.cnf.d/galera.cnf\nINFO:__main__:Copying /var/lib/kolla/config_files/src/root/.my.cnf to /root/.my.cnf\nINFO:__main__:Writing out command to execute\nINFO:__main__:Setting permission for /var/lib/mysql\nsudo: unable to send audit message: Operation not permitted\nsudo: unable to send audit message: Operation not permitted\ninvalid command name \"Y/n\"\n    while executing\n\"Y/n\"\n    invoked from within\n\"expect {\n  timeout { send_user \"\\nFailed to get \u0027Switch to unix_socket authentication [Y/n] \u0027 prompt\\n\"; exit 1 }\n  eof { send_user \"\\nFailed to get \u0027...\"\n    (file \"/usr/local/bin/kolla_security_reset\" line 12)\n\nERROR 1045 (28000): Access denied for user \u0027root\u0027@\u0027localhost\u0027 (using password: YES)\nERROR 1045 (28000): Access denied for user \u0027root\u0027@\u0027localhost\u0027 (using password: YES)\n\u0007mysqladmin: connect to server at \u0027localhost\u0027 failed\nerror: \u0027Access denied for user \u0027root\u0027@\u0027localhost\u0027 (using password: YES)\u0027\n2021-09-09 11:58:57.719994 |                                      |    WARNING | ERROR: Can\u0027t run container mysql_upgrade_db\nstderr: Version check failed. Got the following error when calling the \u0027mysql\u0027 command line client\nERROR 1045 (28000): Access denied for user \u0027root\u0027@\u0027localhost\u0027 (using password: YES)\nFATAL ERROR: Upgrade failed\n2021-09-09 11:58:57.721410 | fa163ec0-ca3c-1d89-2ca8-0000000013e7 |      FATAL | Create containers managed by Podman for /var/lib/tripleo-config/container-startup-config/step_2 | undercloud | error\u003d{\"changed\": false, \"msg\": \"Failed containers: mysql_bootstrap, mysql_upgrade_db\"}\n2021-09-09 11:58:57.722924 | fa163ec0-ca3c-1d89-2ca8-0000000013e7 |     TIMING | tripleo_container_manage : Create containers managed by Podman for /var/lib/tripleo-config/container-startup-config/step_2 | undercloud | 0:07:40.587420 | 128.37s\n\n```","accounts_in_message":[],"_revision_number":1},{"id":"87ae64bcfe1d05ae56b42ee2ed0de80af3fc2669","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-09-09 12:47:56.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n... snip ...\n\n\u003e stderr: Version check failed. Got the following error when calling the \u0027mysql\u0027 command line client\n\u003e ERROR 1045 (28000): Access denied for user \u0027root\u0027@\u0027localhost\u0027 (using password: YES)\n\u003e FATAL ERROR: Upgrade failed\n\u003e 2021-09-09 11:58:57.721410 | fa163ec0-ca3c-1d89-2ca8-0000000013e7 |      FATAL | Create containers managed by Podman for /var/lib/tripleo-config/container-startup-config/step_2 | undercloud | error\u003d{\"changed\": false, \"msg\": \"Failed containers: mysql_bootstrap, mysql_upgrade_db\"}\n\u003e 2021-09-09 11:58:57.722924 | fa163ec0-ca3c-1d89-2ca8-0000000013e7 |     TIMING | tripleo_container_manage : Create containers managed by Podman for /var/lib/tripleo-config/container-startup-config/step_2 | undercloud | 0:07:40.587420 | 128.37s\n\u003e \n\u003e ```\n\nThe problem is that the container still has 10.3 mariadb:\n2021-09-09 11:56:54 0 [Note] /usr/libexec/mysqld (mysqld 10.3.28-MariaDB) starting as process 288 ...\n\nI presume that is not expected?","accounts_in_message":[],"_revision_number":1},{"id":"9733d8f74e03a739bcb850b654bf790d9fc3d692","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-09-09 12:55:15.000000000","message":"Patch Set 1:\n\nit happened due to this https://review.opendev.org/c/openstack/tripleo-common/+/808044/1/container-images/tcib/base/base.yaml#16\nand mariadb10.3 is installed \nhttps://e500f5844c497d7c1455-bb0af7d0ed113130252cfd767637324e.ssl.cf2.rackcdn.com/808044/1/check/tripleo-ci-centos-8-content-provider/9fb1788/logs/container-builds/dc5247ad-e80a-401f-ad1b-f4ae34437e84/base/mariadb/mariadb-build.log\n```\n mariadb                          x86_64  3:10.3.28-1.module_el8.3.0+757+d382997d quickstart-centos-appstreams        6.0 M\n mariadb-backup                   x86_64  3:10.3.28-1.module_el8.3.0+757+d382997d quickstart-centos-appstreams        6.1 M\n mariadb-server-galera            x86_64  3:10.3.28-1.module_el8.3.0+757+d382997d quickstart-centos-appstreams         61 k\n mariadb-server-utils             x86_64  3:10.3.28-1.module_el8.3.0+757+d382997d quickstart-centos-appstreams        1.1 M\n```","accounts_in_message":[],"_revision_number":1},{"id":"f66fb1584866a37cc7e63e0905151dd9f6fd030a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-09-09 13:01:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c4129e7fd24717fb2b82d91c2a0ed77e46b76934","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 15:18:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f86ce2101a4480184fdd739767e164b : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31809fd07abf43f2afaf8d7a4a871d77 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/836026c7c8834d639d95936572aea2f9 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/706b0357a58e48e9bb29fc44f06b7e0e : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d39a83f0e28b4c78bc29ce6e0fdd1afa : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/371476d393634716ab3a6f39886c3be1 : SUCCESS in 5m 59s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/0fd79d0e60e842d593479d716b682e16 : SUCCESS in 2h 16m 52s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1980afc009134849913271c84b4c7ea3 : FAILURE in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6250b0adbbf42c2b78ef214c4cc3d9c : SUCCESS in 4m 26s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/8bf26864d5774731ae36cc536d8cbbed : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":2},{"id":"04faf1b604aafd9f23cdf703afaf3928e6f936b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-09-09 15:19:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0c83ec07b2ffba0239198ce598cd19325dd2a0a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-09-09 15:32:37.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e7177df448af4da580876f9cd9a3e7f4c942fd4d","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2021-09-09 15:36:49.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"82defdca0a3a9261092aef099d3086141e194c6d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-09 16:22:19.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9ac8004f944d4c27ccdb0dba5e2322f17bf88890","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-09-09 18:24:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9fb45e0e541745e1a76fc0b4ca023f6f : SUCCESS in 2h 50m 56s","accounts_in_message":[],"_revision_number":4},{"id":"fbd474f844505e8a2051709422f978f19068b431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 19:23:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c42708a6bb34b6890dd71fe06d82271 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0231fdf38714ca687e1c8674f484f10 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ede52438cb914cd180c0e27a8604148e : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e3ab16e9e484470abf2171b6b70095b : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15394639397c4d9ea21cab4cc9277b62 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6dfea64e47423ba7ba24f7dbfeee3d : SUCCESS in 5m 30s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/eab2ebad769f41e8907ab9177f371035 : SUCCESS in 3h 04m 53s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9cf503be910b4f379a02682c75aa1e12 : SUCCESS in 1h 38m 28s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/56c1967a39de48a3ab1c5e5c332a1568 : SUCCESS in 47m 33s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/cc3f15dc30b9429982663b33e231634a : SUCCESS in 1h 12m 08s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/c31af76d46ad45a29317a314c094087d : SUCCESS in 2h 58m 42s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6953e9e6a0eb476a81a829f6cbfffd27 : SUCCESS in 1h 27m 24s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/cfa0700e73ce44178f5f80c45fdfda24 : SUCCESS in 1h 10m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e6b98ba965f34d18946cc6a08269e498 : SUCCESS in 6m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bcfe341648c4de982dfb893b122328e : SUCCESS in 5m 37s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/f54590f8e98b43f7b45dec381c51ce53 : SUCCESS in 6m 10s","accounts_in_message":[],"_revision_number":4},{"id":"3e144da964d84632a9a3f6380e520a7d6d252c80","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-09 20:45:12.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"71ca6bd919c71d91573884875f57cc90fe882642","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 20:46:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3487498efcf37c52a38bf671c25d9d366d3b8b65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 01:57:27.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5188530f1261425cbe3d8c805d56c4f1 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a79d0da7406441680d8abe36f4310cc : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e71f0671607401f811426c83c26b9e4 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ffae49b304a4f0ea9dec9d1e41b6c3a : SUCCESS in 5m 15s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/eb0e3f45453742afb2d34cf453de94f7 : SUCCESS in 3h 09m 43s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/bb65ac1ff593491bbff89fc4afb95ed1 : SUCCESS in 1h 40m 32s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/326f6fd6c11943c9bf5cf2ba8ea0e074 : SUCCESS in 57m 39s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c032c53bffe1412cb3a8f7824a8ccf1a : SUCCESS in 1h 38m 16s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/13bf4797aae7475c87432062fa1507db : SUCCESS in 2h 59m 43s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/07090c6729f947b1b81dee435be90ea6 : SUCCESS in 1h 42m 10s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/75520923161447f4a4d61d7e2065c14c : SUCCESS in 1h 21m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f33308ff793b46d3b6dde423db571217 : SUCCESS in 5m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f3cde10592c045faabf9a11fa3e70543 : SUCCESS in 4m 54s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/becd7c18605b4ec9a6fa3ee350942555 : SUCCESS in 6m 22s","accounts_in_message":[],"_revision_number":4},{"id":"948b09d69ecfbaad433bed929879e69384e80ab9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 01:57:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"95eecb96bbf12c517f695d200732ad69b89e0f68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 01:58:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3892409bae4420eabc58645aaa73a3f : SUCCESS in 57s","accounts_in_message":[],"_revision_number":4},{"id":"65e0659fc59e75ea9466fdc1f4a40c5fb8b6388d","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2021-09-10 05:20:09.000000000","message":"Patch Set 4:\n\nhttps://2c7c22777014beabc415-9d89cba2eeb12e627930ba775ed5fbca.ssl.cf2.rackcdn.com/808044/4/gate/tripleo-ci-centos-8-content-provider/eb0e3f4/logs/container-builds/9cf18e2d-8759-49b6-a144-3776ab052234/base/mariadb/mariadb-build.log\n```\nmariadb                          x86_64  3:10.5.9-1.module_el8.5.0+732+7afc82e7 quickstart-centos-appstreams        6.2 M\n mariadb-backup                   x86_64  3:10.5.9-1.module_el8.5.0+732+7afc82e7 quickstart-centos-appstreams        6.8 M\n mariadb-server-galera            x86_64  3:10.5.9-1.module_el8.5.0+732+7afc82e7 quickstart-centos-appstreams         65 k\n mariadb-server-utils             x86_64  3:10.5.9-1.module_el8.5.0+732+7afc82e7 quickstart-centos-appstreams        1.2 M\n```","accounts_in_message":[],"_revision_number":4},{"id":"bdf61c3924346d8aa4029aae81762ee2b95ebfe9","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-09-11 14:04:53.000000000","message":"Patch Set 4:\n\nFWIW this created https://bugs.launchpad.net/tripleo/+bug/1943330","accounts_in_message":[],"_revision_number":4},{"id":"3c16457e10ed2b668a1425556c18b896a4a1623c","tag":"autogenerated:gerrit:revert","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2021-09-12 22:56:44.000000000","message":"Created a revert of this change as Idae80ada596cfc19b78e198a2580ca277a151b02","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"03abc41cacf6cafdcb12e22bf56c9a29f4b02614","revisions":{"e8d5d4fc9ada71cb66bacd91d1a72aa86c33ad34":{"kind":"REWORK","_number":1,"created":"2021-09-09 09:59:40.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/44/808044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/44/808044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/1"}}},"commit":{"parents":[{"commit":"e08affe816a700c996975181b84594cf02431b63","subject":"Add security_reset.expect.10.5 from kolla for mariadb\u003e10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e08affe816a700c996975181b84594cf02431b63"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-09-09 09:47:53.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-09-09 09:59:37.000000000","tz":330},"subject":"Switch to mariadb:10.5 module","message":"Switch to mariadb:10.5 module\n\nIn EL9, we are consuming mariadb:10.5, It adds the same for\nEL8 also.\n\nIt also copies the required security_reset[1] file\nneeded for mariadb:10.5\n\n[1]. https://review.opendev.org/c/openstack/tripleo-common/+/807577\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I26174f6a450885fd65f8b1af5622d7d68c0aaf40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e8d5d4fc9ada71cb66bacd91d1a72aa86c33ad34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e8d5d4fc9ada71cb66bacd91d1a72aa86c33ad34"}]},"branch":"refs/heads/master"},"4fb92738d1d24a223b20ef42adee549cf2894888":{"kind":"REWORK","_number":2,"created":"2021-09-09 13:01:13.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/44/808044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/44/808044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/2"}}},"commit":{"parents":[{"commit":"e08affe816a700c996975181b84594cf02431b63","subject":"Add security_reset.expect.10.5 from kolla for mariadb\u003e10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e08affe816a700c996975181b84594cf02431b63"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-09-09 09:47:53.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-09-09 13:01:02.000000000","tz":330},"subject":"Switch to mariadb:10.5 module","message":"Switch to mariadb:10.5 module\n\nIn EL9, we are consuming mariadb:10.5, It adds the same for\nEL8 also.\n\nIt also copies the required security_reset[1] file\nneeded for mariadb:10.5\n\n[1]. https://review.opendev.org/c/openstack/tripleo-common/+/807577\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I26174f6a450885fd65f8b1af5622d7d68c0aaf40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4fb92738d1d24a223b20ef42adee549cf2894888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4fb92738d1d24a223b20ef42adee549cf2894888"}]},"branch":"refs/heads/master"},"87cadf95617410ec8451bc85572314ed0e911c73":{"kind":"REWORK","_number":3,"created":"2021-09-09 15:19:57.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/44/808044/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/44/808044/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/3"}}},"commit":{"parents":[{"commit":"e08affe816a700c996975181b84594cf02431b63","subject":"Add security_reset.expect.10.5 from kolla for mariadb\u003e10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e08affe816a700c996975181b84594cf02431b63"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-09-09 09:47:53.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-09-09 15:19:43.000000000","tz":330},"subject":"Switch to mariadb:10.5 module","message":"Switch to mariadb:10.5 module\n\nIn EL9, we are consuming mariadb:10.5, It adds the same for\nEL8 also.\n\nIt also copies the required security_reset[1] file\nneeded for mariadb:10.5\n\n[1]. https://review.opendev.org/c/openstack/tripleo-common/+/807577\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I26174f6a450885fd65f8b1af5622d7d68c0aaf40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/87cadf95617410ec8451bc85572314ed0e911c73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/87cadf95617410ec8451bc85572314ed0e911c73"}]},"branch":"refs/heads/master"},"03abc41cacf6cafdcb12e22bf56c9a29f4b02614":{"kind":"REWORK","_number":4,"created":"2021-09-09 15:32:37.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/44/808044/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/44/808044/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/44/808044/4"}}},"commit":{"parents":[{"commit":"e08affe816a700c996975181b84594cf02431b63","subject":"Add security_reset.expect.10.5 from kolla for mariadb\u003e10.5","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e08affe816a700c996975181b84594cf02431b63"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-09-09 09:47:53.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2021-09-09 15:32:26.000000000","tz":330},"subject":"Switch to mariadb:10.5 module","message":"Switch to mariadb:10.5 module\n\nIn EL9, we are consuming mariadb:10.5, It adds the same for\nEL8 also.\n\nIt also copies the required security_reset[1] file\nneeded for mariadb:10.5\n\n[1]. https://review.opendev.org/c/openstack/tripleo-common/+/807577\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I26174f6a450885fd65f8b1af5622d7d68c0aaf40\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/03abc41cacf6cafdcb12e22bf56c9a29f4b02614"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/03abc41cacf6cafdcb12e22bf56c9a29f4b02614"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
