)]}'
{"id":"openstack%2Ftripleo-common~809486","triplet_id":"openstack%2Ftripleo-common~master~I628abc4cb8c46c3b2e5bfd3cd16bb6d5cc95fabe","project":"openstack/tripleo-common","branch":"master","topic":"bp/whole-disk-default","hashtags":[],"change_id":"I628abc4cb8c46c3b2e5bfd3cd16bb6d5cc95fabe","subject":"Detect image format for qemu-nbd --format","status":"MERGED","created":"2021-09-16 23:09:33.000000000","updated":"2021-09-23 13:27:36.000000000","submitted":"2021-09-23 13:25:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"809486-bp/whole-disk-default","meta_rev_id":"55949a988a00d6561a55d700df71d4d81ee225d5","_number":809486,"virtual_id_number":809486,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-23 13:25:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-09-21 00:30:16.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":2,"date":"2021-09-23 09:45:21.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2021-09-21 20:00:55.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":1,"date":"2021-09-23 09:45:21.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-17 00:32:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-17 01:53:14.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-21 20:00:55.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-09-23 09:45:21.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"}],"messages":[{"id":"ba122567adcd97c2e3e6134c5c94d3a92f0e2167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-09-16 23:09:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38250f7cd1428466f74bd1d681dba67a7d13b726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-17 00:32:49.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7b431ca22544f8f9253ae7a687942fa : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20bfdde557504ce3a4a5d3e5568e0414 : SUCCESS in 4m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1cfbf477a8645a3921b0c350499c703 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ee82b1802a64f6082a74a09726a5b3c : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f4932bbf52847ac8b7b08abd670f9f1 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/598c1de5d3d34388b43e056645fc10de : SUCCESS in 4m 30s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/2a161ad31a814177bdee6a140d24b017 : SUCCESS in 1h 16m 53s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/412e3913e7104fe3a46b6c269e54089c : FAILURE in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dd4bd52cc87142cd88296582b6ebed5c : SUCCESS in 4m 32s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/cfb5f5682ceb4a939dcf44ab98344807 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"dd0d9094fa9e363c589ada1ab3078831e5a62de3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-09-17 01:53:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3f9afcb8da2a4c6d91ca43bb4c90b357 : SUCCESS in 2h 42m 54s","accounts_in_message":[],"_revision_number":1},{"id":"06e0395407c8a12a72fdf62d0c4bfe2f3a106865","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2021-09-20 21:56:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fd334d9ede344c5a03bde996f8b48e01a8909e2b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-09-21 00:30:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8b49c0b1ee73480a81af285f6ae8558f : SUCCESS in 2h 33m 06s","accounts_in_message":[],"_revision_number":2},{"id":"344b48100294eff8db83d403dddae054ba680c1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 00:45:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad113290b4f24a94bc48aaeee5b19336 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5b0c279359a4272886b9ac57012b0af : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e303a33620f4459d9f5df8e485a148bb : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08b7fe0401bc44eca06ef69b8d3be78c : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6108f57442546e196722cf41e46988e : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3487879d1dd04e86a188e222d80df22f : SUCCESS in 4m 17s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3d52cabc766d48afa08c756ef6d266a7 : SUCCESS in 2h 37m 44s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/a579909799ef40e69b008a9c46aac133 : SUCCESS in 2h 48m 02s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/170b4d9ff2bc479392671a173f725bf1 : SUCCESS in 1h 19m 48s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/4c51df2589f2479eb20b4a81eedf2460 : SUCCESS in 1h 10m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/71980f00041e471f8add291255d21f7f : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae9876146c054c338b1379b0763b4e5f : SUCCESS in 4m 53s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/4c287714ce8e4a4993b401f5c588155e : SUCCESS in 7m 50s","accounts_in_message":[],"_revision_number":2},{"id":"17074ffed94a5efe2775a2d22800b847f97a9715","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-21 20:00:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"52e11a9581edbc871bd62c02e30ba6d849f021b5","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2021-09-23 09:45:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"22d6cf959b074982c59335c2be71ed6995bb54a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 09:46:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14cf420fb9c4257cf08beac8ac134d2fa5c43046","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 13:25:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30cffcc7f5ac48fc93dd82d800c21bf0 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f618999c08c24766a79d4821e9f1bbbe : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cad747ef93242f4801608f3b66f512c : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6032aab4e1b64a81ac1ae6ec8b7180b0 : SUCCESS in 5m 05s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/798821d0b7dd4218942d04186c76147b : SUCCESS in 2h 46m 23s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/c990a1a178764b68b3d0fd7ffb2c412f : SUCCESS in 2h 35m 09s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0d76739b1142438ea1397b7a43b522be : SUCCESS in 1h 15m 52s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/963053165b4d49a9bb5357bbb4e7e6a5 : SUCCESS in 1h 05m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a2625f5848924610850f7cd1dccdfce1 : SUCCESS in 5m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/878b1cf866454e7ab985956e84ae408c : SUCCESS in 4m 57s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/8f008abdee3e471da0c31dab92f6fbdd : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":2},{"id":"8a44f665749a5fa19ff73798c4da1910b3ca150f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 13:25:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"55949a988a00d6561a55d700df71d4d81ee225d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 13:27:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/478b91732e6b41648f71b8f1fa9657c8 : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"db7280e653da1d7eb5b3a345ee22a93ceb1e3710","revisions":{"c22809ca4b22ec908a05036a048e9249be1822db":{"kind":"REWORK","_number":1,"created":"2021-09-16 23:09:33.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/86/809486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/86/809486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/809486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/809486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/809486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/86/809486/1"}}},"commit":{"parents":[{"commit":"91eeabeecf64e0cc72ace76e7b08059a23aaa8c1","subject":"Merge \"Revert \"Switch to mariadb:10.5 module\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/91eeabeecf64e0cc72ace76e7b08059a23aaa8c1"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-09-16 23:03:15.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-09-16 23:07:05.000000000","tz":720},"subject":"Detect image format for qemu-nbd --format","message":"Detect image format for qemu-nbd --format\n\nqemu-nbd will error when the image format is raw unless there is an\nexplicit --format raw argument provided.\n\nThis change adds format detection for raw and qcow2 using qemu-img\ninfo. Any other image type will depend on qemu-nbd auto-detection.\n\nBlueprint: whole-disk-default\nChange-Id: I628abc4cb8c46c3b2e5bfd3cd16bb6d5cc95fabe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c22809ca4b22ec908a05036a048e9249be1822db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c22809ca4b22ec908a05036a048e9249be1822db"}]},"branch":"refs/heads/master"},"db7280e653da1d7eb5b3a345ee22a93ceb1e3710":{"kind":"REWORK","_number":2,"created":"2021-09-20 21:56:39.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/86/809486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/86/809486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/809486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/809486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/809486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/86/809486/2"}}},"commit":{"parents":[{"commit":"049f9a9659d09248f3b15d4e152de8d9323b6c6d","subject":"Remove /etc/rabbitmq/rabbitmq.conf","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/049f9a9659d09248f3b15d4e152de8d9323b6c6d"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-09-16 23:03:15.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2021-09-20 21:56:24.000000000","tz":720},"subject":"Detect image format for qemu-nbd --format","message":"Detect image format for qemu-nbd --format\n\nqemu-nbd will error when the image format is raw unless there is an\nexplicit --format raw argument provided.\n\nThis change adds format detection for raw and qcow2 using qemu-img\ninfo. Any other image type will depend on qemu-nbd auto-detection.\n\nBlueprint: whole-disk-default\nChange-Id: I628abc4cb8c46c3b2e5bfd3cd16bb6d5cc95fabe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/db7280e653da1d7eb5b3a345ee22a93ceb1e3710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/db7280e653da1d7eb5b3a345ee22a93ceb1e3710"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
