)]}'
{"id":"openstack%2Ftripleo-common~812486","triplet_id":"openstack%2Ftripleo-common~master~I3b6afbc51c69a4c51e464ddc830b2e055ca4f0c3","project":"openstack/tripleo-common","branch":"master","topic":"constants-cleanup","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-10-11 12:57:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3b6afbc51c69a4c51e464ddc830b2e055ca4f0c3","subject":"Remove unused constants for TripleO UI","status":"MERGED","created":"2021-10-05 11:42:43.000000000","updated":"2021-10-11 12:58:50.000000000","submitted":"2021-10-11 12:57:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"812486-constants-cleanup","meta_rev_id":"83ab833d42461c8cabf51ba09a618a12456a582b","_number":812486,"virtual_id_number":812486,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-11 12:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-05 14:41:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-10-05 20:04:03.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-11 10:03:22.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-11 10:03:22.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-05 14:41:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-05 14:51:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-05 20:04:03.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-10-11 10:03:22.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"}],"messages":[{"id":"e3a05a143988849226ca1a1deaa9ff50545cff99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-05 11:42:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f827f849ec840f188421b77f6ac4ff2e6c1ee0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-05 11:52:23.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3fb662d20e6134587ba5a579e101816117defc41","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-10-05 14:41:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/fe94fabee63b40c69518a515bf498b81 : SUCCESS in 2h 48m 24s","accounts_in_message":[],"_revision_number":2},{"id":"43bd04405619920f11d919bbadbaa8e37b702182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 14:51:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/475907e8e7dc4780880329fc70427977 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc650549691a41358e346498f3f50346 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0165e4342ff14a0bbb41e0cdb941e86f : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39aea9fee1e6467a83477b78dd441000 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75cbec47137942f0862cd4e55bb4d5aa : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fff4bba062cd46c38850b551230b4c80 : SUCCESS in 5m 36s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0e669875ce684537aa6628d3f83fbcb0 : SUCCESS in 2h 46m 47s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/644515866d8e4970a9517354b6ff7895 : SUCCESS in 1h 29m 19s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2a77b843c206492daa066cbecfce81f0 : SUCCESS in 42m 17s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/44666b0ec24041b585fb759d237ffd96 : SUCCESS in 1h 20m 53s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/100c2d61938a4ecd8d897dddee6f4cb8 : SUCCESS in 2h 34m 56s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1d08d261c574463eb77ca754db17a63b : SUCCESS in 1h 23m 15s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/e240b79d45054b688e897ec43d05b055 : SUCCESS in 1h 08m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/903c2291b32043f6acc7fb9b5a46aed4 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/66cab8770ccc4551a7b9913f22cda101 : SUCCESS in 5m 04s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/62fbdd6d03f34547b9de8134f482cc8f : SUCCESS in 6m 16s","accounts_in_message":[],"_revision_number":2},{"id":"13fe895d832fd2b96660da3d277a9ac628cba251","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-05 20:04:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"861eeb792460189ff0c95dd4bea0d189545ab0c2","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-10-11 10:03:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f3224d56c604a7335d35aa08512f43e6281b32f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 10:03:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fe2bddb7644876a02843ad273c9661e9293e07b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 12:57:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d3d560543884c86a73aade143c60427 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db98a707c74b44e295d4572b6eeb6bd7 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f07f7c20bf71423bb315e62af249f548 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a98423a85ddd48df8f7ff3a5e6751153 : SUCCESS in 5m 06s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e3dcd16b322044ae91d8792c83223a65 : SUCCESS in 2h 52m 20s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/ca689fe4147c4fcfb75e3c7ea4afa349 : SUCCESS in 1h 28m 46s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/373f31dbd83341d3aa0a7f64545d193c : SUCCESS in 41m 37s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4021e69b50b04f23b59f540ada63de11 : SUCCESS in 1h 11m 36s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/650957526ef54ba392865028415334bc : SUCCESS in 2h 31m 56s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6824b8461e4a43f082f4605d2c54623f : SUCCESS in 1h 19m 10s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/a79daef408544e1fa73024ed7bd94714 : SUCCESS in 59m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1cf49f6a2a894bccafdba5dace9300c5 : SUCCESS in 6m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d35b3bedb82a443b97b3058c0aa759cb : SUCCESS in 5m 17s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/253d7ceca7334131a4cb1df6c022910d : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":2},{"id":"0e4cbaea4a5c9a8017abe64539f11ab3076da753","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 12:57:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"83ab833d42461c8cabf51ba09a618a12456a582b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-11 12:58:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d2e6d23df944b0a5ecb39c9523b863 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2ebe8f0b7967f6c6055d226996563c7c3c0b5776","revisions":{"f756f9066fd1e94b51d59b30e080a1c39e4de4de":{"kind":"REWORK","_number":1,"created":"2021-10-05 11:42:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/812486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/86/812486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/812486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/812486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/812486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/86/812486/1"}}},"commit":{"parents":[{"commit":"8bf82d79a710539c9d236102ba150f30d90e404c","subject":"Merge \"Allow for OCI manifests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8bf82d79a710539c9d236102ba150f30d90e404c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-05 11:36:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-05 11:41:47.000000000","tz":540},"subject":"Remove unused constants for TripleO UI","message":"Remove unused constants for TripleO UI\n\n... because these constants are no longer used since TripleO UI was\nremoved during Rocky[1].\n\n[1] 5566697549bfd6bbd55691ff4f290079f8085084\n\nChange-Id: I3b6afbc51c69a4c51e464ddc830b2e055ca4f0c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f756f9066fd1e94b51d59b30e080a1c39e4de4de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f756f9066fd1e94b51d59b30e080a1c39e4de4de"}]},"branch":"refs/heads/master"},"2ebe8f0b7967f6c6055d226996563c7c3c0b5776":{"kind":"NO_CHANGE","_number":2,"created":"2021-10-05 11:52:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/812486/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/86/812486/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/812486/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/812486/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/86/812486/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/86/812486/2"}}},"commit":{"parents":[{"commit":"8bf82d79a710539c9d236102ba150f30d90e404c","subject":"Merge \"Allow for OCI manifests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8bf82d79a710539c9d236102ba150f30d90e404c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-05 11:36:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-05 11:42:56.000000000","tz":540},"subject":"Remove unused constants for TripleO UI","message":"Remove unused constants for TripleO UI\n\n... because these constants are no longer used since TripleO UI was\nremoved during Rocky[1].\n\n[1] 5566697549bfd6bbd55691ff4f290079f8085084\n\nChange-Id: I3b6afbc51c69a4c51e464ddc830b2e055ca4f0c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2ebe8f0b7967f6c6055d226996563c7c3c0b5776"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/2ebe8f0b7967f6c6055d226996563c7c3c0b5776"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
