)]}'
{"id":"openstack%2Ftripleo-common~819151","triplet_id":"openstack%2Ftripleo-common~master~I9b1af1b53c7648b92ef8dfdb53f94229ca21448e","project":"openstack/tripleo-common","branch":"master","attention_set":{},"removed_from_attention_set":{"27419":{"account":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"last_update":"2021-11-25 13:50:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9b1af1b53c7648b92ef8dfdb53f94229ca21448e","subject":"Removing erlang-hipe","status":"MERGED","created":"2021-11-24 17:43:52.000000000","updated":"2021-11-25 13:51:53.000000000","submitted":"2021-11-25 13:50:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819151","meta_rev_id":"c13902407601936cdbd4624ca0978df6976b138b","_number":819151,"virtual_id_number":819151,"owner":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-25 13:50:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-24 21:33:58.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"all":[{"value":2,"date":"2021-11-24 17:45:40.000000000","_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":1,"date":"2021-11-24 19:15:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-25 09:09:13.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"all":[{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-25 09:09:13.000000000","_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-24 17:45:40.000000000","updated_by":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"},{"updated":"2021-11-24 19:15:54.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2021-11-24 20:48:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-24 21:33:58.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-25 09:09:13.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"}],"messages":[{"id":"0c523c930220e8fa8e9c677b003063a73f8596e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-11-24 17:43:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0dcad11b6e000c652935df5418702f483cdaef6d","author":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"date":"2021-11-24 17:45:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2ff2eff3b0903ae95617e2f973761fd5175810d8","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2021-11-24 19:15:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c9439bf3e730ee2d89dd35d51c36c6f9bb8d9f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 20:48:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/400d5efbf76f4675bdf36fbde6ae69d0 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04cb22f20b1b4121abea6d75dc9a3de8 : SUCCESS in 3m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91b622d1737b4665aa85b68213d88b42 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b71d59d57414cc3ac6986e0e3ab6ec6 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e2e56fcd4462409b9ac680ba00a27cfa : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c9e37574c149ae8b566f3702d52444 : SUCCESS in 4m 28s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3138507e67ba4d2e9aaa4739152d163d : SUCCESS in 2h 55m 48s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/77b8e9c7ce1d4519874947e1a858b8fa : SUCCESS in 1h 34m 30s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bfd0b8873ad9438c8253061371c00444 : SUCCESS in 44m 54s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/043e70dd34b84e3aa38580d319e78399 : FAILURE in 1h 36m 15s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/874972fd850047b79ed8f71a94b0c090 : SUCCESS in 1h 18m 34s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/9c1bfbaac5444a46b4f2b7348dbca5ce : SUCCESS in 2h 37m 11s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d31daaf94efd46ac936a5229b4c69572 : SUCCESS in 1h 14m 11s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/23f034f077fe4122ae3f824a67d0652b : SUCCESS in 1h 10m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/948f0b3869de41008b77a93e3a77c8a7 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8e1fd445c51e4c4a9c3743dfccec1af8 : SUCCESS in 4m 26s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/3d8773f72dae43fa867983a7da9a51fb : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"91fd89e3a034c5e15a3a0c8e2b2f796b8ef0b8c4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-11-24 21:33:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/24f0304ba25e46c4918d1e06e269cc00 : SUCCESS in 2h 55m 39s","accounts_in_message":[],"_revision_number":1},{"id":"a823fd790869b48817e7a9494a6454073b97a013","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-11-25 02:04:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"983ea1a399a256461410bd72c60b38316a263a6e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2021-11-25 04:43:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/084aec2551bb46cd9af6627fdf010945 : SUCCESS in 2h 29m 44s","accounts_in_message":[],"_revision_number":1},{"id":"c883ea7da80f6e2583fa34f2b747040d6c883955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 06:43:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10e0de5f977b4ff7814942d63fe90539 : SUCCESS in 8m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c2bb8b8cc8146c8af07786626ba76bd : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07ae1a8d0ec7470b9ab5cf509e00f36c : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5de3b7e812a24299b5f27987cc0a3536 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/926bea0839ae43d79de9a85f310ce9e2 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c674091de3004b1fbf9d2a93e63bdbf7 : SUCCESS in 6m 17s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/38d1f49598574a3fb8c2ae5cc8a5fb27 : SUCCESS in 3h 34m 55s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a4fcf8848add410dbba6ac70895dd69e : SUCCESS in 1h 50m 19s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/73c9d9c4ca8045fab9c5f94504342096 : SUCCESS in 56m 47s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/647fd2e4e9974f9bb9e5eb85acf75347 : SUCCESS in 1h 43m 17s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f395507ba4244340b4675fd2cf60b219 : SUCCESS in 1h 20m 55s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/5ffee12317244b2eae7a8c6495270705 : SUCCESS in 3h 19m 40s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b03dc403172c4fccad10b62af2a242be : SUCCESS in 1h 24m 51s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/18f77dff59764422b10bd504d026db6f : SUCCESS in 1h 23m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/644101015a6d4205a1841af65a171a2c : SUCCESS in 7m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4c1633e9220d48d087b4fbb7eefa7211 : SUCCESS in 7m 26s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/3ed2ba799f6c4cb487276f982c7342fc : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":1},{"id":"601217a04a60e660750e1c023dfb0a3e4329d334","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2021-11-25 09:09:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7cdd310b4ffe01c553fe113c44ca3135bea75f29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 09:09:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb56e4fde0643a243f9b6a1507baa9ba4a9ac028","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 13:50:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6d97161ad0e450abf50c6d563852cdf : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00137fd97c4a41c8abc0785c0ccbd089 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3e1bb380e3b483a89273c7e38771b92 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6ff729d373a477a891f6715805fee90 : SUCCESS in 4m 22s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/11f0673d42244ab7bdfef27ae8ce84c7 : SUCCESS in 3h 42m 20s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/9199bee2b3fe408eb34f214349e58a15 : SUCCESS in 1h 34m 31s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f32c78f172214799859289548a1b4381 : SUCCESS in 55m 58s\n- tripleo-ci-centos-8-scenario001-standalone https://zuul.opendev.org/t/openstack/build/ac556fe18b2a4c309914d4aa8d536a4f : SUCCESS in 1h 59m 39s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a48b3dbd81b4448487ba58bf587bff63 : SUCCESS in 1h 23m 04s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/2e3a9d365dbd4daca35e3405081d7f83 : SUCCESS in 2h 52m 49s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1073137513d24729951370349510e399 : SUCCESS in 1h 16m 15s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/e86691360a3c45f788bfe6f12ce9fce3 : SUCCESS in 1h 13m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b9c2043104e4e3eac59ce6449cc3cab : SUCCESS in 4m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/023e5eed34ef4a31bc1157f184ac453f : SUCCESS in 4m 23s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/bcc903007aff4d4b84bac18b21122c44 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":1},{"id":"420a005a9d4895306cd6c2d65a467239ca94c353","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 13:50:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c13902407601936cdbd4624ca0978df6976b138b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 13:51:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49d55833abd14a58a77d27c0e31e13cd : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62713d731bb91ba0f6858e2dac6c88f172fa2b99","revisions":{"62713d731bb91ba0f6858e2dac6c88f172fa2b99":{"kind":"REWORK","_number":1,"created":"2021-11-24 17:43:52.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/51/819151/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/51/819151/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/51/819151/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/51/819151/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/51/819151/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/51/819151/1"}}},"commit":{"parents":[{"commit":"bdd016b8ff5267004f873772a23ee89c8143a643","subject":"Merge \"Running ansible-config before writing default config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bdd016b8ff5267004f873772a23ee89c8143a643"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-11-24 17:42:40.000000000","tz":-300},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-11-24 17:42:40.000000000","tz":-300},"subject":"Removing erlang-hipe","message":"Removing erlang-hipe\n\nThis is no more shipped, at least not in CentOS9.\n\nThis is breaking the container build process.\n\nChange-Id: I9b1af1b53c7648b92ef8dfdb53f94229ca21448e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/62713d731bb91ba0f6858e2dac6c88f172fa2b99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/62713d731bb91ba0f6858e2dac6c88f172fa2b99"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
