)]}'
{"id":"openstack%2Ftripleo-common~823608","triplet_id":"openstack%2Ftripleo-common~stable%2Fwallaby~Ic946d0ef686cb178e0f02db012f2767c9ec5d6d4","project":"openstack/tripleo-common","branch":"stable/wallaby","topic":"remove_libcgroup-tools-stable/wallaby","attention_set":{},"removed_from_attention_set":{"17216":{"account":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"last_update":"2022-01-09 16:09:27.000000000","reason":"Martin Schuppert replied on the change"},"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2022-01-10 06:25:45.000000000","reason":"Rabi Mishra replied on the change"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-01-10 10:28:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic946d0ef686cb178e0f02db012f2767c9ec5d6d4","subject":"Remove libcgroup-tools from libvirt container","status":"MERGED","created":"2022-01-06 15:30:08.000000000","updated":"2022-01-10 10:29:37.000000000","submitted":"2022-01-10 10:28:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"823608-remove_libcgroup-tools-stable/wallaby","cherry_pick_of_change":793976,"cherry_pick_of_patch_set":1,"meta_rev_id":"1ae690a7899b20130f7863fe3d8d834771f346ae","_number":823608,"virtual_id_number":823608,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-10 10:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-01-06 17:43:12.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"recommended":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"all":[{"value":1,"date":"2022-01-09 16:09:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":2,"date":"2022-01-10 06:25:45.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-07 11:39:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"value":2,"date":"2022-01-10 06:59:33.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"value":1,"date":"2022-01-10 06:59:33.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"CC":[{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-06 15:30:08.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"state":"REVIEWER"},{"updated":"2022-01-06 17:43:12.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-06 19:14:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-07 11:39:04.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"CC"},{"updated":"2022-01-07 11:39:04.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"CC"},{"updated":"2022-01-07 11:39:04.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"CC"},{"updated":"2022-01-07 11:39:04.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"CC"},{"updated":"2022-01-07 11:39:04.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2022-01-07 11:48:03.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"state":"CC"},{"updated":"2022-01-10 06:25:45.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"},{"updated":"2022-01-10 06:59:33.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"}],"messages":[{"id":"a95bb8fe94c870e98bbeee1f27a976dafcb54b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-06 15:30:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8c15274e379d2b4417fc073e0a2a537e83bf19bc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-06 17:43:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/49e5ede2615b4efc87dddaa88bd62e21 : FAILURE in 2h 12m 04s","accounts_in_message":[],"_revision_number":1},{"id":"0826f5c5fa713df3e9c775d885d3a2ad2fb50c60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-06 19:14:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec660c9897d24ef69d0f41e351219719 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d09c38ec84f43ecbcba1065a88d49f3 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d88f6ca38a4c497db79fa9cf77380fdb : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb0879fb72c04835b34ef99251acd3a6 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f17ab0dd34e14d61b76607996ff30d2d : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc1b49ad33ab4e03bef18de88af6aee6 : SUCCESS in 4m 37s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b27cba3dc7654ff7b2c8fbd5fb711429 : SUCCESS in 3h 30m 55s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/82562de2966d4b2393098fb692c9ed46 : SUCCESS in 1h 53m 47s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/16a116df023c4a5f806d6a68b582b5d2 : SUCCESS in 1h 08m 42s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/a5d330e32cdd4d948033efdc5bc7f934 : SUCCESS in 1h 29m 21s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/5e58f196ea49430d9b21c08f49067e9d : SUCCESS in 3h 25m 22s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-wallaby https://zuul.opendev.org/t/openstack/build/60f6bc9902ce4a1a9ee8fc3f1b1a178d : SUCCESS in 1h 28m 06s\n- tripleo-ci-centos-8-undercloud-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/afbccec3462b4eb9bc03d036d49b99e9 : SUCCESS in 1h 36m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/06d0661c50d04732a046a4ce4bf7dfeb : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cbd81e4e52674b47b5f9876211a93614 : SUCCESS in 4m 46s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/f121daf0189f4416b2d2e2ec3dd18482 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b2264f796dfed579c8f9d9faf5cce9e9aed7398f","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-07 11:39:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"94945d8ee904a028ab0e2715878a218052b85dd9","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-01-07 11:48:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bbba4160c511f209ec4d5a20a2ad9e8568d97712","author":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"date":"2022-01-09 16:09:27.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b37e4b7eca24007f61846343ec611dfe6b1232e3","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-01-10 06:25:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"45b5117a27ed759d3a4093c68e8d6b69437301b0","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-01-10 06:59:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c512ccc4b5e7c777656c63069cbc2ce7c34bf612","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 07:00:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9cab705aa509d79b99395cf174d3ade5a8a1a6c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 10:28:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09154558cdfa48f69728831720925ff7 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc1ecef6b882406884fae9052bff7df9 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41e4148e515c4e6eb1198b50cb5bec3b : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c294f26888040429afdc200b3886b75 : SUCCESS in 6m 13s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/1bee8bdf5d8a4602b1f55c237b1b5ab2 : SUCCESS in 3h 27m 33s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5ec4a3f2b97d4c74aa94063977c801a5 : SUCCESS in 1h 56m 20s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1e8eccdb95424458966c1427fdba2b92 : SUCCESS in 1h 38m 43s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/e58498d87e064bdaa6ae78999c937511 : SUCCESS in 1h 48m 19s\n- tripleo-ci-centos-8-content-provider-victoria https://zuul.opendev.org/t/openstack/build/c364a0927e8d4d9398d94dd3ec8bff1d : SUCCESS in 3h 14m 20s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-wallaby https://zuul.opendev.org/t/openstack/build/a5947ebccd624eb98423a27a1eb1d15a : SUCCESS in 1h 34m 59s\n- tripleo-ci-centos-8-undercloud-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/866ba9d8de6e4a71848d46edf5614c80 : SUCCESS in 1h 39m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d107ddf8519b4eedb1e8e31c85c27719 : SUCCESS in 6m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c375c8480cbe44508a63febe778a9b90 : SUCCESS in 3m 45s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/b24fed4d535a48e7b37c15fe8027b2d6 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"e79749308546ef5d6f25e2779a2b407140cc479c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 10:28:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1ae690a7899b20130f7863fe3d8d834771f346ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 10:29:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cf0959b8d8c4ad99ce8f55204cbe381 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b800f471ee3a38651b95280b936e11af136f1a1f","revisions":{"b800f471ee3a38651b95280b936e11af136f1a1f":{"kind":"REWORK","_number":1,"created":"2022-01-06 15:30:08.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/08/823608/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/08/823608/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/08/823608/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/08/823608/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/08/823608/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/08/823608/1"}}},"commit":{"parents":[{"commit":"e374f9139325fa241258390990c15a22a5b62621","subject":"Merge \"Remove redhat-release only if installed\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e374f9139325fa241258390990c15a22a5b62621"}]}],"author":{"name":"Martin Schuppert","email":"mschuppert@redhat.com","date":"2021-06-01 14:12:11.000000000","tz":120},"committer":{"name":"chandan kumar","email":"chkumar@redhat.com","date":"2022-01-06 15:30:08.000000000","tz":0},"subject":"Remove libcgroup-tools from libvirt container","message":"Remove libcgroup-tools from libvirt container\n\nlibcgroup-tools was added to the libvirt container image\nwith I469b4f51c57e1651f42064b0134aa3916af72d80.\n\nSince OCP on OCP is right now discontinued and libcgroup\nis planned to be removed from RHEL 9, lets remove it also\nfrom the libvirt container image.\n\nIf there is the plan to start it again, we could probably\nuse systemd-run instead.\n\nCloses-Bug: #1956600\n\nChange-Id: Ic946d0ef686cb178e0f02db012f2767c9ec5d6d4\n(cherry picked from commit 5e469e42cf06c86713e54885bfcd344a1b40d662)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b800f471ee3a38651b95280b936e11af136f1a1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b800f471ee3a38651b95280b936e11af136f1a1f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
