)]}'
{"id":"openstack%2Ftripleo-common~824092","triplet_id":"openstack%2Ftripleo-common~master~I26129ad0bf40275b6ad580793e071937e3e61b8c","project":"openstack/tripleo-common","branch":"master","topic":"cs9","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-01-13 05:36:30.000000000","reason":"Marios Andreou replied on the change"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-01-14 11:55:04.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2022-01-12 16:42:56.000000000","reason":"Alex Schultz replied on the change"}},"hashtags":[],"change_id":"I26129ad0bf40275b6ad580793e071937e3e61b8c","subject":"Remove remove-machine-id element","status":"MERGED","created":"2022-01-11 08:17:58.000000000","updated":"2022-01-14 11:57:10.000000000","submitted":"2022-01-14 11:55:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"824092-cs9","meta_rev_id":"4f8ba3e16bec8ca50799f5298888ec9de06368b3","_number":824092,"virtual_id_number":824092,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-14 11:55:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-11 11:40:07.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2022-01-13 05:36:30.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2022-01-12 16:42:56.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2022-01-13 11:54:07.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"CC":[{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-11 08:26:53.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-01-11 08:26:53.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-01-11 11:40:07.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-11 12:03:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-13 11:54:07.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2022-01-14 05:39:21.000000000","updated_by":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"reviewer":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"state":"CC"}],"messages":[{"id":"f492dbcc4e2c70a1f420dd4573a7ea36cc2f1056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-11 08:17:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8964695f981af31ea3ddee89f5cd948e05834a0d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-11 11:40:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/59905c91124247208b89814df867467b : SUCCESS in 3h 21m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e6783a3c220d8e5adc301b5fa8e23c9c0a256fa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 12:03:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfc236d407d44581b1bbbd653adf64dc : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c9faed0e9ac48db813dea16856927bd : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da4640b6eb7b400e877b232913ee9f9b : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51b9f8689f314d268149ce731999914a : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/798f78fb16254760bbbb8a8a4573d0f2 : SUCCESS in 4m 24s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ae038b74c4ae40518bbf0686aa351d68 : FAILURE in 29m 48s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/ffbd97c316ba4e2faa44ffa4f1d01bec : SUCCESS in 3h 31m 14s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/c648a40026d6400098a0a857b63a0eb8 : SUCCESS in 1h 42m 21s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f5aeeff439564b5d85ef73f45dde00f3 : SKIPPED (non-voting)\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/652955820fab4720b0e832dcb0b9b569 : SKIPPED (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ba6033f16f2842ef9d623a7b81437edc : SKIPPED (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3e1483f5bb304b25b86f1e8bf2207607 : SUCCESS in 55m 28s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/3bcdfc48263f407e8657dce02fb1a615 : SKIPPED (non-voting)\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/cfb5e561fe6a4af08fe0644f4a7723ca : SUCCESS in 1h 29m 04s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/6bd4450672e749ca8e655da987b122f7 : SUCCESS in 36m 09s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/7fb944ada1064204b4fbd12a13dbd873 : SUCCESS in 51m 16s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/37f18eca09314c51a984ba792bcdb927 : SUCCESS in 40m 13s (non-voting)\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/e9f6bfd76f5341ffbc62b7830bb1cd1a : SUCCESS in 33m 15s (non-voting)\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/dc2d1163ff6844c680a491a68d1805ca : SUCCESS in 41m 10s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ace5ff82e2834039b93e961c8a9ac0b5 : SUCCESS in 1h 35m 38s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/321c17f1cdf94496bc9e54bbd8d00129 : SKIPPED (non-voting)\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/788c07b26e4043fcb19868a27f9a84d6 : SUCCESS in 3h 32m 22s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3338e7803b314a6daff1b13c6f635e7c : SUCCESS in 1h 26m 35s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/46908b10d0e044da97809c578786fcc8 : SUCCESS in 1h 34m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c549204728cc41bd935fd5e0886b37de : SUCCESS in 4m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f644b96bd444e06946d12914e782e67 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/596a0cde5514495d9cc0adf08b1bb6ce : SUCCESS in 4m 22s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/069e698431744c6ea3cf41df342805cb : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"706abff90dbb6db752a32b2fada5bdbfae1d02ae","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-01-12 16:42:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a45a02840e082a66b626d66a5c6975e101ac5d9","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-13 05:36:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f5b96f51a9a5eb052f1a3b98cc91411bc623f0a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-13 09:19:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5e21b19446d44ce6b584a1e29b0c67a3 : SUCCESS in 3h 41m 37s","accounts_in_message":[],"_revision_number":1},{"id":"6f5845acc647e4fb37f270f0df7b2cdb2fad3359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 09:24:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/533dc942699c4e19873a1f6b56936d32 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0535058295ab452b8e4e8f9a9889d5bc : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d35079732f0a4d9c872cbe9354d297bb : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f948875c6f0d4c7788e1b1d9449423ab : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8073af2f0a24105a2a6df070e5ffd99 : SUCCESS in 4m 33s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/8551999a0d6c4011b1ebf9871a5991ee : SUCCESS in 2h 09m 46s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8b7839f4658447cbab5b5f418b64515a : SUCCESS in 3h 35m 30s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/69a16ccf19a641a785adff9bacc2783f : SUCCESS in 1h 52m 08s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/52c21156d01742858d16e8bb65ed9858 : SUCCESS in 1h 14m 29s (non-voting)\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4c2c1c7b06b347b789832ea43d43ccb4 : SUCCESS in 1h 07m 47s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7d70ec4f50b045299014405e658db7d1 : SUCCESS in 1h 06m 46s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/f961bf91aee24818857b6b3b4cff4948 : SUCCESS in 59m 50s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/82b07b38e8ea42f8a616a29c2f81984c : SUCCESS in 37m 14s (non-voting)\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/e1a830dd03ac4a5fb2b403d9a044bfce : SUCCESS in 1h 25m 43s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/81312233e0ab4f9a92b46950925456b4 : SUCCESS in 44m 00s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/ac1a51db484842ffaf72932d93b28ecd : SUCCESS in 52m 19s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/99a98ec980b14c66bbc972963c1a90b0 : SUCCESS in 31m 25s (non-voting)\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/b3c83b2a99ed4d8fb045dc0e594416c5 : SUCCESS in 25m 31s (non-voting)\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/c77908e6b4d24fd8a606f03a4d8fca5c : SUCCESS in 34m 48s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/02b7eb3f601d488cb39d64428a258ec3 : SUCCESS in 1h 45m 38s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ebdbb3e796ec49a6b44dbff163d7e4c1 : SUCCESS in 1h 05m 00s (non-voting)\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d2f63122396b484f9fdfca9064b026a3 : SUCCESS in 3h 04m 31s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/458fd821cf014c108563454f77f50794 : SUCCESS in 1h 24m 17s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/a1abaec59efb4d898fac1f2ceebdc325 : SUCCESS in 1h 15m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d2535cf352974e2786bf826f599008a6 : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4879d3eba66e49fe897e3bb56625a44d : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7c51cec45fa4b04b0bb99d65058e4ad : SUCCESS in 4m 55s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/d7f0893dae4c434191d9eef63e1427be : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"aaf78a260a7ff37638a349fff51616425cd8998d","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-13 11:54:07.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cb0f07783646881df08b5cd8e73d7b125a1f0441","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 11:55:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09c9f9c3467ffbd05109ad23318e3b2524a5b989","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 17:17:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7802e47e12a24f4e971c4931878ab209 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b70892af88e44a74811acecefb560840 : SUCCESS in 9m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8023c4ae27ce4568ba4de56f9ff30a25 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f07651e89c5d4da4ae1b1186616cf464 : SUCCESS in 5m 45s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cde74cc2130b4e2f9e01f4701de66f4d : SUCCESS in 4h 07m 07s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/1cd331929eba40c281f517174e8495b6 : SUCCESS in 1h 53m 03s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ef0cb5be9bab4802a77cfc51c64c736f : FAILURE in 1h 12m 13s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/76baa74c15ae47b7af4b20f336f406e1 : SUCCESS in 57m 41s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/28e3968018d645368681ba1440e50a2a : SKIPPED\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/a72784df994641d8a5c19964ab8b90da : SUCCESS in 1h 19m 18s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/ab05b45ffd6d45c4acd2101dbd89c1c8 : SUCCESS in 42m 41s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/39767e87179d4202adbccf1cea84305c : SUCCESS in 1h 02m 30s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3bb66a1912a642c78beed27e538490c5 : SUCCESS in 2h 17m 27s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/071c1db121a949409c4e40c9b441c97a : SKIPPED\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/266629e39ce44c41af0c966265bac37a : SUCCESS in 3h 09m 40s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3349154d82014aa2a10801a7492c7cf4 : SUCCESS in 1h 28m 20s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/540f2ade3743459e93b8b7122d866ca4 : SUCCESS in 1h 24m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dc60ebf996824f5a905f895ec647c15b : SUCCESS in 6m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/413cb9a74ec9417c873f9b8817f5c3f9 : SUCCESS in 6m 41s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/dee75c850c1749cd8978a2ed7a38fa48 : SUCCESS in 7m 58s","accounts_in_message":[],"_revision_number":1},{"id":"d87575134aadea1c3ab3d99d77ec5d327cc722f8","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-13 19:40:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2ab5ade9a8e285b2ecd17787abd534c1b2564be2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-13 23:10:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f2ac80b0ad134e1f94e78594ad463c5e : SUCCESS in 2h 39m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ed04cd51c0da74ba8534c8e84360b04458f140d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 23:54:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3271292e41674af29e2b10e964a25422 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a8ca6981ef3411a9fca1f859c268c50 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0790a5efa134d66a2fcfac66b759d46 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cf2ba1c4d444b6eab4351fe0dcb0cee : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b1044b14304993aa88b404258d931f : SUCCESS in 4m 32s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/545b59f9dc554b238790d23df99f6991 : SUCCESS in 3h 12m 02s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5ea7647b831d4c6eb5c5964e78dbd38a : SUCCESS in 3h 55m 34s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/f9c02620270146efb5a8c4862d953065 : SUCCESS in 1h 59m 52s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/344d3c5c564f4a21b5e0f58f1374ef7c : SUCCESS in 1h 45m 21s (non-voting)\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/850d89833e194c3e9890da6d850fe3e7 : SUCCESS in 1h 23m 16s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9f86b3eede01497fa63e77fb766bed3e : SUCCESS in 1h 17m 13s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6f5fd5846fcb484891ad4e0b2828642f : SUCCESS in 56m 26s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8654f4e6a2a8455886deb77cb7675f41 : SUCCESS in 47m 50s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/cd4436eb034e47c182ae292e9828ed03 : SUCCESS in 1h 13m 03s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/db484726121448ff853462bbeac6de86 : SUCCESS in 36m 47s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/81dcc85c86fc450e9fc6aeff9cc943f6 : SUCCESS in 50m 00s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/ed7eb220e0e4448ca07f7bf91443ce84 : SUCCESS in 41m 50s (non-voting)\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/cab35315ed72444db03a2e994e0f1137 : SUCCESS in 36m 31s (non-voting)\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/2fa4bacfb15c406280dbb4fb3dc032d8 : SUCCESS in 34m 57s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/1b011dd43e3d4b528ae0f8a5ee5f22b9 : SUCCESS in 1h 28m 40s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/aa5ef1c0f9ae4767b50c15b10119fe91 : SUCCESS in 1h 18m 35s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/edd7489c0d034799bbd31f0dae34d583 : SUCCESS in 3h 26m 51s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1cb74895df2944878ef91fdab0c6f562 : SUCCESS in 1h 23m 06s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/f18e6a2f5de3499da615ad41e7fb7a47 : SUCCESS in 1h 21m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b563af56fef74b48b789cf2343011aa9 : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c0c0df33f19546bd80ed37b952b2899e : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89fe0e06842a49229b1f5ec0528b509d : SUCCESS in 3m 57s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/3c4304558a03409d9d013368f51739e1 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e6d3543f9af0a9d5b60cc0b33b3fac4cf2c0f582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 23:55:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2a1227d8bf9f3fddae545a659e74775978666d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 04:29:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05d7686f5c224e5a840bb92410fcde7a : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc0bd4e1d9e044d99d9d24aa2a1fc522 : SUCCESS in 8m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4f87e0e72cd477e90c3d60dfcab72a9 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e42960b915a4a31b952fe9818c4a637 : SUCCESS in 4m 56s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/adb962042923478f9cacdc5a8fd629d2 : SUCCESS in 3h 35m 04s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b220458f13c5435098ae33ab9876711a : SUCCESS in 1h 54m 00s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/5f6ba73cf0ec42bd92a1db671550ca11 : SUCCESS in 2h 40m 24s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/8e26c17a52294a8fbb99a7333f2b43dc : SUCCESS in 57m 30s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/592f9c0bbc4049e8b54f478eaf2b0931 : SUCCESS in 57m 36s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/a53d7ecaf4354f4591a87c6714fb5b7e : TIMED_OUT in 1h 33m 38s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/a5975fc2735c4a28a6e4e19159a6ccc4 : SUCCESS in 45m 49s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/f05f27d93f00455c9312b2b6d71740ba : SUCCESS in 1h 00m 43s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/3de1070cf6a244d4bf2d09a31beed74a : SUCCESS in 1h 31m 02s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ab6bd37090394dd096c72263e1c99c27 : SUCCESS in 1h 27m 48s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/04749a0f79f64a918074fc3b5e2b1a57 : SUCCESS in 2h 59m 17s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d1bce8565db04b06b360c6e9c8294e6b : SUCCESS in 1h 24m 50s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/bf5da0ee82ea4ab6bf5bbe240e488119 : SUCCESS in 1h 21m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3aa39a8c0e084a6282f47a0ff5f287e6 : SUCCESS in 5m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0ab27d7abc6d41f9a01fcadfbb67b9c0 : SUCCESS in 5m 18s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/4bf9716f60e648bea687e25d36f4e977 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"adfdfce0401310dd7277bc84dd899831b0e1caea","author":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"date":"2022-01-14 05:39:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"099818e0350f2bcb92a75a8ee054f0873a830a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 08:32:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f60e8d305b84475d800b7e6e30a6a371 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b0551273d0a45328be39b7b1c56e3ce : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c0caa99be654f03a4734e8b2bb7a3bf : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32946b904ca849a08cf665126f89f79b : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce687ce76b0741869b32267f24c947a2 : SUCCESS in 4m 38s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/785fa9e186a24491b8f7621d321c0036 : SUCCESS in 1h 55m 11s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/4456a990cda14ce4a3cc42ee99f9b080 : SUCCESS in 2h 41m 18s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/a45efeae84d34b55b1ccd713c84ad22c : SUCCESS in 1h 29m 22s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/0ab45f718a55427384e4612cd0b6c8bf : SUCCESS in 1h 13m 07s (non-voting)\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/cc77b07c6d434ce4a69b3c95af6a4e21 : SUCCESS in 1h 10m 21s (non-voting)\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/aab6e89a99ba4d28a66b80e0c8db4273 : SUCCESS in 1h 02m 28s (non-voting)\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/af70247d415b4f41bb873d3a9cc65dea : SUCCESS in 42m 51s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/0b0bd90a362542478acf587d6e5a88f2 : SUCCESS in 40m 10s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/430a5ac6e09246b4be59f88f95301f75 : SUCCESS in 1h 03m 05s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/bab0f96488714f108d42645c5a0a8d4d : SUCCESS in 31m 51s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/741f51b1632848e9b100277e3ff94666 : SUCCESS in 40m 53s\n- tripleo-buildimage-overcloud-full-centos-9 https://zuul.opendev.org/t/openstack/build/8827575641964c4a82463015dc833122 : SUCCESS in 52m 27s (non-voting)\n- tripleo-buildimage-ironic-python-agent-centos-9 https://zuul.opendev.org/t/openstack/build/e8de73628f104a2382133da276d3dc9b : SUCCESS in 31m 29s (non-voting)\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-9 https://zuul.opendev.org/t/openstack/build/27b17bd252424a67a0b363752c36d382 : SUCCESS in 26m 46s (non-voting)\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/fb58dc53e8ab499fbfda7bb4a3eb7de6 : SUCCESS in 1h 12m 28s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/72edf241251d4f66afc6d44d5cb7b3a1 : SUCCESS in 1h 02m 48s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/9bdf5cf1e7f84f249b79d1f8b1f5ef20 : SUCCESS in 2h 33m 47s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1e283c303ca8434c8ca0112137d0f050 : SUCCESS in 1h 06m 45s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/4955dc4978134133bb3aa600a95857a8 : SUCCESS in 1h 10m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5dcd6829c44844c6a7496c6120e4f543 : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8bc662653f6c4c529cd0c95e6b696998 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/216c1cb7dd164706978901752be4edd2 : SUCCESS in 3m 46s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/59150e37ced14e87bdd244ffd2bf4ccf : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"14c336974255deefcfc8a2573c9022658669b254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 08:33:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e37b6d83b864a2150beaa9478c2aab74f1c459e2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-14 08:35:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2696db89799f49b49e38ba3084b6f57e : SUCCESS in 2h 56m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2efd7720b886c23689d5ffbcf9070b369c0e6d41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:55:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0e708846acb46708034b55f07552b52 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c4f7f819adb4edf9a81d2b5b9fc7038 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81d4944b60be43cab7f633deedd865a9 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f6f1bb3325141ae8bb3c156cb6a4feb : SUCCESS in 4m 34s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8906d22fd27d44fab6e5425c51e19bf4 : SUCCESS in 3h 20m 43s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5a4410fe9ed944f396474f54b2024553 : SUCCESS in 1h 38m 21s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/a23618116a2a497fa9a660729ee182ee : SUCCESS in 2h 14m 22s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6f103d356fa448fbb04fd79176e86853 : SUCCESS in 47m 39s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/b8b6f2cbbd3c4b23bdbed5b5c62845d6 : SUCCESS in 44m 36s\n- tripleo-buildimage-overcloud-full-centos-8 https://zuul.opendev.org/t/openstack/build/d36cdfd2fe12481d875ea37c185d0672 : SUCCESS in 1h 03m 09s\n- tripleo-buildimage-ironic-python-agent-centos-8 https://zuul.opendev.org/t/openstack/build/d470662b88cf493595864f6bb29b5fbf : SUCCESS in 43m 03s\n- tripleo-buildimage-overcloud-hardened-uefi-full-centos-8 https://zuul.opendev.org/t/openstack/build/af4750ec56e449c3816d5b7bc77043a7 : SUCCESS in 47m 43s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/9b02e1e221aa462fb8243c6c81db7282 : SUCCESS in 1h 26m 44s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/722584d33f0c4b69a6f55954303a4588 : SUCCESS in 1h 12m 56s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/20319ac420ec41ecb5eec12386d65334 : SUCCESS in 2h 58m 07s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6135d20b1bc14c338ebc44e3a111dab6 : SUCCESS in 1h 19m 07s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/071630c49d804687afcd13fc6894ec5c : SUCCESS in 1h 14m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0738a50abfbf44a78e8c4764f46c5ad5 : SUCCESS in 4m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/85befae8faaa4f17a6b3509b06c58a48 : SUCCESS in 4m 52s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/e779eaaed6e743c49552a77eea21fc50 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6cc92c7a7706f4608fb058437e58bbe3ba86ab28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:55:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4f8ba3e16bec8ca50799f5298888ec9de06368b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 11:57:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/422f46f88e5d46c8aa71d2c993d0ec19 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8df691d0f781bf4d2c95b1925a135f8e1fecab51","revisions":{"8df691d0f781bf4d2c95b1925a135f8e1fecab51":{"kind":"REWORK","_number":1,"created":"2022-01-11 08:17:58.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/92/824092/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/92/824092/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/92/824092/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/92/824092/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/92/824092/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/92/824092/1"}}},"commit":{"parents":[{"commit":"6688b56221842eb40b0191243640dc96f4c14ff9","subject":"Don\u0027t enable module stream in rhel9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6688b56221842eb40b0191243640dc96f4c14ff9"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-11 08:17:10.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-11 08:17:10.000000000","tz":330},"subject":"Remove remove-machine-id element","message":"Remove remove-machine-id element\n\nhttps://review.opendev.org/c/openstack/diskimage-builder/+/818851\nadds the same in DIB and\nwe no longer need this hack\nhttps://bugs.launchpad.net/tripleo/+bug/1954456/comments/22\n\nRelated-Bug: #1954456\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: I26129ad0bf40275b6ad580793e071937e3e61b8c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8df691d0f781bf4d2c95b1925a135f8e1fecab51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8df691d0f781bf4d2c95b1925a135f8e1fecab51"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
