)]}'
{"id":"openstack%2Ftripleo-common~825038","triplet_id":"openstack%2Ftripleo-common~master~Ia06c6770ea567656b5ce380826bdb120ca32a1a8","project":"openstack/tripleo-common","branch":"master","topic":"cs9","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-01-19 13:08:48.000000000","reason":"Marios Andreou replied on the change"},"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-01-18 12:48:08.000000000","reason":"Douglas Viroel replied on the change"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-01-19 00:54:56.000000000","reason":"Takashi Kajinami replied on the change"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-01-21 08:17:33.000000000","reason":"Chandan Kumar replied on the change"},"10969":{"account":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"last_update":"2022-01-18 11:22:53.000000000","reason":"Shnaidman Sagi replied on the change"},"32458":{"account":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"last_update":"2022-01-18 11:23:00.000000000","reason":"Ananya Banerjee replied on the change"},"29775":{"account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"last_update":"2022-01-18 11:04:51.000000000","reason":"Sandeep Yadav replied on the change"}},"hashtags":[],"change_id":"Ia06c6770ea567656b5ce380826bdb120ca32a1a8","subject":"Reinstall centos-stream-release from centos repos","status":"MERGED","created":"2022-01-18 05:50:42.000000000","updated":"2022-01-21 16:20:24.000000000","submitted":"2022-01-21 16:18:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":30,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"825038-cs9","meta_rev_id":"4c10967f354db233d5cec7479c60013462487d12","_number":825038,"virtual_id_number":825038,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-21 16:18:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":0,"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2022-01-19 13:08:48.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":1,"date":"2022-01-19 13:43:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":2,"date":"2022-01-20 08:11:04.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"value":1,"date":"2022-01-20 08:11:04.000000000","_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-18 07:18:42.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2022-01-18 07:18:42.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"state":"REVIEWER"},{"updated":"2022-01-18 07:18:42.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"state":"REVIEWER"},{"updated":"2022-01-18 07:18:42.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2022-01-18 07:18:42.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-01-18 09:37:29.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"REVIEWER"},{"updated":"2022-01-18 10:47:57.000000000","updated_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"reviewer":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"state":"REVIEWER"},{"updated":"2022-01-18 12:20:18.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 12:27:43.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2022-01-18 12:50:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-18 14:53:44.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-01-19 00:54:56.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"d3dbcd10c8b6c30c4de89ee7adac167939d795e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 05:50:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3589a9827cf7cbc07c38278ec24e813e704acd35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 06:33:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f88340e96d47af0371b9fe2edbb744df19253a17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 07:45:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"459e4bd45fadfd1a52e587fa5d330aa7d3d52875","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-18 08:16:42.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"97360e1f5ae2596363500ec2df183578ec01701e","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 08:18:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4d080b519184908c162008ea9260622c0b14e1d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 08:26:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d5b02ae7d80ee71183eb69dd5c90825f30611a1a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 08:27:39.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3d74301e82b5c142d41558fd55ea66530254b4bd","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 09:20:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"93c336b354d81206e9a8e67e601fd927835d726a","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 09:24:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"70a7a5a367762f66dc9aaf94fcb8a41ae3c851e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 09:35:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"24d0d9a91ba3501d022d5d957a809d51d15ea4b3","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-01-18 09:37:29.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ec147b13b6b78728cb7f34e0695eb0c911c182af","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 10:47:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"06e8e96a93ca0991c4c1352431241252fe967a21","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-01-18 10:47:57.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a642f121c090309fa775308b581486a07a240d54","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-18 10:52:51.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"23bdb84a03252672a8cd77d0113498ef833a5852","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-01-18 11:04:51.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4887020a4d350856511b5b6299d9d8b635670a9e","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2022-01-18 11:22:53.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"1d41c05c87c09a21dafb217d2a151a230372040d","author":{"_account_id":32458,"name":"Ananya","display_name":"Ananya Banerjee","email":"anbanerj@redhat.com","username":"frenzyfriday"},"date":"2022-01-18 11:23:00.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bff6151fa59fa55418e236e1ea3ffee8f58c8483","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-18 12:20:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e5003874d3fb4f5496bcdceaedba1840 : SUCCESS in 2h 44m 01s","accounts_in_message":[],"_revision_number":5},{"id":"49721f8231b948ec52491ef512bdf24e2fa05af1","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-18 12:27:43.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"f031f0a52a8d88e5f605a68c8955d87c8d4a8e51","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2022-01-18 12:48:08.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"be2b4ad043b184ad48f495717e6a4bfc964a92c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 12:50:11.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56759cf0b2a9469f865b5484d2932cc5 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a02da9ab487471788d66ad9901e654c : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c16d7b90deb4467591039ea55fc3c3be : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/955ee147591d4bda9a327fd089450f81 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b6e2868f1e84de7983900de93c73aa3 : SUCCESS in 5m 13s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/3c102f8af41648f9ba5745e96c450d6c : SUCCESS in 2h 56m 05s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/25dca809398c48729d6e7c47cb03d034 : SUCCESS in 2h 44m 52s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b717f74e1eca4738b7d99c6d369679fe : FAILURE in 50m 20s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/6ed44527be6c4c65860ddc309588557b : SUCCESS in 1h 39m 59s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3fc6fcce30c244fdb1d16c189e092ec3 : SUCCESS in 1h 31m 04s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4286533ce9904cce9ed1b5ea58798a5d : SUCCESS in 1h 30m 21s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/0eae2e6ec33a4037bc2d4a9ac3b144e7 : FAILURE in 47m 04s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/4927ffc5269f46a59f4adff89c655bcb : SUCCESS in 52m 03s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ccae626e25f44abcb3768c07abe11719 : FAILURE in 52m 16s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/119bc680c6ef43988d72c4e480e44faf : SUCCESS in 1h 24m 20s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/75815ef3aced4ed28f397ec60e116aaf : SUCCESS in 2h 50m 17s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f1af8e45dfc447eda007b22ebef9a116 : FAILURE in 52m 29s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/914f4468a01043008990a92bfd3035b9 : FAILURE in 1h 17m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dc418bac27d048508ed8704eb3586131 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5614d9e41fc049428157748a92566077 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b996f20f3d5e4d168a8edb5da215a6f3 : SUCCESS in 6m 02s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/709bc47e9e864a6f8e73ab2d065fe5f6 : SUCCESS in 7m 02s","accounts_in_message":[],"_revision_number":5},{"id":"66024af4fa781cfb3a7bb89008d39a8ee962d2b8","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 12:57:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"39323668cea9c69bb9c9f303dc73599a317828ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 13:46:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8029d7d479365a5fc4cb5a978e044ef024dd34ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 14:46:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"a1a4e0fa2c5ebbf2e1accb9d5ae0bdc162f16798","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-18 14:53:44.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"523ee022b058c1269c28d9adbeed5577927f9033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 15:45:55.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0546afccbc22fe4e47172a97c180af4741c3c4f2","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-18 15:46:43.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c585dcdff07f9d3ebdfe34fd319a79bcb9724929","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-01-18 18:53:13.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c7c7bf55d6224832acb7bfd6fbb0a1c3 : SUCCESS in 3h 03m 53s","accounts_in_message":[],"_revision_number":8},{"id":"a3e2f1dbbbe54a21185ba0ee7198bfb9f6a70be8","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-18 19:02:50.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"cb6b8e667a72a14008e5351e2bcc05cbf8a0617e","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-18 19:02:54.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"fe11f9214955dff961b05ee7f62f14f37c7cc098","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-18 20:05:28.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28d59d3f600e453a84f2630991d4a319 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6dc8b693d974a49b082a8193b6d71ff : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d88b958a76c4cf6b342556389ed1e9a : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da21225916634f9cbc3237da658cf21d : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e24240304862435d911d0e3183f99913 : SUCCESS in 6m 03s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/05223fd1c1f6422a8ca852aa9359e57d : SUCCESS in 2h 51m 56s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d6fac491311c41c98d3e897fbfd66b3d : SUCCESS in 3h 12m 24s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/904eb16ae7604f139c2af0b2e9c4e423 : SUCCESS in 1h 43m 39s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/5814b1dc89d04dff9d7270164a04ae38 : SUCCESS in 1h 37m 09s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/1ba55692117a4109b24d9c5c22336cd3 : SUCCESS in 1h 28m 02s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/82edce49043544c684498b9ba9ce0994 : SUCCESS in 1h 15m 57s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/c2a59fadabcc41d8aa98d70fd7ae808a : SUCCESS in 46m 25s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/73b4b171513c4c3ebbe1d2a5df39d2ad : SUCCESS in 43m 49s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c7245259fc894c999fd9e93550965cfe : SUCCESS in 1h 23m 40s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/d11d63b67e4c4c27a502246bb46e9c3c : SUCCESS in 1h 12m 48s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/263a6e6cac104470988583859de87f33 : SUCCESS in 3h 42m 22s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d255940814c74494ba010625fa710790 : SUCCESS in 1h 21m 31s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/c08afc4c0cec4099b24a250f35da1d4a : SUCCESS in 1h 19m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ed7ba107111c40ba8d0dee01fd52391a : SUCCESS in 5m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/87a5e56b27f545d9909d5035f2b9890e : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6001061b073e440a8b3aadc54d87db40 : SUCCESS in 4m 43s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/256e31ff319d48dd8512adbdb9cce184 : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":8},{"id":"8211e9d08ae9d23c4ba255627070a5c0397b1625","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 00:54:56.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0899f75a580019e41a5203d2fbc6be299ce29e53","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-19 08:15:04.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"7394cce7ed38d795ce9c318fab0d2563aafaff27","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-19 08:22:00.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"614be717ea72567fff48b14657bf08d5d7cdea97","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-19 11:46:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"4166f7f5459043a5415f8cc3fe2ede21ec3f6ca9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-19 13:03:25.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"33a52bce72f5a94341cdb57b42941c8f612d8d80","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-19 13:03:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"525f0b0edd9ad425ab9ef9c634f2b3cf41cd7938","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-01-19 13:08:48.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0d7a2b0be9e2962c214969f952373e34bc84049d","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-01-19 13:43:25.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"aa422030ef58737059b4c418baf2fea325dc67cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 16:32:08.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7e8a8b185464bcd8790c2aed1dd840e : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/806c28ffa09147f58119cfba860f3b5b : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ed131c53b124ccdbebabd5ab2d196ff : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd44e237dca94d1ba16842825466f8fe : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e2beb2ba16f4bc98dba083e77d13863 : SUCCESS in 4m 13s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/25890ab09f9b43a0a2607ff7a1a352b7 : SUCCESS in 3h 03m 29s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e536d34dff1c45299c589c505ca2588f : SUCCESS in 3h 15m 28s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/25d15a2d6662426caceb488652f5128f : SUCCESS in 1h 39m 12s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/9ec40d32d6e840e49723819dad15835e : SUCCESS in 1h 30m 13s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5f491e25e3d34cb8992382248f4b7bbf : SUCCESS in 1h 18m 00s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d3fa9ec6c3f34f38b42fe757ecb13d24 : SUCCESS in 1h 21m 56s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/fec23971fe684f93a222e31edf7f418d : SUCCESS in 51m 42s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/bc48fc56a7554957a3e841eb5ff3cdc5 : SUCCESS in 38m 56s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/233d8fdfdc5d47068fc340de9aa9b88e : SUCCESS in 1h 21m 45s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/13dd6d45fa7643ec8770a0f86324785d : SUCCESS in 1h 11m 58s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/519fffabecf841b09ca52fd8342a53be : SUCCESS in 3h 07m 07s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/19b13f2488874b8589ddb5cadff0cc8d : SUCCESS in 1h 20m 33s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/a613a337b7e44a13b8d4db1068fa1d3e : SUCCESS in 1h 19m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6ba2302ae9b04df38b9713c19fee3b13 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b79085a543374122b18ba6cd6f97bee2 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fafc7fe2168044f99af212afdaa3af70 : SUCCESS in 4m 26s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/c579e0d1be4a4a8f8c8f8483f0196eaf : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":9},{"id":"20db4f192f17884aac3809fdc66af03e1684de26","author":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"},"date":"2022-01-20 08:11:04.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"6ad3a4eb527a139ed53f9ff35bcd9ffaee849ef2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 08:11:47.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"25e71826b1ed53f4a7dac14f7bf6b0824b27ed96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 10:20:23.000000000","message":"Patch Set 9: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7707f315771d43b799b93eed33db67e7 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b5ef3651679e45f38f092711294ef40a : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a21b84e066544d29c1461e76b22cc75 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/028046b3f0eb483d8461cf07db06d2e2 : SUCCESS in 4m 15s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/aeb1fd5ec2754594985316dfc78138cc : SUCCESS in 2h 08m 24s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c4283f3069524da19b40c76fecbe0a33 : RETRY_LIMIT in 7m 13s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f61dbabdd663404580204bbe49a5d611 : SUCCESS in 1h 20m 02s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/64433fd458064528afc1725c50dde6cb : SKIPPED\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/af9a969875184c328c0d2f088cb67bd9 : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/12e67814adbd4ed5961ee38bc8d30625 : SUCCESS in 35m 50s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b8c092e2e9554003a8d65d9376c92e0f : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/f0138ce387a643eba87904cf10f74c9f : SUCCESS in 59m 33s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/2ec6ee28444e4debaf2841fc47963b2c : RETRY_LIMIT in 5m 07s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e3161963bce2435fa938e754edc4e1eb : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/6f45dd8272f74bbebb96dee124ce26c8 : SKIPPED\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/527299c78cdc4cd89800c8111fd18e0f : SUCCESS in 4m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1d6411c2b10440c3b29a47760506c2c3 : SUCCESS in 4m 20s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/9531e86e7c48472485d39258803a529b : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":9},{"id":"2ebc6300f04d63e7a296ff58ac0a6e8f4105f434","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2022-01-20 10:39:59.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ff8bef4a1bcee16bd79c4ed8aa9456fab6150b8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 13:36:12.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/005448b6146e4cef9b03c4fc039b3fa7 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/198e250adae444dba2afe903d620f6c9 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea96d83146ef4885bea94d2696fa3e65 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b978e3edbb74e3d99295e1c99ec15f5 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cddff134306d41ee956d77f7c1767cc5 : SUCCESS in 5m 46s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/f6f2abdd769d4d33a90e288098bcd02b : SUCCESS in 2h 54m 19s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e96fa6aeec4449d6b5bae3a2705a5d2c : RETRY_LIMIT in 6m 00s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/291594620a8243809155da773f849761 : SKIPPED\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/c23c3ce2179749f299c830e9576aa70d : FAILURE in 1h 45m 53s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5f8bcf792af849d59f7c24368911115c : SUCCESS in 1h 25m 45s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/0270b1b5419b4c88a992e0c2de0b9c3d : SUCCESS in 1h 11m 38s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/03759dc7f20b4dd6b123fdd1e3e5644d : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/937a16ea1d214126bd738be32f0d3a4c : SUCCESS in 46m 15s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/03c843e8b6224d4dadc1e0269ddc68b9 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/640b716b820145668e9f04e058e95541 : SUCCESS in 1h 16m 54s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3d6a23b14829410881e2f068e050415d : RETRY_LIMIT in 5m 14s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6dc7737a524f45d1ab52213d7e438189 : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/35ab419d21bb4a98b3c9e1ae65ebb8b4 : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/82ebacd3f9e146139279f268f1029309 : SUCCESS in 4m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5970ddb9f9264eca97d80f2a39aab444 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00621119d40349b19e27e407331efbb6 : SUCCESS in 4m 27s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/93b5194bac3743c8a16d5790d502dfdd : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":9},{"id":"41cf4370d28f3e02589d628814a8cd4b3abf9726","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-21 05:11:13.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"c4ecb22023bd39b8beb91921042bd8a846245908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 08:05:03.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/018cf482df1a46c5ae362ff383833704 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/731a22edf3e046198359678a0d09014b : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ca152e126e04f5298a55a992cde4b16 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30d2683506df474288c296b77e9bd547 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc5b683794e94b0aac56af8eac6628ad : SUCCESS in 3m 39s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/4b6babb470964821bf1bed4b2321e4e5 : SUCCESS in 2h 52m 34s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b65072ce867e4ecdb095affddd63939f : RETRY_LIMIT in 5m 02s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/e78df48f6d19412bbb4bc8738f3d4147 : SKIPPED\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/b8cffd3084cf4ba795a3fb58f1c5d52a : SUCCESS in 1h 32m 49s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a081bd4ecce14f6bb58e52c04c2e1bc9 : SUCCESS in 1h 34m 14s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/871d2e2910744dbd8e22f73efed86c7d : SUCCESS in 1h 22m 18s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/6db8b5b597504c62ba64ab73e75bd131 : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8b31b25b853646f1ac37385df7a5cefa : FAILURE in 43m 26s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ba1435ef722142b88692f002adb041c3 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/4d7c0746bdc2478c8f32cf22fce012e8 : SUCCESS in 1h 18m 33s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/1de738f0af4643b6a3d0b3ce129f7c27 : RETRY_LIMIT in 5m 53s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f857011e0970420e807d5140604c7c34 : SKIPPED\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/7a7bcecb6465407499d54c83eb16cc58 : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b062ac84a8964f71a33840d19df360e0 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2e6b92f096a14c6489600b6cf04b5d93 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4fbf11b4aa1409092d2afc050ea1c25 : SUCCESS in 5m 22s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/21403c26516d4bfa9162c2300d609c75 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":9},{"id":"6f81804cdfb74fa391de9ab8a2cac4d1f169938f","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-01-21 08:17:33.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"14a919628fb790cfac5d0db1d295572571f187f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 11:54:56.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df54dcd35c794da5b178f2b16fcd29bc : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e4cb7f6be1c46aba0ae672951db3e37 : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37980a67a8d74d84bec2dfda075395fc : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23d88c7c934f48a0906057c45e588185 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bf6d9dbac384ecdadaae19b030637a4 : SUCCESS in 4m 58s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ca99d7463807498dbb0eae27ff6168f5 : SUCCESS in 3h 03m 06s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5f058b3f56fe4022ac7db9abc71234d5 : SUCCESS in 3h 24m 27s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/5bd8518b599d4df7a6553661c5058654 : SUCCESS in 1h 45m 31s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/86da563c4aed4e93b68220d87bfe3b55 : SUCCESS in 1h 39m 16s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8f07a70400844404ad6ee4f93d481fcd : SUCCESS in 1h 31m 38s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9973255f8fd64d09a8ea28194cf97a22 : SUCCESS in 1h 21m 35s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/01eccec712f142018f7636835af7467d : SUCCESS in 53m 27s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/51370cdbd4d04acc8c2477ca358b0c11 : SUCCESS in 48m 26s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/dc8545d7c0cb4e4bbd13879b7520557b : SUCCESS in 1h 34m 15s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/88f701ef558b499682ffc277f051032c : SUCCESS in 1h 13m 25s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/1ac5e3f7e34a4c8381b4e8fc9f63ed11 : SUCCESS in 3h 12m 50s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b309e537352547a9845af90670f14847 : SUCCESS in 1h 23m 13s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/b3f74f18c05b42b292e7d2d3ab1c4e9d : SUCCESS in 1h 19m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3b120adccf0a41f68d7b64174badbcbd : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/027fd3106e0c4e0583491369ab9a2ecc : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c7fff901e624ddfb1bde5f79e16ef5f : SUCCESS in 4m 31s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/dbfd0bc10a624870992daf0d27bfdb90 : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":9},{"id":"6c00d00471f4b35ea3f5445f0916b362b45a0ff8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 11:55:44.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"68cc808eebd4a9e7826be4f442d145d5222ff3ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 16:18:19.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/714b1289535d44aa835bf4a5e6597ee3 : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/296d5c12c0de4a85b3a4c46730a4ca14 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eca4ad2ee2cd43d09f8156298bbdd79d : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/609d763693cc477f8a967bbd21733158 : SUCCESS in 4m 18s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/61bbe26f7aa3455489ac58e18fe04175 : SUCCESS in 3h 03m 16s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/aebbb54a50b14cca9ee6cb9acd5803b0 : SUCCESS in 3h 45m 04s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/8c8004806767463da951d937270eea4c : SUCCESS in 1h 47m 29s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/6837f44c44ba4b559ce7304132c6403c : SUCCESS in 2h 00m 30s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ef812fd3b3c742008ae242419a691eb6 : SUCCESS in 1h 03m 29s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/716b2514e2f149cda197990b0d54f381 : SUCCESS in 50m 12s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f2d78bd6c5be4e2fbf9c2afb1876c3a7 : SUCCESS in 1h 37m 49s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/5e6c20e29b234144ab920e4db45d90b8 : SUCCESS in 1h 16m 21s\n- tripleo-ci-centos-8-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/8bfc277af02f45fcafd4c87d2af55e7b : SUCCESS in 3h 07m 13s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/88245eeff6d6475080ee701990dca11a : SUCCESS in 1h 31m 19s\n- tripleo-ci-centos-8-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/34271a3d04d341549ff07405f128af19 : SUCCESS in 1h 24m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/996cbf2008ad4ab5ad1d17c7859533ed : SUCCESS in 4m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/13d940c5f0f84b97993c0c88a8fb2f3f : SUCCESS in 4m 24s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/4102da1b4bb84ea1ae3b72d49dec587a : SUCCESS in 5m 24s","accounts_in_message":[],"_revision_number":9},{"id":"b4ae6c366a7cb90b8f02d659a6c9db6cd0e3a138","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 16:18:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"4c10967f354db233d5cec7479c60013462487d12","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-21 16:20:24.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b6cc926ba5412eae17fa40e2650f1d : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"b09fb3fef0896e812ca9ce3e6dc74f297494203a","revisions":{"a873ea95d2b691afbc1130d5510a867bef6116dd":{"kind":"REWORK","_number":1,"created":"2022-01-18 05:50:42.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/38/825038/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/825038/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/1"}}},"commit":{"parents":[{"commit":"ebf54cb5a0cdd13efb57bbd137e317bab2765b97","subject":"Merge \"Remove selinux-permissive from rhel8 image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ebf54cb5a0cdd13efb57bbd137e317bab2765b97"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 05:50:07.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 05:50:07.000000000","tz":330},"subject":"[WIP] Install centos-stream-release","message":"[WIP] Install centos-stream-release\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ia06c6770ea567656b5ce380826bdb120ca32a1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a873ea95d2b691afbc1130d5510a867bef6116dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a873ea95d2b691afbc1130d5510a867bef6116dd"}]},"branch":"refs/heads/master"},"5a76fa0161a33aa917b3ffa64f714ca1917b56df":{"kind":"REWORK","_number":2,"created":"2022-01-18 06:33:47.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/38/825038/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/825038/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/2"}}},"commit":{"parents":[{"commit":"ebf54cb5a0cdd13efb57bbd137e317bab2765b97","subject":"Merge \"Remove selinux-permissive from rhel8 image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ebf54cb5a0cdd13efb57bbd137e317bab2765b97"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 05:50:07.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 06:33:32.000000000","tz":330},"subject":"Install centos-stream-release from centos repos","message":"Install centos-stream-release from centos repos\n\nIn CentOS Stream 8 and 9 container image, centos-stream-release\npackage is already available in the container image. But in UBI,\nit is not available. It can be installed from the base OS repo.\n\nIt removes the logic of downloading the centos-stream-release\nrpm and installing it. As it was added earlier when we were\nadding the initial support for EL9.\n\nCloses-Bug: #1958202\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ia06c6770ea567656b5ce380826bdb120ca32a1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5a76fa0161a33aa917b3ffa64f714ca1917b56df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5a76fa0161a33aa917b3ffa64f714ca1917b56df"}]},"branch":"refs/heads/master"},"bdcb27639d6b282d730d410e4bba722aa3034c29":{"kind":"REWORK","_number":3,"created":"2022-01-18 07:45:29.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/38/825038/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/825038/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/3"}}},"commit":{"parents":[{"commit":"ebf54cb5a0cdd13efb57bbd137e317bab2765b97","subject":"Merge \"Remove selinux-permissive from rhel8 image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ebf54cb5a0cdd13efb57bbd137e317bab2765b97"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 05:50:07.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 07:44:44.000000000","tz":330},"subject":"Install centos-stream-release from centos repos","message":"Install centos-stream-release from centos repos\n\nIn CentOS Stream 8 and 9 container image, centos-stream-release\npackage is already available in the container image. But in UBI,\nit is not available. It can be installed from the base OS repo.\n\nIt removes the logic of downloading the centos-stream-release\nrpm and installing it. As it was added earlier when we were\nadding the initial support for EL9.\n\nsince redhat-release is a part of centos-stream-release\npackage so we donot need to install explicitly.\n\nCloses-Bug: #1958202\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ia06c6770ea567656b5ce380826bdb120ca32a1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bdcb27639d6b282d730d410e4bba722aa3034c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/bdcb27639d6b282d730d410e4bba722aa3034c29"}]},"branch":"refs/heads/master"},"d8b334ec27ec15e698f24d724fa880c2a0e75d21":{"kind":"REWORK","_number":4,"created":"2022-01-18 08:26:58.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/38/825038/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/825038/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/4"}}},"commit":{"parents":[{"commit":"ebf54cb5a0cdd13efb57bbd137e317bab2765b97","subject":"Merge \"Remove selinux-permissive from rhel8 image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ebf54cb5a0cdd13efb57bbd137e317bab2765b97"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 05:50:07.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 08:26:40.000000000","tz":330},"subject":"Install centos-stream-release from centos repos","message":"Install centos-stream-release from centos repos\n\nIn CentOS Stream 8 and 9 container image, centos-stream-release\npackage is already available in the container image. But in UBI,\nit is not available. It can be installed from the base OS repo.\n\nIt removes the logic of downloading the centos-stream-release\nrpm and installing it. As it was added earlier when we were\nadding the initial support for EL9.\n\nsince redhat-release is a part of centos-stream-release\npackage so we donot need to install explicitly.\n\nCloses-Bug: #1958202\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ia06c6770ea567656b5ce380826bdb120ca32a1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d8b334ec27ec15e698f24d724fa880c2a0e75d21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d8b334ec27ec15e698f24d724fa880c2a0e75d21"}]},"branch":"refs/heads/master"},"915fc4c2d7851a4bcad6fe8b3ed02f380c3db5f5":{"kind":"REWORK","_number":5,"created":"2022-01-18 09:35:37.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/38/825038/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/825038/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/5"}}},"commit":{"parents":[{"commit":"ebf54cb5a0cdd13efb57bbd137e317bab2765b97","subject":"Merge \"Remove selinux-permissive from rhel8 image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ebf54cb5a0cdd13efb57bbd137e317bab2765b97"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 05:50:07.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 09:31:29.000000000","tz":330},"subject":"Install centos-stream-release from centos repos","message":"Install centos-stream-release from centos repos\n\nIn CentOS Stream 8 and 9 container image, centos-stream-release\npackage is already available in the container image. But in UBI,\nit is not available. It can be installed from the base OS repo.\n\nIt removes the logic of downloading the centos-stream-release\nrpm and installing it. As it was added earlier when we were\nadding the initial support for EL9.\n\nsince redhat-release is a part of centos-stream-release\nIn UBI 8, redhat-release is a seperate package, so it\u0027s rpm\nneeds to be removed otherwise it will end up with skip broken\npackage.\n\nCloses-Bug: #1958202\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ia06c6770ea567656b5ce380826bdb120ca32a1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/915fc4c2d7851a4bcad6fe8b3ed02f380c3db5f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/915fc4c2d7851a4bcad6fe8b3ed02f380c3db5f5"}]},"branch":"refs/heads/master"},"69f2a78e259e1ec4951a67eeb704dd23a2db89c0":{"kind":"REWORK","_number":6,"created":"2022-01-18 13:46:29.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/38/825038/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/825038/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/6"}}},"commit":{"parents":[{"commit":"ebf54cb5a0cdd13efb57bbd137e317bab2765b97","subject":"Merge \"Remove selinux-permissive from rhel8 image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ebf54cb5a0cdd13efb57bbd137e317bab2765b97"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 05:50:07.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 13:45:01.000000000","tz":330},"subject":"reInstall centos-stream-release from centos repos","message":"reInstall centos-stream-release from centos repos\n\nIn CentOS Stream 8 and 9 container image, centos-stream-release\npackage is already available in the container image. But in UBI,\nit is not available. It can be installed from the base OS repo.\n\nIt removes the logic of downloading the centos-stream-release\nrpm and installing it. As it was added earlier when we were\nadding the initial support for EL9.\n\nsince redhat-release is a part of centos-stream-release\nIn UBI 8, redhat-release is a seperate package, so it\u0027s rpm\nneeds to be removed otherwise it will end up with skip broken\npackage.\n\nOn removing redhat-release, /etc/os-release file also gets deleted.\nSo we need to first remove redhat-release rpm if present and then\nreinstall centos-stream-release package.\n\nCloses-Bug: #1958202\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ia06c6770ea567656b5ce380826bdb120ca32a1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/69f2a78e259e1ec4951a67eeb704dd23a2db89c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/69f2a78e259e1ec4951a67eeb704dd23a2db89c0"}]},"branch":"refs/heads/master"},"165dd6c1de536b382e3d5ca078348eff806d172b":{"kind":"REWORK","_number":7,"created":"2022-01-18 14:46:29.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/38/825038/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/825038/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/7"}}},"commit":{"parents":[{"commit":"ebf54cb5a0cdd13efb57bbd137e317bab2765b97","subject":"Merge \"Remove selinux-permissive from rhel8 image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ebf54cb5a0cdd13efb57bbd137e317bab2765b97"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 05:50:07.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 14:46:03.000000000","tz":330},"subject":"Reinstall centos-stream-release from centos repos","message":"Reinstall centos-stream-release from centos repos\n\nIn CentOS Stream 8 and 9 container image, centos-stream-release\npackage is already available in the container image. But in UBI,\nit is not available. It can be installed from the base OS repo.\n\nIt removes the logic of downloading the centos-stream-release\nrpm and installing it. As it was added earlier when we were\nadding the initial support for EL9.\n\nsince redhat-release is a part of centos-stream-release\nIn UBI 8, redhat-release is a seperate package, so it\u0027s rpm\nneeds to be removed otherwise it will end up with skip broken\npackage.\n\nOn removing redhat-release, /etc/os-release file also gets deleted.\nSo we need to first remove redhat-release rpm if present and then\nreinstall centos-stream-release package.\n\nCloses-Bug: #1958202\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ia06c6770ea567656b5ce380826bdb120ca32a1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/165dd6c1de536b382e3d5ca078348eff806d172b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/165dd6c1de536b382e3d5ca078348eff806d172b"}]},"branch":"refs/heads/master"},"c5aeb58cf868811a3be6b2becc9c8d2a05864e4b":{"kind":"REWORK","_number":8,"created":"2022-01-18 15:45:55.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/38/825038/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/825038/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/8"}}},"commit":{"parents":[{"commit":"ebf54cb5a0cdd13efb57bbd137e317bab2765b97","subject":"Merge \"Remove selinux-permissive from rhel8 image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ebf54cb5a0cdd13efb57bbd137e317bab2765b97"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 05:50:07.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 15:44:15.000000000","tz":330},"subject":"Reinstall centos-stream-release from centos repos","message":"Reinstall centos-stream-release from centos repos\n\nIn CentOS Stream 8 and 9 container image, centos-stream-release\npackage is already available in the container image. But in UBI,\nit is not available. It can be installed from the base OS repo.\n\nIt removes the logic of downloading the centos-stream-release\nrpm and installing it. As it was added earlier when we were\nadding the initial support for EL9.\n\nsince redhat-release is a part of centos-stream-release\nIn UBI 8, redhat-release is a seperate package, so it\u0027s rpm\nneeds to be removed otherwise it will end up with skip broken\npackage.\n\nOn removing redhat-release, /etc/os-release file also gets deleted.\nSo we need to first remove redhat-release rpm if present and then\ninstall centos-stream-release package.\n\nIt also removes dnf-core-plugins package installation as we donot\nneed to download the centos-stream-release rpm.\n\nCloses-Bug: #1958202\nRelated-Bug: #1958240\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ia06c6770ea567656b5ce380826bdb120ca32a1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c5aeb58cf868811a3be6b2becc9c8d2a05864e4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c5aeb58cf868811a3be6b2becc9c8d2a05864e4b"}]},"branch":"refs/heads/master"},"b09fb3fef0896e812ca9ce3e6dc74f297494203a":{"kind":"REWORK","_number":9,"created":"2022-01-19 13:03:25.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/38/825038/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/38/825038/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/38/825038/9"}}},"commit":{"parents":[{"commit":"ebf54cb5a0cdd13efb57bbd137e317bab2765b97","subject":"Merge \"Remove selinux-permissive from rhel8 image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/ebf54cb5a0cdd13efb57bbd137e317bab2765b97"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-18 05:50:07.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-01-19 13:03:09.000000000","tz":330},"subject":"Reinstall centos-stream-release from centos repos","message":"Reinstall centos-stream-release from centos repos\n\nIn CentOS Stream 8 and 9 container image, centos-stream-release\npackage is already available in the container image. But in UBI,\nit is not available. It can be installed from the base OS repo.\n\nIt removes the logic of downloading the centos-stream-release\nrpm and installing it. As it was added earlier when we were\nadding the initial support for EL9.\n\nsince redhat-release is a part of centos-stream-release\nIn UBI 8, redhat-release is a seperate package, so it\u0027s rpm\nneeds to be removed otherwise it will end up with skip broken\npackage.\n\nOn removing redhat-release, /etc/os-release file also gets deleted.\nSo we need to first remove redhat-release rpm if present and then\ninstall centos-stream-release package.\n\nIt also removes dnf-core-plugins package installation as we donot\nneed to download the centos-stream-release rpm.\n\nCloses-Bug: #1958202\nRelated-Bug: #1958240\n\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ia06c6770ea567656b5ce380826bdb120ca32a1a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b09fb3fef0896e812ca9ce3e6dc74f297494203a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b09fb3fef0896e812ca9ce3e6dc74f297494203a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28223,"name":"Cedric Jeanneret","display_name":"cjeanner (Tengu)","email":"cjeanner@redhat.com","username":"cjeanner"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
