)]}'
{"id":"openstack%2Ftripleo-common~839515","triplet_id":"openstack%2Ftripleo-common~master~I19d8f1c0b9f6d4cfe8cbf98e86cea51c76ff369b","project":"openstack/tripleo-common","branch":"master","topic":"ooo_c7_teardown","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-05-02 10:42:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I19d8f1c0b9f6d4cfe8cbf98e86cea51c76ff369b","subject":"Remove tripleo-build-containers-jobs template from zuul layout","status":"MERGED","created":"2022-04-27 11:52:49.000000000","updated":"2022-05-02 10:44:52.000000000","submitted":"2022-05-02 10:42:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839515-ooo_c7_teardown","meta_rev_id":"c12ac8581b229249a7e7ded6dcb0f65381e381f4","_number":839515,"virtual_id_number":839515,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-02 10:42:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-04-27 15:11:02.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"recommended":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"all":[{"value":1,"date":"2022-04-28 13:29:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},{"value":2,"date":"2022-04-29 13:47:20.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2022-04-28 19:30:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-28 03:10:20.000000000","_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":0,"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":1,"date":"2022-05-02 08:17:12.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 14:43:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 15:11:02.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-28 03:10:20.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2022-04-28 13:29:10.000000000","updated_by":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"reviewer":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"state":"REVIEWER"},{"updated":"2022-04-28 19:30:23.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2022-04-29 13:47:20.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2022-05-02 08:17:12.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"}],"messages":[{"id":"efd485b8dd710e7ac96c92f80f7c9cab6df2c006","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-27 11:52:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d058d52f1abea0dcd689661bc54f576afe9a9cfb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-27 12:14:49.000000000","message":"Topic set to ooo_c7_teardown","accounts_in_message":[],"_revision_number":1},{"id":"ca434bbead9f2d796cf0ed8294c4c25ed11b458d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 14:43:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f14500cb1cbb4b12bfdfee8fcfab79bf : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21704b41444346ed87e42b22987b3766 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d50c231d60524168855424b54cdd8cd2 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/090ce166b4b840f99bfcc73fd621090b : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c42569eb6b448268d79c71b012ab368 : SUCCESS in 4m 56s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/1c08af56ff694b6c937957e44d95d817 : SUCCESS in 2h 39m 10s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/acc1118c5c0d4d0c947255e2b33a6682 : SUCCESS in 1h 27m 15s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/97e082acff9d4cdbbd065f705b52e259 : SUCCESS in 56m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e534ed9fc6ca411d860d7403386dd76a : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7aa7b3a65f1042cf8278a4cfd85add5a : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1ba848e94ec4d598f2dcbb1e8be55df : SUCCESS in 4m 24s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/d82b78695f8c4181946f88efa7d58a75 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"ab59380aea6e1c69159e4b5035c6315d2ee90a92","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-04-27 15:11:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/867a20498bdb4e76b6ea1f7eba2e92c8 : SUCCESS in 3h 16m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6c7d5f64a75d9409dc342aee9ef0b29447871a15","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-04-28 03:10:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fff8b835f670d9a665ef7d045597f930f4aee5b","author":{"_account_id":34208,"name":"Rafael Castillo","email":"rcastill@redhat.com","username":"rcastill"},"date":"2022-04-28 13:29:10.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3430b6e633f9b8c326ce85dc979e3c46727d945d","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-04-28 19:30:23.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ad51dcd4bef90e9c0f49e8a71f749b5e98e72c52","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-04-29 13:47:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb3bb3ecfdb18268f080b8ca74cdadbcd073f3b5","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-05-02 08:17:12.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"af9f0b99725ddfebacb358116e4134a407b39bc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 08:18:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8a21e69bf21e0343b5e144882eac0538ddb60c13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 10:42:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4472fbb44c94736b603a60ba18e338c : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c352c75c456d4e06958e441696ee1a63 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38755bc253c84344bb3adc8d3f941474 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5318b3cbc2e94a338285c23303d14004 : SUCCESS in 4m 22s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/5bcbd43785ed48b0bf68d0ff8a5ac56b : SUCCESS in 2h 22m 53s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/e1d026edd56b4bd18473435020a0b278 : SUCCESS in 1h 16m 31s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/cf27f27e346641fa8a7e826baacc0570 : SUCCESS in 52m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b36d84bb5a0943cf90dc72a9dbe2119e : SUCCESS in 4m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9e179d0c4daf481ba676e5ae8db7a254 : SUCCESS in 3m 29s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/edf9a0bbf1804ecc8dd4ff43ff3f3ed9 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"59f59dfcf875136d46b9fa486e9cc26ae01916a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 10:42:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c12ac8581b229249a7e7ded6dcb0f65381e381f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 10:44:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a248dceb2094bb29d761593b2509d19 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"852a2d8ed01b2d6da56d56c59fe53e724aa46f78","revisions":{"852a2d8ed01b2d6da56d56c59fe53e724aa46f78":{"kind":"REWORK","_number":1,"created":"2022-04-27 11:52:49.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/15/839515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/15/839515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/15/839515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/15/839515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/15/839515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/15/839515/1"}}},"commit":{"parents":[{"commit":"8898142b3322f01dc60b6b8855c841c65e7712d4","subject":"Merge \"Switch from grub2 to bootloader element for overcloud-full\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/8898142b3322f01dc60b6b8855c841c65e7712d4"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-27 11:13:43.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-27 11:48:53.000000000","tz":180},"subject":"Remove tripleo-build-containers-jobs template from zuul layout","message":"Remove tripleo-build-containers-jobs template from zuul layout\n\nThis removes tripleo-build-containers-jobs as this only contains\ncentos-7 jobs [1] and they are being removed with [2].\n\n[1] https://opendev.org/openstack/tripleo-ci/src/commit/15761b77d91ab3e398f6fa1d10d2f5da267b7931/zuul.d/build-containers.yaml#L119-L127\n[2] I70d2e5dbe995bc8dfe06817d6460f3f2f3732e7e\n\nChange-Id: I19d8f1c0b9f6d4cfe8cbf98e86cea51c76ff369b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/852a2d8ed01b2d6da56d56c59fe53e724aa46f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/852a2d8ed01b2d6da56d56c59fe53e724aa46f78"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
