)]}'
{"id":"openstack%2Ftripleo-common~839532","triplet_id":"openstack%2Ftripleo-common~stable%2Fvictoria~I19d8f1c0b9f6d4cfe8cbf98e86cea51c76ff369b","project":"openstack/tripleo-common","branch":"stable/victoria","topic":"ooo_c7_teardown","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-05-02 12:45:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I19d8f1c0b9f6d4cfe8cbf98e86cea51c76ff369b","subject":"Remove tripleo-build-containers-jobs template from zuul layout","status":"MERGED","created":"2022-04-27 11:55:50.000000000","updated":"2022-05-02 12:49:23.000000000","submitted":"2022-05-02 12:45:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839532-ooo_c7_teardown","cherry_pick_of_change":839531,"cherry_pick_of_patch_set":1,"meta_rev_id":"b2ff4f5597b446b0db65445048ad5032f05341e1","_number":839532,"virtual_id_number":839532,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-02 12:45:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":2,"date":"2022-05-02 08:17:38.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-28 03:06:29.000000000","_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":1,"date":"2022-05-02 08:17:38.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 16:31:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-28 03:06:29.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2022-05-02 08:17:38.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"}],"messages":[{"id":"f134c0912e256929a6c840dc7082434de97dda1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-27 11:55:50.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"5c750a33263165f733f8ddee4c517c733dc50e20","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-27 12:14:42.000000000","message":"Topic set to ooo_c7_teardown","accounts_in_message":[],"_revision_number":1},{"id":"4d33ea4c6a2cb9254eabce578a3148fd902aea31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 16:31:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5bc13960a4c45e0ac37be49584829cc : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ab1c05fe7c94dc4ba6fab38b7ce9dfe : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/633b26bcbe5242a58b937912977293fb : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44ff2578f35a4e6597e2c3cbf65ff296 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82128ee5e2dd4ec28f4936373149f3f6 : SUCCESS in 4m 11s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/49917df9352b491fb290547aa05ade19 : SUCCESS in 4h 20m 47s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c64fe303b432442685b5920146a22db9 : SUCCESS in 1h 46m 07s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/ed41468d71f04900aeec385d27438bbc : SUCCESS in 4h 29m 53s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/7f07b9fdfe1a420ca9a7a0c1ed7dd107 : SUCCESS in 2h 44m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/02b3f309b120472c855324b72c592797 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/974dd12323df45d2b43ae31f15055292 : SUCCESS in 5m 15s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/06d2115d7cd04096a9e4efe547153c43 : SUCCESS in 5m 34s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/04c57d6955974bd3bc09fc5cd45a9658 : SUCCESS in 48m 34s","accounts_in_message":[],"_revision_number":1},{"id":"88fd445838e5ff771dc7d2ae93cc82764a8bca53","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-04-28 03:06:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32154b6aef8808f387e060039e474024b19d6771","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-05-02 08:17:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9840823520f9c4d79fd58cab64bff564f15eed60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 08:18:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cf08a026f50d40abb1582a0f88f0825a0a752ce2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 12:45:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a055cec2d83b402d836781780177f332 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4f66f93ee36442cbc27d8cd69bdf6d5 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad7c76125ad54d3db5f0309b6545fba5 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b7601d498d4b548e3083f275e69554 : SUCCESS in 4m 37s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c263adb7ec5748c796cdb758a1eeaced : SUCCESS in 4h 26m 53s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4fb645f96f0f44b3a61b14b8b5d86bf6 : SUCCESS in 1h 51m 16s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/fe530d0e64454ff09c1ce9281740ce61 : SUCCESS in 4h 21m 04s\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/15716c11dcfc4c35a1ff0d8c857a3ec6 : SUCCESS in 2h 45m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/383e9ba7a48e40c792d13bb799f3337d : SUCCESS in 4m 55s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/6cee353b84e14c34bd6f0f1f26dccddc : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"0bafc392713abea5a9e735c8b6e13f26a52ef9b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 12:45:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b2ff4f5597b446b0db65445048ad5032f05341e1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 12:49:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d6bc9b3ea004b3b90b65de4b5f7b28b : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fdd948671523d0207bd5c7f3fff6b749b08b15c8","revisions":{"fdd948671523d0207bd5c7f3fff6b749b08b15c8":{"kind":"REWORK","_number":1,"created":"2022-04-27 11:55:50.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/32/839532/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/32/839532/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/839532/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/839532/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/32/839532/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/32/839532/1"}}},"commit":{"parents":[{"commit":"a7c9952bc839bb21433bbbe0e3c82a04f4bb4b94","subject":"Merge \"Fix reference to undefined variables when heat-config times out\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a7c9952bc839bb21433bbbe0e3c82a04f4bb4b94"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-27 11:13:43.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-27 11:55:50.000000000","tz":0},"subject":"Remove tripleo-build-containers-jobs template from zuul layout","message":"Remove tripleo-build-containers-jobs template from zuul layout\n\nThis removes tripleo-build-containers-jobs as this only contains\ncentos-7 jobs [1] and they are being removed with [2].\n\n[1] https://opendev.org/openstack/tripleo-ci/src/commit/15761b77d91ab3e398f6fa1d10d2f5da267b7931/zuul.d/build-containers.yaml#L119-L127\n[2] I70d2e5dbe995bc8dfe06817d6460f3f2f3732e7e\n\nChange-Id: I19d8f1c0b9f6d4cfe8cbf98e86cea51c76ff369b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fdd948671523d0207bd5c7f3fff6b749b08b15c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fdd948671523d0207bd5c7f3fff6b749b08b15c8"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
