)]}'
{"id":"openstack%2Ftripleo-common~839533","triplet_id":"openstack%2Ftripleo-common~stable%2Ftrain~I19d8f1c0b9f6d4cfe8cbf98e86cea51c76ff369b","project":"openstack/tripleo-common","branch":"stable/train","topic":"ooo_c7_teardown","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-05-02 11:09:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I19d8f1c0b9f6d4cfe8cbf98e86cea51c76ff369b","subject":"Remove tripleo-build-containers-jobs template from zuul layout","status":"MERGED","created":"2022-04-27 11:56:10.000000000","updated":"2022-05-02 11:13:34.000000000","submitted":"2022-05-02 11:09:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"839533-ooo_c7_teardown","cherry_pick_of_change":839532,"cherry_pick_of_patch_set":1,"meta_rev_id":"c2ae3554e3e52c843221f9b73a720ae3df6e5cb6","_number":839533,"virtual_id_number":839533,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-02 11:09:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-04-27 14:46:02.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":2,"date":"2022-05-02 08:17:21.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-28 03:10:42.000000000","_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"all":[{"value":1,"date":"2022-05-02 08:17:21.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-27 14:46:02.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-27 14:46:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-28 03:10:42.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"REVIEWER"},{"updated":"2022-05-02 08:17:21.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"}],"messages":[{"id":"5540eb3202e906427e42d86f03c79bea6c73e2b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-27 11:56:10.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"f563793bb97c055b51355d5e6334ce12bf0c74d2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-04-27 12:14:39.000000000","message":"Topic set to ooo_c7_teardown","accounts_in_message":[],"_revision_number":1},{"id":"85a283ad9afb1be44c5fa2e51f97d05c2afc53f3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-04-27 14:46:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/fbd96d6f8161448d80ba89eb39ce6b43 : FAILURE in 2h 49m 22s","accounts_in_message":[],"_revision_number":1},{"id":"ab61c8946cc655a6a6947752836162de1baa0e5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-27 14:46:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/207f75b16649437982d76eb14f201b4b : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29e316ed3a7f488e93e1279d2f5619db : SUCCESS in 4m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/df79d796e1d549629f957559201e18f4 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/634286398d934317a8cc1e13fee46076 : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c70941a1f8ba4e8782734cd8c14de7a5 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c668c10bbd415fae7ddcc5977b095a : SUCCESS in 4m 15s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/e3b3a53ead6c4984b23a02a6ecddd6e7 : SUCCESS in 2h 36m 40s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/5f3c51b682e645c3b678f9c811ed4c24 : SUCCESS in 2h 02m 01s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/0639f658560e420696e6d7e3da0e9f24 : SUCCESS in 1h 24m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c59763f773be48fab25e1fcbd9e37335 : SUCCESS in 5m 43s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/1b3edd52d296437b997108f0808af490 : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":1},{"id":"d2a623cfb3f872b32a14175eff2a4a6f2e6d6618","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-04-28 03:10:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ee529a4ce599ab4d694b3bd648c02c56846c936","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-05-02 08:17:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"baed10272771ab51097aefbdf0fdcd08fbef8a5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 08:18:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a39dc1be29c1a103661c4b293bbb773c3d363fcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 11:09:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aec85d359ed54968bda273f74a3b04cb : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e2166ba6894c4203b6c2a83f31b35c3f : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb37c0fbc9464b01beb4fd62f7b28d03 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dca0b0ada6ad4db785de4f612545d876 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a25cd45a01425096f66700817deffb : SUCCESS in 4m 14s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3e979e00536248d890dc7924071f92b6 : SUCCESS in 2h 50m 27s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/71c55cb824244bd09b5db100bfaaca8c : SUCCESS in 1h 46m 39s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c346551a449b4703bdbe9dbfaa679ed3 : SUCCESS in 1h 39m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/abfa71b7b8b842de91a9e03a50c73bc7 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f4c0d45d0b3908b2aebd757d5559ddb11676c20e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 11:09:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c2ae3554e3e52c843221f9b73a720ae3df6e5cb6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 11:13:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee6eb49d63934a439e11dedfeb395e08 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7fcd60cd9c5185bed03324d23817145b93d1eb54","revisions":{"7fcd60cd9c5185bed03324d23817145b93d1eb54":{"kind":"REWORK","_number":1,"created":"2022-04-27 11:56:10.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/33/839533/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/33/839533/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/839533/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/839533/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/33/839533/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/33/839533/1"}}},"commit":{"parents":[{"commit":"b5ef9a568e7247aeb74ea20832be5d88a24bd124","subject":"Ensure failures on the undercloud leads to a complete stop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b5ef9a568e7247aeb74ea20832be5d88a24bd124"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-27 11:13:43.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-04-27 11:56:10.000000000","tz":0},"subject":"Remove tripleo-build-containers-jobs template from zuul layout","message":"Remove tripleo-build-containers-jobs template from zuul layout\n\nThis removes tripleo-build-containers-jobs as this only contains\ncentos-7 jobs [1] and they are being removed with [2].\n\n[1] https://opendev.org/openstack/tripleo-ci/src/commit/15761b77d91ab3e398f6fa1d10d2f5da267b7931/zuul.d/build-containers.yaml#L119-L127\n[2] I70d2e5dbe995bc8dfe06817d6460f3f2f3732e7e\n\nChange-Id: I19d8f1c0b9f6d4cfe8cbf98e86cea51c76ff369b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7fcd60cd9c5185bed03324d23817145b93d1eb54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/7fcd60cd9c5185bed03324d23817145b93d1eb54"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
