)]}'
{"id":"openstack%2Ftripleo-common~841549","triplet_id":"openstack%2Ftripleo-common~stable%2Fvictoria~I5673ff0480341cce4e0680587648c10c82ab1285","project":"openstack/tripleo-common","branch":"stable/victoria","topic":"ceph_monitoring_containers","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-05-12 13:04:15.000000000","reason":"Marios Andreou replied on the change"},"18002":{"account":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"last_update":"2022-05-12 13:34:56.000000000","reason":"John Fulton replied on the change"},"25402":{"account":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"last_update":"2022-05-13 00:42:27.000000000","reason":"Change was submitted"},"29775":{"account":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"last_update":"2022-05-13 00:42:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5673ff0480341cce4e0680587648c10c82ab1285","subject":"Move the ceph monitoring stack containers to quay.io","status":"MERGED","created":"2022-05-12 07:07:00.000000000","updated":"2022-05-13 00:44:22.000000000","submitted":"2022-05-13 00:42:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"841549-ceph_monitoring_containers","meta_rev_id":"137e1b35a638ca06d54c6010c752aaa543c58e4d","_number":841549,"virtual_id_number":841549,"owner":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-13 00:42:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"all":[{"value":2,"date":"2022-05-12 13:04:15.000000000","_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2022-05-12 13:34:56.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":1,"date":"2022-05-12 15:08:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":1,"date":"2022-05-12 18:06:07.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-12 12:53:25.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"state":"REVIEWER"},{"updated":"2022-05-12 13:01:23.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2022-05-12 13:01:55.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"CC"},{"updated":"2022-05-12 15:08:06.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"},{"updated":"2022-05-12 17:21:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70169637e1f6bd5418c10dc60ac4b650f40e6e5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-05-12 07:07:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f8b188a14060b28ebfb4d27181178a048ec1c19","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-05-12 07:07:28.000000000","message":"Topic set to ceph_monitoring_containers","accounts_in_message":[],"_revision_number":1},{"id":"bdb5bd05176e4c94d654fbeacad4a0f5e19e93d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-05-12 09:22:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e5f0ded059056fb352f656e0ed4beb2feda9eca9","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-05-12 12:53:25.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"551b95e07dbf000048f4ae9a580d3ea3974a4386","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-05-12 13:01:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5d6bea31817849a8f4d8607b601a7b02fa301b9b","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-05-12 13:01:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a81b39a50ff4d96fb9f51bb91cbb27137dd695a8","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-05-12 13:04:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"151a5b4f76b9b831c0f954e9c9267a31a7efeb63","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-05-12 13:05:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f12cf655951d535b1a095ecb2729193fc2ca5246","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-05-12 13:34:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"fa4c896d8081d6176a34d934221c662829af5e9c","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-05-12 15:08:06.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"24e0972bfc6b4572b90ebc9821015883c5aa4637","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 17:21:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/590d898378ec49699bf8f8b016a923d2 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a59d1b4722be41be98c0dbb037442abd : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa823979772841918a31cc85b8fd03a9 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c39ac0210ec644aab40ae9b7dafe65a6 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b680d9d26e2c45e0a6cd9a0abd487bc1 : SUCCESS in 4m 32s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6b0eeab39de043488089b15df636743c : SUCCESS in 4h 04m 11s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/cbb08b54db2648d68fab6b8e40a91f47 : SUCCESS in 2h 23m 41s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4a6b9cde0aa646ac964daccc229aa6e3 : SUCCESS in 1h 34m 35s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/13a14d5019bc490b8c01b8f50fed6c8e : SUCCESS in 1h 55m 34s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/59e81bfbd0704ac592cd6c5e58c02982 : FAILURE in 1h 08m 02s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/7b460d9e31a04a5d802de753c92d8b9a : SKIPPED (non-voting)\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/dcd476f4f364441b811eda3d2bdb51cd : SKIPPED (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/68aa3da19a8d461d8a96059c6437060c : SUCCESS in 2h 24m 42s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/4c236f8049c14ef8ae76cb24a131a619 : SUCCESS in 2h 17m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2e9f072b25b94ce3a0c2c4c1f473e2f7 : SUCCESS in 4m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/59f4b2b4287a4f7a9e36c33336436d89 : SUCCESS in 5m 01s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/3fb441bc2f3c4ac1b523263da12c60f7 : SUCCESS in 5m 40s\n- tripleo-buildimage-overcloud-hardened-full-centos-8 https://zuul.opendev.org/t/openstack/build/4a4fdc52ff944599bf8f806c259ac85e : SUCCESS in 54m 33s","accounts_in_message":[],"_revision_number":3},{"id":"5cf585b5fc60d0582b204dc9e0cbe71bee964b80","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2022-05-12 18:06:07.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"676c22ad4e18f1ad9adbab9f69d997439627242e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-12 18:07:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"89b10ea46e3f6eb31cf3f5caa11ccf4f3c69379b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 00:42:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73783fe9437144dc94337836a4416a58 : SUCCESS in 3m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e21b6d121cd445a88e3637a4e19843a7 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae4fac4df0fe4c4cb27517fa0e654c7e : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/221978caa7cf458a9dcb4edbfc0def6b : SUCCESS in 3m 51s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5b2df17fc94143dc965464412a3ed8bd : SUCCESS in 4h 02m 52s\n- tripleo-ci-centos-8-containers-multinode https://zuul.opendev.org/t/openstack/build/b300478c91b7468a9796d7f34b10b9d5 : SUCCESS in 2h 03m 18s\n- tripleo-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/179fb0186abc4ef3bff2431404f0d702 : SUCCESS in 1h 20m 05s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/4d0b245352f745eb9ff3efd9c4e61a21 : SUCCESS in 1h 27m 42s\n- tripleo-ci-centos-8-content-provider-ussuri https://zuul.opendev.org/t/openstack/build/3aafc9f0ce8a4546b820798f020d7943 : SUCCESS in 3h 58m 39s (non-voting)\n- tripleo-ci-centos-8-standalone-upgrade-victoria https://zuul.opendev.org/t/openstack/build/a83ef7c85c5e43939221042062a0d8e9 : SUCCESS in 2h 24m 00s (non-voting)\n- tripleo-ci-centos-8-undercloud-upgrade-victoria https://zuul.opendev.org/t/openstack/build/bed749e181614231a6edd49116497e10 : SUCCESS in 1h 37m 34s (non-voting)\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-updates-victoria https://zuul.opendev.org/t/openstack/build/765869fbde004537862eb63828f751cd : SUCCESS in 2h 13m 08s\n- tripleo-ci-centos-8-scenario000-multinode-oooq-container-upgrades-victoria https://zuul.opendev.org/t/openstack/build/1ca36de2f90d4822a1cbb3fee1562cf2 : SUCCESS in 2h 03m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5390dea9614046ad912c1320d6df4c4d : SUCCESS in 4m 22s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/e2b4075e9f564c4a8b0437dfe1f8ff6a : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":3},{"id":"d4f39f5802d48bf4950fae78cd5f290a1430bd06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 00:42:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"137e1b35a638ca06d54c6010c752aaa543c58e4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-13 00:44:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f8e9f5bd6194eea84633eff79a92ea9 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f20606909ecc6d363b9e2e7ad721221ebaad4169","revisions":{"a75b38b20da3cc501d49f7e953aa5be5e57dbcf9":{"kind":"REWORK","_number":1,"created":"2022-05-12 07:07:00.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/49/841549/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/49/841549/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/841549/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/841549/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/841549/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/49/841549/1"}}},"commit":{"parents":[{"commit":"fdd948671523d0207bd5c7f3fff6b749b08b15c8","subject":"Remove tripleo-build-containers-jobs template from zuul layout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fdd948671523d0207bd5c7f3fff6b749b08b15c8"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-05-12 07:06:41.000000000","tz":120},"committer":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-05-12 07:06:58.000000000","tz":120},"subject":"Move the ceph monitoring stack containers to quay.io","message":"Move the ceph monitoring stack containers to quay.io\n\nquay.ceph.io is no longer accessible and the Ceph Org moved to\nquay.io. This patch just updates the container image references\nto point to the new registry.\n\nCloses-Bug: #1973115\nChange-Id: I5673ff0480341cce4e0680587648c10c82ab1285\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a75b38b20da3cc501d49f7e953aa5be5e57dbcf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a75b38b20da3cc501d49f7e953aa5be5e57dbcf9"}]},"branch":"refs/heads/stable/victoria"},"a203ffe38f0abbe8f329747c8cb4482629381979":{"kind":"REWORK","_number":2,"created":"2022-05-12 09:22:58.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/49/841549/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/49/841549/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/841549/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/841549/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/841549/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/49/841549/2"}}},"commit":{"parents":[{"commit":"fdd948671523d0207bd5c7f3fff6b749b08b15c8","subject":"Remove tripleo-build-containers-jobs template from zuul layout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fdd948671523d0207bd5c7f3fff6b749b08b15c8"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-05-12 07:06:41.000000000","tz":120},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-05-12 09:22:54.000000000","tz":330},"subject":"Move the ceph monitoring stack containers to quay.io","message":"Move the ceph monitoring stack containers to quay.io\n\nquay.ceph.io is no longer accessible and the Ceph Org moved to\nquay.io. This patch just updates the container image references\nto point to the new registry.\n\nWe are also marking Ussuri content-provider and victoria upgrade jobs\nnon-voting because Ussuri content provider is failing with similiar\nerror - Unable to pull containers from quay.ceph.io and its not\npossible to fix Ussuri as its EOL and not accepting patches anymore.\n\n[1] https://zuul.opendev.org/t/openstack/build/3997166729c9416b8ed634db844e0302\n\nCloses-Bug: #1973115\nChange-Id: I5673ff0480341cce4e0680587648c10c82ab1285\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a203ffe38f0abbe8f329747c8cb4482629381979"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a203ffe38f0abbe8f329747c8cb4482629381979"}]},"branch":"refs/heads/stable/victoria"},"f20606909ecc6d363b9e2e7ad721221ebaad4169":{"kind":"REWORK","_number":3,"created":"2022-05-12 13:01:02.000000000","uploader":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"ref":"refs/changes/49/841549/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/49/841549/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/841549/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/841549/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/841549/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/49/841549/3"}}},"commit":{"parents":[{"commit":"fdd948671523d0207bd5c7f3fff6b749b08b15c8","subject":"Remove tripleo-build-containers-jobs template from zuul layout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/fdd948671523d0207bd5c7f3fff6b749b08b15c8"}]}],"author":{"name":"Francesco Pantano","email":"fpantano@redhat.com","date":"2022-05-12 07:06:41.000000000","tz":120},"committer":{"name":"Sandeep Yadav","email":"sandyada@redhat.com","date":"2022-05-12 13:00:03.000000000","tz":330},"subject":"Move the ceph monitoring stack containers to quay.io","message":"Move the ceph monitoring stack containers to quay.io\n\nquay.ceph.io is no longer accessible and the Ceph Org moved to\nquay.io. This patch just updates the container image references\nto point to the new registry.\n\nWe are also marking Ussuri content-provider and victoria upgrade jobs\nnon-voting because Ussuri content provider is failing with similiar\nerror - Unable to pull containers from quay.ceph.io and its not\npossible to fix Ussuri as its EOL and not accepting patches anymore.\n\n[1] https://zuul.opendev.org/t/openstack/build/3997166729c9416b8ed634db844e0302\n\nCloses-Bug: #1973115\nChange-Id: I5673ff0480341cce4e0680587648c10c82ab1285\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f20606909ecc6d363b9e2e7ad721221ebaad4169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f20606909ecc6d363b9e2e7ad721221ebaad4169"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
