)]}'
{"id":"openstack%2Ftripleo-common~844147","triplet_id":"openstack%2Ftripleo-common~master~Ia502a5a1b3451dc9096a94b52a5f04422386c942","project":"openstack/tripleo-common","branch":"master","topic":"secure_boot","attention_set":{},"removed_from_attention_set":{"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2022-06-28 02:39:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia502a5a1b3451dc9096a94b52a5f04422386c942","subject":"Install grub2-efi, shim in ironic-pxe image","status":"MERGED","created":"2022-05-31 23:35:29.000000000","updated":"2022-06-28 02:42:15.000000000","submitted":"2022-06-28 02:39:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844147-secure_boot","meta_rev_id":"fa25d19c157caba7a510182f658bcdc7ada2a682","_number":844147,"virtual_id_number":844147,"owner":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-28 02:39:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-06-27 16:24:35.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":2,"date":"2022-06-14 21:59:46.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2022-06-21 07:50:40.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2022-06-21 07:50:40.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-01 01:17:42.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-01 02:20:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-02 04:02:03.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-06-14 21:59:46.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2022-06-26 22:44:13.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"CC"}],"messages":[{"id":"9a5f9142bd8559791ecf2555e6f8a54fa258ac47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-05-31 23:35:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a21bfcc1ca507fd280d58ef4713e33b7f560d0fe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-01 01:17:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2906e06d41fb4167bec396f0526d7bdb : FAILURE in 1h 40m 44s","accounts_in_message":[],"_revision_number":1},{"id":"5cdcc978241a39f81c4f1bd1b571e4030204f40f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 02:20:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3373814533db43caaaa166c2390029a3 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99f9b19ea280410a9bf2c7c6011ad658 : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/657e88dd8fc54a7a85e3d2e8e7ab7c8a : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/005a68b91f9946bd8c73c4ab9800be15 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f854cbc1e78d4885b02887b53d118c65 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/717dc4bb609a4d2ea79c7ad8feb1197f : SUCCESS in 4m 06s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/a7cbf36a02af4d7cb20c9dbff53708b9 : SUCCESS in 2h 32m 54s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f326343fddfb4f0d889078cf50ca2c99 : SUCCESS in 1h 39m 04s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6a96c70b0b7147a4b74e0f42682473f8 : SUCCESS in 1h 33m 52s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/eb3e172c8e974df7915f31679acad1da : SUCCESS in 35m 14s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/0372663e4baf4f009c803f190733165f : SUCCESS in 1h 25m 46s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/a18193880fcb4b0480c8065a4c46d023 : SUCCESS in 2h 07m 58s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b03620975bcb4c928cdaa037ccc69c4b : SUCCESS in 1h 19m 08s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/ae0fc2a39d384e549a7a95f6b159b32f : SUCCESS in 1h 08m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/90135be355774da48dcb5d908a87420d : SUCCESS in 4m 08s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/5a66e285bec940f383ed825f97b75c99 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"299f423935939efd08d2d4f6d40ab8a8a8763333","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-06-02 01:40:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f65c87be82e5df031a09b00bf0c1c8472b6511b5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-02 03:56:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/60b30d687df945a5a19b6e9483fc5b5b : SUCCESS in 2h 14m 31s","accounts_in_message":[],"_revision_number":2},{"id":"13db75cd3359aa8d7a6c4a1245c556580436a566","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-02 04:02:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b61b0be6eca41a2fc35e24195c417289ca4c727b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 05:48:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8cec399d7d75463fab49dd0814421c95 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6293967b650f417084f3fc1dd5760fdb : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba4ac0c0774e465cadc51b64807f525e : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3edfae0b10043ed8c6b5f2b97e1d179 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bfc0c72300fd4988a6261cd8060e2b10 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4d0076bbf714e51b7d907acc63504b4 : SUCCESS in 4m 24s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7ab0fe1508d7415096a29fa065743ff6 : SUCCESS in 3h 55m 04s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/3baaecff27594008b53c16fb97da887e : FAILURE in 1h 24m 54s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/8434cbe00ab0452bbf69b3be7b42e4af : SUCCESS in 1h 39m 57s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/52b23d20184a453eae1a0ef9cd177344 : SUCCESS in 1h 06m 39s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/3d5b2c3248ad4720bb0a6328122c881e : SUCCESS in 1h 28m 59s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ae992751e6674445a8e15ee6eae6bf3f : SUCCESS in 3h 35m 46s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/69a6607c20d04efeb41b3d2ac8dffba7 : SUCCESS in 1h 30m 05s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/e9128b441339408e8211f1acf7e8594f : SUCCESS in 1h 18m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00500ead27ed4cec908a12f6b687ec6b : SUCCESS in 4m 33s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/a63866e48122441ea04153ed5b7404ae : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f07ea1e6e7462e9c8d2704e1767d0c33a2f2a9fa","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-06 13:05:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f92fbe5efb8f56c84830eb7401f0936953262cdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-06 17:07:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e155dea0b7542b0aa543fc0229bcab7 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b7a2bf315f41e8b78b3de473a6a860 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74c68b72db64447bb71ac6411d929de2 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37cfcac1c6404b33a44b6da242a0432c : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61a29211e30c49c1a40a0b5a0f4bf8d6 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b303a25146d54d2d8072be9d47576571 : SUCCESS in 4m 26s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/616e0ea1720f4c298cb7b73e5b37d7ca : SUCCESS in 3h 47m 16s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/23f6dcf5b8334b1bb67682e563683685 : SUCCESS in 2h 31m 08s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/271408d433884cbea2e858a6885e8c66 : SUCCESS in 2h 25m 18s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/c190757a284d42c1ae1fe3a20aee9522 : SUCCESS in 1h 07m 32s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/38d6fbb251dd416da2a8d8567b134548 : SUCCESS in 1h 45m 38s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/cf3632b4044a4648a0ede437c5ba300f : SUCCESS in 2h 55m 11s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/be7cd91cd8834e1cacb6b6c422f4c3f6 : SUCCESS in 1h 25m 22s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/36016e5fae0348a0bc11993680a4bfc1 : SUCCESS in 1h 18m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f9ef6c502c0542f5913e5f6af0d933a3 : SUCCESS in 4m 25s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/dca1a5131bcd4bf5912e65cc2c78d2b7 : SUCCESS in 5m 46s","accounts_in_message":[],"_revision_number":2},{"id":"63b8ebd84230bfd54e11a12c145d54519bbce3c4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-06 17:58:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/683245cdae9146b08bad62393e9448f2 : FAILURE in 2h 16m 42s","accounts_in_message":[],"_revision_number":2},{"id":"884f024ca5bb4e283d33d1641189ba7c47c79ca7","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-06-14 21:59:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"983b564a4661fbb12afd765c63f618815642728b","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2022-06-16 03:32:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ac15b579f03461a05d1b865e7f873feff05e5124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 04:19:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6c91b2e88db490d8651ce5fd6a7133f : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e331e2786c418fbff79ef53f067cf6 : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abda74a41f5a4f428c51c9426b686cfe : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1bc0eaee1924ef4bd1e98fe82417721 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95071fd51c6a4b478d3a0432f8c57962 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d103c870d6174fa99d73532a9d83b3e3 : SUCCESS in 4m 27s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/46d11da27080486192c26fc67e53ea09 : FAILURE in 34m 48s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f078281fb7e44ecf8614320ea6521f84 : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/4a2194ae7a024d38bc00d8254f0c916e : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/cda88937c0f847ea965bd45842e126b8 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/7df09bbe1dff4f34b72e42186c75572d : SKIPPED\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/5e82947ca7b74882acb2222c88cf6deb : FAILURE in 43m 04s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/12756037aca84940b2e9789864236be6 : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/1ac800be53f341b69ff943fa2924cd70 : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fdc8f8a21ac04e7ba89f1a091d9d2054 : SUCCESS in 4m 08s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/37cbfe660a9748159b3af484cc9161af : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":2},{"id":"a489fbe86248e1201c08b7a731d1917b7d3bece8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-16 04:29:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2719d49b18894764935642b10b1f73d7 : FAILURE in 51m 02s","accounts_in_message":[],"_revision_number":2},{"id":"fb0dfad95be11f5021131e0db59d23a75ffaa0a9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-21 02:24:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"12dd6f0e557ecd2cadbb6e83d8f5dc9af180e75b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-21 04:35:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f49dd766680c461e99d3aee051cbc585 : SUCCESS in 2h 10m 39s","accounts_in_message":[],"_revision_number":2},{"id":"685599337ef6042bf9fbe4d1ff6a2277d6d738b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 05:25:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d05a32fec394953951c5f98ef5b0073 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb7d9353094f42ffb21805a28855f750 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22abd78db6c34c928259d35e5001f192 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2400055e90ae4ff5a04dc8b73db3bb4f : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c2fb19b1b6441d995b2839bfb9af52f : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2729d23e73114ebb929b03c9b4db3846 : SUCCESS in 4m 32s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/e916169a61ab420b8bc43350aedf8495 : SUCCESS in 2h 15m 44s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/885523549b084054b611856e6accdd3c : SUCCESS in 1h 24m 05s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/7a01a8af63b643afac48c3cc549c98c9 : SUCCESS in 1h 15m 16s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/ca88976cfdb245bc8c8496cc21d36c69 : SUCCESS in 40m 05s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/0841cae2d5d94707a5a8cdc73cf1de47 : SUCCESS in 1h 04m 30s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/87f8372eba1f41b99d53ead32e94c997 : SUCCESS in 2h 03m 41s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/82ec6296d1cb4ec39381fb80c3ff0f70 : SUCCESS in 1h 04m 45s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/6b0c2cbecc8f4371951e84887de81581 : SUCCESS in 1h 00m 25s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2dd5c1a4f36d4d45ada04916b4d34b8f : SUCCESS in 5m 03s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/1f06042902c24977bf227528991f8965 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"05bafd5dba17f319151e4393d9805fe44b55090b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-21 07:50:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"13119d8d27756627823421d3f03de431704041c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 07:51:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"57c6bfd3b39d546f27c14e02d992f6f0882fdd8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 11:28:01.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bb514f038ae4117afc06ab857fec369 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/049cb991b6114789bddba8d4d34123cf : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69c1c55664bd49228a077ee4707c9e79 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac2db44067446139612bb00ac41a754 : SUCCESS in 3m 29s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/29b08733879447879c9d29a62bb0a9a6 : SUCCESS in 3h 32m 17s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/51f887f249494a2b8277df2bad6ec7e3 : FAILURE in 2h 11m 25s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/ec52bf644b694b338eead3f5c1e479cc : SUCCESS in 50m 42s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/4d8d6340fbed4b719c42ec63b71fbad4 : FAILURE in 1h 02m 12s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/97be40d15e2d4239a19df8a4fe25d5fc : SUCCESS in 2h 33m 57s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/93ac91e4aec0440fbe7d295525ca96f2 : SUCCESS in 1h 13m 48s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/cb0140839d4545be860ded67f6526679 : SUCCESS in 1h 30m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1cccf0ac94bd4ae1ad4cdb24cbb6ee66 : SUCCESS in 4m 22s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/d5d89d45f04342a1a786991f6f75e973 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"fc4d3d6691805922dd32097c1d9b123d7a9677fd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-25 02:17:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"29e08acb8bdda2c57cf92f210e1b4a0e50d899f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 04:26:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27361f7341bb4049862160e546c9da05 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2ac05cf612f445aaeac91c624c5657d : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/223b5953cc8848edabb78d63bf69632a : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dde29dc51964a848f69bdf05061b4c8 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/796fc7d7e1944a68b3fb9fd6673eea5b : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/660182c7d27445e3a59cff92602e2a12 : SUCCESS in 4m 32s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7d51491951f04baa9f627e81b178811e : FAILURE in 1h 10m 04s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/3bcb83710c5b426d8dfef15426451b09 : SKIPPED\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3813ea7cb4994cdfbc24b7e13ea97876 : SKIPPED\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/779498f6390e41adb608c2b233388783 : SKIPPED\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/54b665e438da473c9e95b8a6c8edeaf3 : SKIPPED\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/e31bd1bf221c44f39fc5ae8ad08a2cd3 : SUCCESS in 1h 40m 44s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/86c96b41ae22490585d48b71d3ce77a8 : SKIPPED\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/99f615742e684e1da2aed515c3bf2106 : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6fc4a5c52bdb4042971693d1241da7cf : SUCCESS in 4m 08s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/efee627850b84f189a410df533552326 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"cfb7fc08252949e0a6b676dcb9e2b22e05b9b27e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-25 04:29:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/f9fa08e15030445c8cca3a4ab0acaf2d : SUCCESS in 2h 11m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0b88e74e82e29fe1dd470c91fa77502203d2a897","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-26 13:49:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ce53fb1cf2e12266c5a58e725ce46bf9f01863c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-26 16:46:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/20c2b39e756a45e0bd1224408c2a0c98 : SUCCESS in 2h 55m 41s","accounts_in_message":[],"_revision_number":2},{"id":"f230201436f7e6f585fae9ddbd12d1cf87f30c90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-26 17:03:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcb66dba2f9d47e6a1e538952d388d78 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc0f89ede2ed42e697c9c29e1ca48ac0 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46b04689699b4268a995fb9f39e270c1 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f71cd716ac2d40218d08efcd0c8c8a89 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cfd439932eb4b37a7a35facfe50a748 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db998d746bf741afa298f40a6c9b06d1 : SUCCESS in 3m 48s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/8ccb6df58c314d8c8d03cbb3f9075841 : SUCCESS in 3h 03m 34s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/ca0ace3b523542fdbf9e541af9b4d492 : SUCCESS in 1h 46m 03s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9abb9a1b1a1a4da69a8daa238fa58015 : SUCCESS in 1h 36m 24s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/1b6c6bc390a247738aebc62973c9adf8 : SUCCESS in 47m 53s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/9db11b0389764ed4a25c77c9d2c83910 : SUCCESS in 1h 39m 41s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/47e0ef8d6afd434ab52295ea5ae084a5 : SUCCESS in 2h 36m 54s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/363bbbc31e94490b97653fe98f24b2d0 : SUCCESS in 1h 31m 35s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/0b86b59571b4460d8f5c6f987d695d33 : SUCCESS in 1h 11m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4111a50d80ab47d2bd13bf883d142411 : SUCCESS in 3m 43s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/af966b0054c9472aa552ffb5b0bc4188 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"d1ab0ba3f143a5e45dccb84f5c02341031ec8a99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-26 17:04:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f67589508a747716218d447a76dd04143e33aace","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-26 22:35:52.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a3fba9781d0474d923b161db0704658 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87f3b2c8b3f34176896ee60295519892 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d045ee0dfae4b61b2dfa63d86ac5289 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93913117b4024d42a64c11fc7947650d : SUCCESS in 3m 40s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/fecef9736a814109942219afed24d7fd : SUCCESS in 3h 01m 17s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/14dd73df32444731a846d9d94225a55c : POST_FAILURE in 2h 09m 39s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/13524e324d6542d3b05ddaa2c961ec6c : SUCCESS in 34m 39s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/8e23ecb6667e4ad2aae5ba325d586f4b : SUCCESS in 1h 07m 10s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ecf7b12c8a16467b9704f25c25f410ad : SUCCESS in 2h 15m 37s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/76375aff3b25442d8ce7a9f4df552d74 : SUCCESS in 1h 28m 39s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2b7312ebda4f4fe8bce5d4e3d8fb6e0e : SUCCESS in 1h 11m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6c3418c0286d43da93556266a4c09155 : SUCCESS in 3m 45s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/b2a8860df30b48c8abb024aa5648a3d7 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"777b8e582333acc80e6a4bbeae76885ea0abba75","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-26 22:44:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95fef0d5c317f00f56842f1eb651dac3000d92ec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-27 00:55:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/72caddc9b3ca4b4a95d6df35a719ab40 : SUCCESS in 2h 10m 15s","accounts_in_message":[],"_revision_number":2},{"id":"467a602ef2df5ffe1ead5dfcfb15da3ac0651ad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 02:21:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1e0b50022da462c878b09369138a36b : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be7be6b6e5f2472885aaa3a745f7d4b5 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64fcadae5b074611a765c86567fd711a : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8faf1a956d5442128dee8d6250bf36ec : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f16b236c26a84db399f598aba3a9173b : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5b51bc89ee44be4a0e43e97a0a1fdba : SUCCESS in 3m 52s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/0f225bb49b6c465296b67f6977c840b4 : SUCCESS in 3h 27m 09s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/499af9bc586b482a98979d36161698a8 : SUCCESS in 1h 59m 27s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2e64a9b148d3471c95f4e3b18392795f : SUCCESS in 1h 37m 54s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/39a8a8029ba24da4ab59022dc6d60f7f : SUCCESS in 50m 11s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/db735062d4e84e63aab86028ba947ca9 : SUCCESS in 1h 30m 01s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/9e9d81e29bed497c9f84cc06eb15fd83 : SUCCESS in 2h 25m 32s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/84d8b245bfd64121b95b41902d7e2960 : SUCCESS in 1h 38m 26s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/d1f01497d53c423fa98a7a0394017f9a : SUCCESS in 1h 09m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a83c9ab2cb2d476d953e68b8d6c1c222 : SUCCESS in 3m 56s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/a6662cf137b744bc8a42fe70438be090 : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":2},{"id":"01f77e309c9fc319c260252dd1000452ba6e0042","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 02:22:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a9ff9aff9515c714691965b3a6eb25ee3a3d9d91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 12:41:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac9f01961acd410ca1812249b8039ccd : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb33ed15f38247b7b1b46e5a22c3927d : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bb295ba882d49a1a7c8261eedd2111c : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9058ef443d5548aa9501f3eee843b07d : SUCCESS in 3m 47s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/1eb64e7c4b5b4901a08cf831d7241466 : SUCCESS in 3h 12m 54s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/4f094e46ed694f43b7515a1072d31f60 : SUCCESS in 1h 30m 43s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/90fbcb65a23648ebb34254de5c634460 : SUCCESS in 38m 12s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/4e52a2cb1b25442ca63c0be04e48c803 : SUCCESS in 2h 01m 58s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/1ec5950439ff435dabc811b2d2c272f2 : SUCCESS in 1h 00m 25s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/a92771a7800e4702aef4243d2c51eb85 : FAILURE in 10m 03s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f76e65e481c74389b8d8882e811ce7ef : FAILURE in 1h 09m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/65832997dfbd4fb7addb9dde15adfc55 : SUCCESS in 4m 09s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/d4755b9f1c6843aba0134b2618094f77 : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"20576d495636727c2e85c3e6fe4cb26c20cb10b4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-27 15:16:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"06a054abc90eaf08046085074632c72cff65db83","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-27 16:24:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8f04731baa9d48a089f408ce67fd349b : FAILURE in 43m 59s","accounts_in_message":[],"_revision_number":2},{"id":"48be9c25e593f400ff32b54bd890f0dc989a7aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 17:49:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/beead6396a4345bfb0be3ac7202056eb : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d07d77b68514946b079449ce2731ca2 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ebd17ecedfb4f27ba29fa422007b2b2 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7e7b8d51e9a43369f7c383d6ef25262 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c88d38945ba4340abbc04f166c7e550 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c6a0924105142779ca78a1a651efc66 : SUCCESS in 3m 51s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/24d270f9e0e94681a75de484837c7fce : SUCCESS in 2h 24m 10s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/88c193cea70b49e3a7e1fb4407d84bf2 : SUCCESS in 1h 25m 45s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/dcac6ebb8bef4ee89bfb625f3d799c81 : SUCCESS in 1h 20m 00s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/ba1c25b560a54ad194692e2cd3bdd5fc : SUCCESS in 37m 45s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ad33994a65284e9f80b27572b52e7ddf : SUCCESS in 1h 13m 16s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/cea4b2328fe24096854c0e5c4fd68719 : SUCCESS in 2h 08m 38s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/4a94d5f8a6e347acaea2a0f4f4dc69f1 : SUCCESS in 1h 13m 04s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/4a68f749d9314790a0620410cdbbc587 : SUCCESS in 1h 02m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1e8c0a4dfd6a45d4a3aa902b8ae1f5e1 : SUCCESS in 4m 41s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/ab8109d5d23e4c618017646fa0db1974 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":2},{"id":"c1e87378f883c2bb10d1250ec6922ba787b1705e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 17:51:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"13790d01e251185a5a08c02b5f2d0c311afc0ea8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 02:39:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fd4728792ab41319a7afe87c333af1a : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5139f72b4342429daeb5ee6fa2f4c848 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10b2adbf58da43f4a7455274e8fa891b : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d3aed9f30244e69a514870090fd5dde : SUCCESS in 4m 18s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/be5d7d3022d64817adedc2f2d008f627 : SUCCESS in 3h 06m 40s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/68b7035fd4fc49afb26342619d9af975 : SUCCESS in 1h 50m 47s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/21ee3c36ab60468d8a3bfd18a907428b : SUCCESS in 57m 18s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ca141a81a0e547cd80bb94009f2f0877 : SUCCESS in 1h 28m 37s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/67b16e3d7ac645b98fc3ff5a6a09aae7 : SUCCESS in 2h 41m 12s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/989996bcd71b42c5898add66d9fdc36f : SUCCESS in 1h 20m 30s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/80331f7133ab4b75ba3c2017989681b4 : SUCCESS in 1h 40m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/08c5b4cb8cf644b08c5a98feb26d347c : SUCCESS in 5m 58s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/f206716201c8400e91b58b8e94b6220f : SUCCESS in 5m 08s","accounts_in_message":[],"_revision_number":2},{"id":"1984a3dd51a356fe2f1c10dd2261926452a29ee6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 02:39:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fa25d19c157caba7a510182f658bcdc7ada2a682","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 02:42:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb94cf6656b491ab4f399fd83383566 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"115a740e89b15964367ba79704742d0cc6136277","revisions":{"93e83ec6573764f71f1e09543eaa2171bfa969a9":{"kind":"REWORK","_number":1,"created":"2022-05-31 23:35:29.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/47/844147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/47/844147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/47/844147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/47/844147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/47/844147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/47/844147/1"}}},"commit":{"parents":[{"commit":"b36e292c68b5c6cbac37ba01361d31221f130f64","subject":"Update ceph daemon image pull registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b36e292c68b5c6cbac37ba01361d31221f130f64"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-05-31 23:33:39.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-05-31 23:33:39.000000000","tz":720},"subject":"Install grub2-efi, shim in ironic-pxe image","message":"Install grub2-efi, shim in ironic-pxe image\n\nThese packages provide the efi binaries which enable grub network\nboot.\n\nRelated: https://review.opendev.org/c/openstack/puppet-ironic/+/842034\nChange-Id: Ia502a5a1b3451dc9096a94b52a5f04422386c942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/93e83ec6573764f71f1e09543eaa2171bfa969a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/93e83ec6573764f71f1e09543eaa2171bfa969a9"}]},"branch":"refs/heads/master"},"115a740e89b15964367ba79704742d0cc6136277":{"kind":"REWORK","_number":2,"created":"2022-06-02 01:40:47.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/47/844147/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/47/844147/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/47/844147/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/47/844147/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/47/844147/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/47/844147/2"}}},"commit":{"parents":[{"commit":"b36e292c68b5c6cbac37ba01361d31221f130f64","subject":"Update ceph daemon image pull registry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b36e292c68b5c6cbac37ba01361d31221f130f64"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-05-31 23:33:39.000000000","tz":720},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-06-02 01:40:36.000000000","tz":720},"subject":"Install grub2-efi, shim in ironic-pxe image","message":"Install grub2-efi, shim in ironic-pxe image\n\nThese packages provide the efi binaries which enable grub network\nboot.\n\nRelated: https://review.opendev.org/c/openstack/puppet-ironic/+/842034\nChange-Id: Ia502a5a1b3451dc9096a94b52a5f04422386c942\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/115a740e89b15964367ba79704742d0cc6136277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/115a740e89b15964367ba79704742d0cc6136277"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
