)]}'
{"id":"openstack%2Ftripleo-common~845807","triplet_id":"openstack%2Ftripleo-common~master~I65b32f4ce3fddeb694fb33ca65076d45d23a3bb6","project":"openstack/tripleo-common","branch":"master","topic":"memory-usage","attention_set":{},"removed_from_attention_set":{"21129":{"account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"last_update":"2022-06-23 15:33:22.000000000","reason":"\u003cGERRIT_ACCOUNT_21129\u003e replied on the change","reason_account":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"}},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-07-01 13:10:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I65b32f4ce3fddeb694fb33ca65076d45d23a3bb6","subject":"Reduce memory consumption in Cinder services","status":"MERGED","created":"2022-06-14 18:46:23.000000000","updated":"2022-07-01 15:07:45.000000000","submitted":"2022-07-01 13:10:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845807-memory-usage","meta_rev_id":"1e94f7ebc10728dcf99ec7d706647db01ea6dbbd","_number":845807,"virtual_id_number":845807,"owner":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-01 13:10:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-23 20:18:50.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":2,"date":"2022-06-28 20:00:53.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2022-06-23 19:46:58.000000000","_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":1,"date":"2022-06-28 20:00:53.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"CC":[{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-14 20:55:36.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-14 21:38:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-15 18:55:18.000000000","updated_by":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"reviewer":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"state":"REVIEWER"},{"updated":"2022-06-17 11:34:34.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2022-06-28 20:00:53.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2022-07-01 05:07:40.000000000","updated_by":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"reviewer":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"state":"CC"},{"updated":"2022-07-01 14:56:18.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"state":"CC"},{"updated":"2022-07-01 14:56:18.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-07-01 14:56:18.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2022-07-01 15:01:04.000000000","updated_by":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"3b094d9d124bde999202f8029be128f41236295c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-14 18:46:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"698d844cb862a59528d74c5da5b70fa4cde0947e","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-14 20:55:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c2fbf2310dda46908b7e96897e05538f : SUCCESS in 2h 08m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8042bf29b823d901da387c881707c961a2aa3aab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 21:38:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea6c401b887c4ba0b1a80f0560ab25cd : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4858929bee741b1aa9750033f8d1e61 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9bbf0aafc684df1934ca7242222d833 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a1165525adc4de9979658beee8dd1c1 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fa5c1eff37b459b8849a6750eac1e4b : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef53aead3d384ff79e6ab13fb9f70661 : SUCCESS in 4m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3d47f279b3e4068815667a5c98bbbbb : SUCCESS in 4m 10s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ac9e7f80a243462babe9141640421aaa : SUCCESS in 2h 38m 55s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/3d77acd2a95949589b07e3602ad5fa16 : SUCCESS in 1h 35m 32s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2e097ff3a29548338ad2eb6cf8122749 : SUCCESS in 1h 32m 29s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/22fe795d033941eb93ecb2f70606e54f : SUCCESS in 43m 35s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/1ea1dad568b2441aaf450784c989a1b2 : SUCCESS in 1h 19m 12s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/159e266720dc483aa2f598ebe47e90c4 : SUCCESS in 2h 21m 16s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/229d6450d5274045b61cee33507d1b20 : SUCCESS in 1h 18m 25s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/4430fd7a70c64331a5640d5b662ba1b9 : SUCCESS in 1h 11m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fbec4c93b53240a2b2005983c1d413a0 : SUCCESS in 4m 58s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/c6bad527bd5149a29caacd7ba54d6083 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"8a28e5def3fac9b8ba7cc6b7a2eafa8b4252b88b","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-06-15 18:55:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"aaf932f85b05b2cee529762a9942ff10adf9a434","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-15 19:20:53.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf189d8672390a3ba53136acd5bf04672678bc01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-17 10:43:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"50cdc2acef4ad3712dc42febe83a44e675d92d3b","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-17 11:33:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2647d8b3be5b01536ecaecf381c893a31e0d49dc","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-17 11:34:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"01dad9375ad4444804b7378383ff97aae37f489d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2022-06-17 12:24:56.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91a986bd8f487d629f116106fcb0c94cb6073610","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-17 13:22:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/643646db2e0349dd80c04faae598f810 : SUCCESS in 2h 34m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f4037b19a329d7689dc69b8e7db14fe50dbfb4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 13:42:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b66a91fba424e8b855a1667630fa5aa : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0537c32546814f6182b93ec93620d09d : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73625143487b4887a57dc31e45eb68ef : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/358f1cb4df34479daada31705cf32026 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3346e7271a80491bb8ccf4f6e413b677 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf1769c46a847e3bbc1e76c1970709d : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7f62221cc4c4800adcd81da50ea7eb5 : SUCCESS in 3m 23s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/d128f0b56606450d905a54d31810258e : SUCCESS in 2h 45m 13s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/8e82c4f3578c417cb1418a074d778071 : SUCCESS in 1h 40m 23s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/65a032299e434d1fa00cc4455794d985 : SUCCESS in 1h 38m 09s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/0174160a88e543c4b7ba4e0e1cc1583b : SUCCESS in 44m 20s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/1b287cd8650f49379a546eccdbdd864b : SUCCESS in 1h 21m 36s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/21a3ba5b115d4996bbc3f5479485288a : FAILURE in 1h 31m 31s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/d917762e7b914a2d882ff983bbec4151 : SUCCESS in 2h 28m 44s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9bd02f05835e4b1bbbffa984a339df18 : SUCCESS in 1h 26m 18s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/acc86f8bdf894b028d31da79f7bd30ca : SUCCESS in 1h 14m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/df021c9c819842c3803de3b40464d6a1 : SUCCESS in 4m 24s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/5825144d44c145b78cb949de2c109bf0 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6fdb478d28c785eef84a29789f4ac7f062a48a57","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-20 12:16:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1bc28ed386a486cf57bbbdb894df56c854ed092c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-20 14:36:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ffa514084983405296719f4fa51c9f39 : SUCCESS in 2h 11m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a60571b0598a49f9383196e1e91aa0bacd5a9bf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 15:59:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/794e6e4ddf7541a8bcf65d143e5abf22 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/177d6c3b1f1e4a0abfda1532e031ea44 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a2d05c417fc46f9bab156e496f8f780 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbeba7db7bc84842a97aa1b5a1a01e79 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20e3a8e152104bcf91828bd904a0afef : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf1701beaed4fc999c7441110df36e7 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7fc7d01817c4475fab4a629829aa19bd : SUCCESS in 3m 59s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/8a6ddfd03c2b4e059f03361e7923f898 : SUCCESS in 3h 28m 53s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f55bc1ea06954695a842eaa504d8e6ed : SUCCESS in 2h 11m 50s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/874cf8e8c900482db789eef27a04c32a : SUCCESS in 2h 16m 18s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/58b2e6161f9c4df9a3c9b6dab3d15777 : SUCCESS in 55m 05s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/bff32c419a1f4b5fa85b69ba3974ae9c : SUCCESS in 1h 26m 06s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/67dd354271b84a37a8472ff02fd3550d : SUCCESS in 1h 45m 33s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/3758aecc4bc54dac9a705e5554b4be9a : FAILURE in 1h 02m 29s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7474cad62df4444695a1fe48df8cc587 : SUCCESS in 1h 26m 41s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/b147edc4342a45c0975bc0c101e06b0c : SKIPPED\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dc69945ecda44aa597e26696b29ed03d : SUCCESS in 4m 42s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/b9038188d5bd4a0db5e58b0c1aea250f : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e1cbc96003eb33273d7d5e913edabf2f7f36c51c","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-06-22 15:02:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63081f1247d1cbbcc959e5c7430b94635a2b53a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 17:18:52.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fa62d826b8e4bf88e662f35f1849fa6 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ed761ee0d74f029c8f7d95b44e5b17 : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c13fec2cd6324e5e89a18b1443e31936 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af2af7ba5d534dc4a953d40fc4f09b3f : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06b05e4cfdae4061a1ae7a25c004723b : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e500f1178e742b7b80d5a96c366a6a6 : SUCCESS in 3m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd150250c1ce4dc2b6d28760d74b76fd : SUCCESS in 4m 21s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/5f384f86fec6440c964c7271674af236 : SUCCESS in 2h 07m 11s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/7ec81ab15d7d44f0b1a5292fccd42e60 : SUCCESS in 1h 17m 58s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/f9a8b51bfd32496d8a7287a4ef1b7985 : SUCCESS in 1h 14m 01s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/bdfdb8efada240f3ac28d98f476467af : SUCCESS in 37m 06s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/dcea52e66f6f4a93ab8eed1ea53a7b9f : SUCCESS in 1h 10m 45s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/54dae786b21d4b6198967fbde43482d5 : FAILURE in 44m 45s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/09dbaffbc5844b72be125c6c4c2a4004 : SUCCESS in 2h 04m 41s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/957c9b46483c4eee9c4c5caca9bc0a17 : SUCCESS in 1h 02m 15s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/739be633434b41a98c78265bace53979 : SUCCESS in 1h 08m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a7a769d28c0845a0ae1066e34451ae93 : SUCCESS in 5m 24s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/709331d2e3aa4b7cbacfe297a224a1ca : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"8176a5ab4b0df37cef6d78d4de6917aeb416bccd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-22 18:25:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5a2d85d7dcbe4b4cbe8e67fe5ad9f373 : FAILURE in 2h 36m 27s","accounts_in_message":[],"_revision_number":2},{"id":"f949d3f11c3a8129b87894bc1ceaf9405c527106","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-06-23 15:33:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39c2581ccd2a5c4af3a3ee3c454e6a793a046412","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 19:30:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f902705b1914d39b64c2add9f87a826 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84751f97870046aeaa01c40614579a75 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cb73b55c88b40ddbf30249b92cbc42c : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6350cf828839435f94171618e033753f : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd258288081e4be68eac3bb3cbefc92a : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcb3c1575e3d42bcb707db5a49ee2a39 : SUCCESS in 4m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/adf83e0af38841e5b57693e2228409d7 : SUCCESS in 4m 03s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/aa54455eb69149a08a69bcd9f425f152 : SUCCESS in 3h 43m 54s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/5c11b00dd8ca4f54834b7256d00e26ab : SUCCESS in 1h 54m 24s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/984ac2ce6fd44b1d8a7ed3e6bf7c644c : SUCCESS in 1h 38m 51s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/d356cfe2d5254adeafaeb7043cc34336 : SUCCESS in 49m 09s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/4d7035ec36b74690bd26bfca1984da01 : SUCCESS in 1h 27m 26s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/384e8074f47c4ca1be265b6ffbbed94c : SUCCESS in 1h 33m 42s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/ca8cc1cc62e141e0b9162ac80a140dee : SUCCESS in 3h 04m 18s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/66360c7d45a44843ba5e3018150b7fe0 : SUCCESS in 2h 16m 20s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/25cd9cf3c64345598bfd7ccb06222484 : SUCCESS in 1h 34m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0a4652ea5e8b427592f1891ffe848010 : SUCCESS in 4m 26s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/a24e8901d4ee4195af0e9d4351b26c5b : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"5f5252cbb75be711a042383431448cffdf3fbbb6","author":{"_account_id":21129,"name":"Alan Bishop","email":"abishopsweng@gmail.com","username":"ASBishop","status":"ex Red Hat"},"date":"2022-06-23 19:46:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"43386fa8440bd05a89861afd363be0092d64a29b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-23 20:18:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ec5e700dd3a2488d846c4bc83bcbce27 : SUCCESS in 2h 34m 14s","accounts_in_message":[],"_revision_number":2},{"id":"3039a94c9b4027be2ca70b354aa37d4d2bdb707f","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2022-06-28 20:00:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1854229b9949db36cf125788d5b5887bf3752bd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 20:02:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"15e4a02c04f1d337176342300f04cf1d3f7d9723","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 11:30:17.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c236b5a275c54dde9a05cccde718d7ee : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcbb505ed2f64eb5bfa524934a2009c3 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d25cb7d22e043aeb373ae36677e14ef : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10f872498a52403da33e43837e2b0294 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee582ee1005d4965b9d8e0efeb2d650c : SUCCESS in 3m 46s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/36add5f71ded4f79bfff3b1710e4c5d2 : SUCCESS in 2h 49m 56s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/b6421237057140a8a4869b6b103713dd : SUCCESS in 1h 41m 53s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/ca69f9e96787410dad81705f065f266b : SUCCESS in 46m 57s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/c3aeab6e71524b0598f9e60e1e903de2 : SUCCESS in 1h 20m 48s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/94ee3798996949179c0d9db50da8569d : FAILURE in 31m 08s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/8b4024dc06cb401f9230855e4a7bc9aa : SUCCESS in 2h 23m 56s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/6dfd115562ea401e9d741992be035030 : SUCCESS in 1h 10m 50s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b36d1d621c0d485298ec577b18581d8d : SUCCESS in 1h 30m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/25d8f0be929846d88cfd2d1c195ad56b : SUCCESS in 4m 26s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/d2575fb9c9d442e29db17f50f61234f3 : SUCCESS in 9m 43s","accounts_in_message":[],"_revision_number":2},{"id":"e137764fab4f5ede20c466cf88127a93efd18ea5","author":{"_account_id":29775,"name":"Sandeep Yadav","email":"sandyada@redhat.com","username":"sandeepyadav93"},"date":"2022-07-01 05:07:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1170f5f1d01ffc53ea707e2d11d4f0e6bedf4fdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 07:56:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4093fe11171483580f5ea56dbce2614 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1077f8db801d472a8ad4ced8c603e782 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d08756e84d6b43bca92002943e6910e8 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14709824bdf34d939340c7474c4ea29d : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/587cba36a61d457082b164803557e071 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cde51062113f429581cee8fa6ff247d4 : SUCCESS in 3m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fdb36db7eaa64bcc9bdd9db24339dcc2 : SUCCESS in 3m 23s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/772a999a2d104b6f80816811a7092586 : SUCCESS in 2h 42m 56s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/58631837d1bf47558decd9f3b3bc26ea : SUCCESS in 1h 42m 16s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/a4992228799a4943847682d835406707 : SUCCESS in 1h 29m 27s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/0d51fa4078724eebbb9d1eb01e5ce30b : SUCCESS in 40m 53s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/1c54f3817ed44ba88ef51c05788176b8 : SUCCESS in 1h 15m 55s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/f88bb9f49b8d4feb98dd33a52d401dd0 : SUCCESS in 1h 39m 07s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/34622c5a30e44068a2fdf67758a7ad87 : SUCCESS in 2h 20m 35s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/112a40f2c7464685b00af05a0218279c : SUCCESS in 1h 26m 37s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/9e2cec3862f143dcb7ff0aaea20404bc : SUCCESS in 1h 13m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/543743525d6248cbaba0721948969b5a : SUCCESS in 3m 42s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/020a22dc662547d58a750e1bd21c799a : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"d560c87d73db8030a9f3ad8c801141ca7ea0fea7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 07:58:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c1600e1c0dce5e6265697d3ae1dee7d840292e47","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-07-01 08:39:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/bf66ba4cf06745c8b08bca527455917b : SUCCESS in 2h 17m 46s","accounts_in_message":[],"_revision_number":2},{"id":"85f0e5d23e23b77f2c3ab857143fffce5d33f630","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 13:10:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/828003596e66412cbfe32887fc1e0526 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b11c29fce8c482a898821abdb59b34d : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3071e728ccec442e9f1861399d38a7d3 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a90254f664c24ff79d7b33c244e4c6ac : SUCCESS in 3m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03a8051af9c844cababeafd0eccb9c89 : SUCCESS in 3m 19s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/500d99a7789d4dfd9865375773c144cb : SUCCESS in 3h 18m 30s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f43eef69e0a0483293ff99bdcfcd5a99 : SUCCESS in 1h 55m 28s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/10296e3b782045fdae5e76dfa425a72e : SUCCESS in 52m 14s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/867381297e544afbb931865447a76be3 : SUCCESS in 1h 33m 57s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/99f1e662d3bb4be097bf2c0a6416a9fa : SUCCESS in 1h 56m 08s\n- tripleo-ci-centos-9-content-provider-wallaby https://zuul.opendev.org/t/openstack/build/c3a48d075ce1460795dda0e5479a7941 : SUCCESS in 2h 45m 08s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/bdb1bc8cf326438ebda6f3094c0531ed : SUCCESS in 1h 22m 17s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e5173f199ac648f1b092af9fddc0e68c : SUCCESS in 1h 39m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cfd41eac82e04c3ea10c3fc4d1cf0e4f : SUCCESS in 4m 17s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/0dd5db4d333d40909434e079e8f757ef : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":2},{"id":"99be3c21a10fd5bb68603ad824c1d261a3f404af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 13:10:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3367d35f80b72931fcb22600005e06a663caf9c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 13:12:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/521b5740584c4fac9add38ab4cc5f56f : SUCCESS in 1m 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a8c1d39c42e4573a23896a8f5b5f6f8 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":2},{"id":"224f3a74bff1db0de608a86c5e826c3c03ff3c4b","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-01 14:56:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8a1bbe884d5757253c3346cda8618dca10b81484","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2022-07-01 15:01:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1e94f7ebc10728dcf99ec7d706647db01ea6dbbd","author":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"date":"2022-07-01 15:07:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6d8b07b7f9500850cb08a64e436eeb2f38c00b11","revisions":{"d0cb348bda6213e9fe9021e82476326732518abf":{"kind":"REWORK","_number":1,"created":"2022-06-14 18:46:23.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/07/845807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/07/845807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/07/845807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/07/845807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/07/845807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/07/845807/1"}}},"commit":{"parents":[{"commit":"1ad453cca17e75b87ecc32ea1ec45e18f9b42404","subject":"Merge \"Drop collectd-dbi from collectd base container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1ad453cca17e75b87ecc32ea1ec45e18f9b42404"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-14 18:24:36.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-14 18:46:22.000000000","tz":120},"subject":"Reduce memory consumption in OpenStack services","message":"Reduce memory consumption in OpenStack services\n\nThis patch reduces memory usage of the different OpenStack Python\nservices by tuning glibc.\n\nThe specific tuning consist on disabling the per thread arenas and\ndisabling dynamic thresholds.\n\nThe primary objective of this change is the Cinder Backup service, which\nnot only suffers from high water mark memory usage, but is also uses\nexcessive memory with default values.\n\nJust after 10 restore operations it ends up using almost 1GB or RAM and\ndoes not ever free it afterwards.  With this patch the memory\nconsumption of the service is reduced down to almost 130MB.  If we add a\nrevert from Cinder (Change-Id I43a20c8687f12bc52b014611cc6977c4c3ca212c)\nit goes down to 100MB.\n\nIt\u0027s even worse on real deployments, where we have seen a reduction in\npeak memory usage, going down from 2.9GB to 1.2GB, and fixing the high\nwater mark issue that previously left the service constantly using 2GB\ngoing down to 140MB.\n\nThis glibc tuning is applied to all Python services, since it could be\npotentially helpful without real performance impact.\n\nThis is the equivalent of the devstack proposed patch from Change-Id\nIc9030d01468b3189350f83b04a8d1d346c489d3c\n\nChange-Id: I65b32f4ce3fddeb694fb33ca65076d45d23a3bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d0cb348bda6213e9fe9021e82476326732518abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d0cb348bda6213e9fe9021e82476326732518abf"}]},"branch":"refs/heads/master"},"6d8b07b7f9500850cb08a64e436eeb2f38c00b11":{"kind":"REWORK","_number":2,"created":"2022-06-17 10:43:15.000000000","uploader":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"ref":"refs/changes/07/845807/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/07/845807/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/07/845807/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/07/845807/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/07/845807/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/07/845807/2"}}},"commit":{"parents":[{"commit":"1ad453cca17e75b87ecc32ea1ec45e18f9b42404","subject":"Merge \"Drop collectd-dbi from collectd base container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/1ad453cca17e75b87ecc32ea1ec45e18f9b42404"}]}],"author":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-14 18:24:36.000000000","tz":120},"committer":{"name":"Gorka Eguileor","email":"geguileo@redhat.com","date":"2022-06-17 10:42:58.000000000","tz":120},"subject":"Reduce memory consumption in Cinder services","message":"Reduce memory consumption in Cinder services\n\nThis patch reduces memory usage for Cinder Volume and Backup services by\ntuning glibc.\n\nThe specific tuning consist on disabling the per thread arenas and\ndisabling dynamic thresholds.\n\nThe Cinder Backup service suffers from high water mark memory usage and\nuses excessive memory.  As an example just after 10 restore operations\nthe service uses almost 1GB of RAM and does not ever free it afterwards.\nWith this patch the memory consumption of the service is reduced down to\nalmost 130MB.  If we add a revert from Cinder (Change-Id\nI43a20c8687f12bc52b014611cc6977c4c3ca212c) it goes down to 100MB during\nmy tests.\n\nIt\u0027s even worse on real deployments, where we have seen a reduction in\npeak memory usage, going down from 2.9GB to 1.2GB, and fixing the high\nwater mark issue that previously left the service constantly using 2GB\ngoing down to 140MB.\n\nThis glibc tuning is not applied to all Python services because I\nhaven\u0027t done proper testings on them and at first glance they don\u0027t seem\nto have such great improvements.\n\nThis is the equivalent of the devstack proposed patch from Change-Id\nIc9030d01468b3189350f83b04a8d1d346c489d3c\n\nRelated-bug: #1908805\nChange-Id: I65b32f4ce3fddeb694fb33ca65076d45d23a3bb6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6d8b07b7f9500850cb08a64e436eeb2f38c00b11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6d8b07b7f9500850cb08a64e436eeb2f38c00b11"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}}]}],"submit_requirements":[]}
