)]}'
{"id":"openstack%2Ftripleo-common~847437","triplet_id":"openstack%2Ftripleo-common~stable%2Fwallaby~Ie5288864cd6f346a5bb5b481b4aa5dbd1abb9f47","project":"openstack/tripleo-common","branch":"stable/wallaby","topic":"cherrypick-bug/1978997-aqgasvnmmk","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2022-06-28 02:39:41.000000000","reason":"Change was submitted"},"9976":{"account":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"last_update":"2022-06-28 02:39:41.000000000","reason":"Change was submitted"},"12393":{"account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"last_update":"2022-06-24 13:07:23.000000000","reason":"\u003cGERRIT_ACCOUNT_12393\u003e replied on the change","reason_account":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}}},"hashtags":[],"change_id":"Ie5288864cd6f346a5bb5b481b4aa5dbd1abb9f47","subject":"Downgrade pacemaker and resource-agents","status":"MERGED","created":"2022-06-24 10:39:39.000000000","updated":"2022-06-28 02:42:33.000000000","submitted":"2022-06-28 02:39:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847437-cherrypick-bug/1978997-aqgasvnmmk","cherry_pick_of_change":846287,"cherry_pick_of_patch_set":2,"meta_rev_id":"511ecf9b69e3bbe64fbf9c4e34f416483d833f1c","_number":847437,"virtual_id_number":847437,"owner":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-28 02:39:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-06-24 13:49:25.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"recommended":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":1,"date":"2022-06-27 14:50:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-06-24 13:09:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":2,"date":"2022-06-24 13:07:23.000000000","_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2022-06-24 10:50:03.000000000","_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2022-06-24 23:34:14.000000000","_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":1,"date":"2022-06-25 10:44:57.000000000","_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}],"REVIEWER":[{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-24 10:39:39.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"CC"},{"updated":"2022-06-24 10:45:16.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2022-06-24 10:45:18.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"CC"},{"updated":"2022-06-24 10:50:03.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2022-06-24 13:49:25.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-24 14:20:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-27 14:50:54.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"8ca77ebdcf87a8126f54b5c6ff7fe29adb03f237","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-24 10:39:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a3bfdbe7a9003dc66dcb122c804e3c08ded618e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-24 10:45:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c704cea60ea1a834e7713b48a657961c841bd565","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-06-24 10:45:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9f4f7d1e2893bbf72f32a6a20731ff3c98569f0","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-24 10:47:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"497110259e134f3ef2117aa07c9828d278e4f105","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-24 10:50:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"df68204d7826f55257c7691e962f5faf15154ffb","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2022-06-24 13:07:23.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ae582de87c76395989a4edea6278a00d52822cf6","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-24 13:09:18.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"9520166c813aa984ff36471639faf09251be96e5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-24 13:49:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e9a5d942dccd46e0ae4f74a18b7f0d25 : FAILURE in 46m 35s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/87ffb1fe0c224d2088e6152896af2ab8 : SUCCESS in 2h 43m 27s","accounts_in_message":[],"_revision_number":2},{"id":"cfe53f350650ce4647cdbef2bd658c26f8c5abc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 14:20:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f1b2163f5d048abbb0ef93955218af7 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b14290274b4848b7f776d159006b37 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b66698415eea419da03ef2f8a70f7ee8 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3a59b834669409ab53a8d584bea86d1 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b219186bf9df4d2dbc509c527db0c1ce : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dfd9483375b4f3db2d8d21a7a6e3174 : SUCCESS in 4m 18s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/af633b8bac654bbc9d3a98c030739f42 : SUCCESS in 3h 27m 38s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/17e8de4eb22844229ae65294f87abacf : SUCCESS in 3h 18m 51s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/1e0a514f64fa4f21931063e7d7ab9996 : POST_FAILURE in 2h 25m 12s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/76233e1ab1e54f5591a53f5ef6f7cba9 : POST_FAILURE in 2h 22m 01s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/c8bf5f7cba5043eab9803e49d13c355d : SUCCESS in 59m 40s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b3e3a54702694547b6ba2a06443d7104 : SUCCESS in 1h 28m 23s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/0d3aaed30d974ec187e829ae0bd63313 : SUCCESS in 1h 16m 57s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/4867190a260847858cc3015b68bcdb34 : FAILURE in 1h 14m 44s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3732f918d00b4accbb367aba69821ec5 : SUCCESS in 1h 18m 21s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/eb84eedb31924f22acdc85533de827dc : SUCCESS in 3h 25m 13s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/a4f945b54faa43c0a484862d54164395 : SUCCESS in 1h 54m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9cf1103b03f14cd8aba059cfa95f9bc2 : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b9642bb8db0418e9104f44bf8d395c9 : SUCCESS in 5m 12s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/747dc1ceb6bb401da7710fc90f68a04a : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":2},{"id":"ae04d737bcbbabb2ad6da435cf3d6992e183aa64","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-24 23:34:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"87c81f8adc50308a2cdd0b4caa3e861a8470acbe","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-24 23:34:14.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f1964e41a7900a7106e8cdcdf93508f69c66fef9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-25 01:44:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2c0f542999074e4daa41e94f6b674a13 : FAILURE in 1h 04m 33s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b84d749d8da3470bbf73d07345a8e9ec : SUCCESS in 2h 09m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1794a80e69d6c4a610082c2a274fb17ecdb302bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 03:42:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ed1ef44211a443aaf63ba4b36f0f849 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e123d39dcedd4259920a33e9ecc8bd6d : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/101d78770715455891defb7045634bce : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64b362a38c7849c694627f6f0a8048e4 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/959c3765b9aa49d08be54d5468171df4 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8408115e695f4557a17d1824c4aa3c1d : SUCCESS in 3m 44s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ce852a50918f411ba05f3a19ee055acc : SUCCESS in 3h 55m 17s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/9fa6c8ade075406ba6f4a08ca99d486c : SUCCESS in 3h 47m 03s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/6d3ea1320df740e680dd02ecd7c72a8d : POST_FAILURE in 2h 39m 26s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/bc7218174f7e4b2ebf575a773fd8a784 : SUCCESS in 1h 49m 53s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/d4ce2101f09940379035b7ac3d69fd55 : SUCCESS in 1h 39m 58s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/5708ed7e0eb2458cb4dc5cceb652327c : SUCCESS in 1h 42m 23s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/4620820c24fc4312a62440f4ff12e007 : SUCCESS in 1h 27m 14s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/11af1086a63b49578600400026903637 : SUCCESS in 1h 46m 42s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1f4a24cffcb0486795f1d1bb6f8f09b6 : SUCCESS in 1h 29m 53s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1927d2889caa47dc94cf797bb72b48f9 : SUCCESS in 3h 55m 55s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/23a0d5748c7f4fe48158be461297633c : SUCCESS in 2h 08m 54s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/694a26ffd8e14aa4b3b56335a2295127 : SUCCESS in 3m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/338ad9046d524aadaa4f9521f63dee51 : SUCCESS in 4m 13s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/7088b64c11904159ba5be75e632e7a7b : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"be7636dc99e3e307c8e86f6e843761407d98480c","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-25 06:38:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a9abdb700f7351b5a76642f5a8e029afbf7ddacd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-25 08:39:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/07a45d3a4acc4bfbab4a733e64d5e48e : FAILURE in 40m 33s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9568ac81a89145ed972d1ce7b784ed2c : SUCCESS in 1h 59m 42s","accounts_in_message":[],"_revision_number":2},{"id":"04b8765ed19130ed5291e4e8a1d3fbf86e4e1f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 10:43:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44368c5f430f42fe9c738164403f0068 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee330f2dfe994f709536caa2466573d6 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d668b71a775474facc30630cff17af0 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9689fef7d85406a96f64a77fb28b00c : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58a2aa645ecb4aeebdc9338964b6c0c8 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de05f686d6d40ee8431593f58312884 : SUCCESS in 3m 39s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/a23e215d7ec24b0d9819422aaae6a4b6 : SUCCESS in 3h 19m 25s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/124af57789bc414e803c9e859ff9633e : SUCCESS in 3h 58m 15s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/8b28792459fc4e13876f2b77b83cefee : POST_FAILURE in 2h 18m 01s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/87e46483e72d4f64b2b0e7e25809e0d9 : SUCCESS in 1h 21m 02s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/94625a8f59d14050ac0b419a1a0c7fab : SUCCESS in 56m 59s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/b95b87dc70504606a75d34968caad8b4 : SUCCESS in 1h 21m 03s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ef126c38a8714b14a4d726b30cdf4b2a : FAILURE in 1h 14m 08s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c1cd4ceeb0f742cea3822f1755cea1fa : SUCCESS in 1h 22m 44s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b2f7902d6e794a01af4ac98e7331ba38 : SUCCESS in 2h 00m 15s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/9c43246256be476eb645cfa8d7f9c83e : SUCCESS in 4h 03m 01s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/0a0cf6b8ae6a44a492fe55e913e29b6d : SUCCESS in 2h 34m 38s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/45c934aec5e94cf6abd7c7bd17de88c6 : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19e8625ac2474f6ca4f10c5a1e2c4aea : SUCCESS in 3m 38s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/99e672b20fa842ec97dac1658652966e : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":2},{"id":"581dfa77863851c957a3b1f029a628d858bfe068","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-25 10:44:57.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e9990468df7c2ffda4add182c62899e626c50405","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-25 13:26:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b9c927c567a54defb7ce009317749196 : FAILURE in 38m 01s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3d1b9e83ac2045299652fb5877f1ad2b : SUCCESS in 2h 40m 39s","accounts_in_message":[],"_revision_number":2},{"id":"86bceb3090cd629c45089fdc858f5c23d6ac2fdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 14:49:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/664e5ddacd914f739b369ae1149f89b1 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a962bc4ebd4eae913e7ab74e0304a3 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f643abd94114d11b8a15c7f9702b78f : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82aba2666f9a4dc2b68a30ce93234c22 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7db47dcd3134853a295e759e73fb99b : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/827e2dadc74741efbbdf527cca408c6b : SUCCESS in 3m 28s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/ff1aba0f4e9249479c25a2b463e866c6 : SUCCESS in 3h 28m 12s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3327af1bf20342fdbf0674039bc2d83a : SUCCESS in 3h 54m 20s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f943fbe6df1240c3b4188264797a3d9a : SUCCESS in 1h 26m 41s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5164fdc1294d459d99f3e871d3f66feb : SUCCESS in 1h 24m 16s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/19079a809a5b4a0e8d52b0e237d661ab : SUCCESS in 1h 25m 18s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/69ebb8c757414ccaba572aa8d131da0f : SUCCESS in 1h 12m 18s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/0ca0b18e6b4045b790977ca3e3ffa933 : FAILURE in 1h 15m 17s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/9cd5f43b252a41608083a7b679283623 : SUCCESS in 1h 14m 03s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/6bfe191bf87540d8a77def9184673352 : SUCCESS in 2h 00m 07s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/7d8a4cb254834e3ebe0022ba47790884 : SUCCESS in 4h 02m 44s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/fb9e7616df2a49b4bf579ce0cb67e8a8 : SUCCESS in 2h 24m 43s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5091d6d8286a4e788fe00940f007f11e : SUCCESS in 3m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f9a30c8c0e424c1eb9da5f3aca8c1360 : SUCCESS in 4m 53s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/870781cd0bd143d3a1f2c18e48500249 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"62d10125c058c8f4b6f2fc5c476d9e15d0447340","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-26 14:01:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f8e87e5b6dad22568377be5f8a3247bf166eaf5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-26 16:50:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/67704af013154ac58a44f16a8b70efc0 : FAILURE in 38m 11s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4fa49fdd717944188b28e8f269a96fda : SUCCESS in 2h 13m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a723f633fd29723914397d4149cdc6d88438a110","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-26 18:24:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa86f5a85d064db1a43fb13cb23f5f71 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75e07060fa1b4430a491f8f4419e7f9d : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a62a51296574681bd6db10853da429c : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1bb18ee98f0346b795d3132ac522c7b6 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f07e16c201a54a478fc0cd720b99ec85 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee458d67bc414627b76708e2ee23dbe5 : SUCCESS in 3m 39s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/73462757f772483eb4c22e2db1648dd5 : SUCCESS in 3h 22m 40s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/611bf39e76c845f38261ef7050a3e076 : SUCCESS in 4h 09m 56s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/e6cde8eebb814bd68814f0050b2c6f4e : SUCCESS in 1h 36m 14s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6b0f27541b3e4cfe9538f21c27127b8c : SUCCESS in 1h 36m 49s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/66d982ef934642ca9b7521fc92c577d8 : SUCCESS in 1h 05m 11s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/f9235d7c53fa4feb8e41cbd266c783fd : SUCCESS in 1h 30m 28s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/6c0171d30690482ea6b3eb4dd8ed9be0 : FAILURE in 2h 11m 47s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/97069b0251ed4e0ead1279df2b94aace : SUCCESS in 1h 46m 37s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/a2f81ed05cf945acbf6eb9ac56357467 : SUCCESS in 1h 27m 28s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/96c08fbed6464bc9904e16f43b277437 : SUCCESS in 4h 00m 16s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/cbe1d2de74fc4a46b0fb8b3801c24a82 : SUCCESS in 2h 06m 08s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00ed7128ef854c849fb0424c559ef597 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3464bbcbfe5a419394721a863a16a7ef : SUCCESS in 4m 41s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/1dc9212515294f899ffca82bdc3c67f0 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":2},{"id":"456d45dc45f41ad47a407c973980cfe8d725552f","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-26 20:11:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c82a34d068aebf5545d9a31bc66ef6e625b0af8","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-26 22:50:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/661a361230734963825f861f3656c43e : FAILURE in 39m 39s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/70de05869fb247bcbb7dd6864784cb33 : SUCCESS in 2h 02m 20s","accounts_in_message":[],"_revision_number":2},{"id":"7b7d7064567d3e40e83569a1e5c2d49e80a120b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-26 23:09:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f85710bc8c1d4707a6c17fff326cd4f2 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6d4595f36ee4fb9af0d5573fbb87871 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a38800b7bce4728a5a6dd04b99a2e1a : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b354ebabd6d4caaaf86e5dec8a612c6 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c4decb21262450da473d6498b886719 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcf78962960442faaa3a420f2216887c : SUCCESS in 3m 09s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7afc6e40f689407bae795a154219061a : SUCCESS in 2h 03m 22s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6d45d455741b463bb08c197bb9d4d078 : SUCCESS in 2h 41m 43s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/34908daf8e624dbc8d7d53d2ebbef37c : SUCCESS in 1h 16m 49s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/ce9d0852c554428381a886d15c485bf4 : SUCCESS in 1h 13m 38s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/94fc13fd7b3940969ca51492e066fbd6 : SUCCESS in 43m 05s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/fcba99e2eefa4c20a150936a2d4de412 : SUCCESS in 1h 11m 20s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/dd65f98bcc7d486f8b9719677d118a44 : SUCCESS in 1h 02m 05s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/875f4b38a2fe49b895a3b3a99b0c36b7 : SUCCESS in 1h 12m 45s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/84dc05ad125c42658c1e9ddaccfb547e : SUCCESS in 1h 01m 01s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/26cec4a450594f818cec0542841ec73c : SUCCESS in 2h 56m 08s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/68a49ef265ec49c78ca1f479ac4ec6eb : SUCCESS in 1h 34m 04s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f110b43af7b74a239dcbca1b896d4c38 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea9cdaa9dc4a44d3b271c8a84b4019e9 : SUCCESS in 4m 34s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/24344ad546964ccf91d7b6fdb528a6bd : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"3521c84e529febbe1b0f6db7c59834419f9541f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-26 23:10:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e89c34b36004a835ae2e79a9df5c6d6108bbbbad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 09:53:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d841f7c4a23d4f1f883baba77468d0b6 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/742643105c134284b077dc47d85d9a61 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d6f21e11b884769a4f09a82c6472fbd : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78aac8d7e298413f9108547353e9be31 : SUCCESS in 4m 47s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/71c6a1ad4b5e410b96f4d81d8a52aaaa : SUCCESS in 3h 58m 49s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7a4763b857e04ee8ac31e61d1dbf63a9 : SUCCESS in 4h 09m 45s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/48fd6b428016408290762785e5828055 : POST_FAILURE in 2h 38m 51s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/9eacfdbad7a740189d25faf88d71b6a6 : SUCCESS in 1h 13m 26s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/ebe273bce22843638856546ed2b723d3 : SUCCESS in 1h 40m 38s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/6687bf216a7a442699264b003a50db73 : SUCCESS in 1h 32m 33s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b6cb493457884d02b7aeb1c803374c87 : SUCCESS in 1h 34m 16s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/cf62d4de9cf54cd6a6d5cb128c04ab1f : SUCCESS in 3h 53m 21s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/b143f7ddaa7645ee910ee50dee79684f : SUCCESS in 1h 32m 54s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/cc65b6d9fd5c479f9bb4e746e0d355f5 : SUCCESS in 2h 04m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/066b41f3f43f4b6b83938a03ac17beef : SUCCESS in 4m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7885794bb9b245008f1c4865af0d958e : SUCCESS in 3m 53s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/215479ba973147b88004e1fab2acb738 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":2},{"id":"46ec0cdffe923282ff6b971129d8c6efa719a12b","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2022-06-27 09:55:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4438c4855bb8d1150ea76e0ed9c05ae6a49407b2","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-27 12:13:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0b962c8352734484b8f53b80c48d9b54 : FAILURE in 37m 49s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d6481b4c67a446eab6fc567a874dc0c5 : SUCCESS in 2h 03m 45s","accounts_in_message":[],"_revision_number":2},{"id":"2fa38fb419389336781299a24ba0821799918bbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 13:45:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/609724d86b0d429986c499c88bf016e2 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07c896ce240b441d91ee49ca92e6c09c : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7d8792a9aaf4154b6ebc19a1451fcdc : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3932628ba4df4367878a93f93eb4cbea : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26846e29ea6d495fab9d5596127e8b4f : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a878e10baa234134bce9c9cdb8725cad : SUCCESS in 4m 00s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/379373aeec694ce1b46c2ea340a6d215 : SUCCESS in 3h 19m 06s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5ed00aad6f304587b166e57973ca44a0 : SUCCESS in 3h 40m 23s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f8ffbe87ef394f4d88da7b801252d628 : SUCCESS in 1h 28m 30s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/3b10bd2c49e94883bc9899a23001d4f6 : SUCCESS in 1h 21m 02s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/aeb855ec77cf4dccb44825493288e5fa : SUCCESS in 57m 01s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/2f88e87ec79a4f009800bed98ef7798c : SUCCESS in 2h 10m 35s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/c5b3973ce93841299fff6c6ead845ffe : SUCCESS in 1h 10m 39s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/b5ff11e0269f46e3b3359807168c2ac1 : SUCCESS in 1h 22m 20s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e5bbb9a55c6a40e8a537b9d1e492d6cf : POST_FAILURE in 2h 08m 03s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/51efe6e1514141ec829dd2921dacf00e : SUCCESS in 3h 24m 01s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/19bfb63e77ae4c1db606c555f8286c8c : SUCCESS in 1h 49m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a6d893f3aea04e41b7ae74f49ba7484b : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/62bf44bc64244e75a2a4e5800586baa3 : SUCCESS in 4m 06s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/618a3b3dd21c40689f81b99732bbbd2f : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":2},{"id":"345314cdbed6e3aaa4adf24376c2745155003fad","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-06-27 13:48:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4e5aa01bf7a6f00bb950b8954dc9c33bb375398","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-27 14:50:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"afe98895c3022b6b91353ffd9574d13d35f6ffc5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-06-27 16:04:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/01db26d484a944cd8837845d5bca4d9d : FAILURE in 41m 28s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c105aa157d9a40e18d79dc42878f8e3f : FAILURE in 2h 14m 43s","accounts_in_message":[],"_revision_number":2},{"id":"06c5f57d2967f00f6167f20706e9c171287b1373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 17:34:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/728edde9e5824766bcb10e0ae75c59a4 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45db6af08e0342d5a0cb1a991bf0e307 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ee8e7dc629443c78a6fa98f82bfc1a5 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1e76431961e4cc3b69ec966a41bff1b : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e26819ae729e4a50a0f488c1dc1cce4a : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62bdb0d1fff4911a2a026f4bef70204 : SUCCESS in 4m 09s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7fa47b0846404c44b84c8ba3a345fa28 : SUCCESS in 2h 52m 21s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/56616eaaa97a44f9a84615f2969b7239 : SUCCESS in 3h 38m 39s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/079c838c39cd44c886f884916373fd8c : SUCCESS in 1h 36m 25s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e78c545f6db24a438fd4df562ff8edeb : SUCCESS in 1h 37m 30s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/d8c2e5cd9ce946dab32adb8001fb1ff7 : SUCCESS in 1h 04m 28s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/c792a6cc41aa4cedb8274ee908aa2519 : SUCCESS in 2h 03m 51s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/7230975e496b4082b6a87e8b62869c92 : SUCCESS in 1h 18m 36s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/432fef331279461e94caadbdf22b1d5b : SUCCESS in 1h 28m 15s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e40149c2586b47668cfa11a91579c107 : SUCCESS in 1h 06m 43s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/d88153ee81e84df89ee909b4353826fe : SUCCESS in 3h 28m 22s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/32f55a4f74164b73a2b781d723880683 : SUCCESS in 1h 46m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/77f668cfd802433ebc6819c850e9dd0e : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e27d69e5fec42218e233d11b7938596 : SUCCESS in 4m 53s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/b8326b6c2caa4f0aacd7a88d74c4dc0f : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":2},{"id":"3bb9fc65accb8c6b3382f56185c23fa1c68f3ab3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 17:36:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b29926db9594b95f9628c034684ade4b231e80d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 02:39:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/820d5b7fa32b40c0826910855bc10588 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b1c24bc5e8c4cd5afab2333e2e11666 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f2a14a3d93d41e4aea0ba6944c3b78c : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b4f15bb19544301a2af743620f4434f : SUCCESS in 4m 21s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/9b8c69253ef140a99d0c3034c782f0c5 : SUCCESS in 3h 05m 49s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7f56fd00d08e452eb282dba3e758876f : SUCCESS in 3h 56m 53s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/188eb64a56df4a1f9511993626c7ba33 : SUCCESS in 1h 42m 39s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/c3652cd4403e4bb4b21492718dd40c28 : SUCCESS in 1h 07m 19s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/cabdc84a7ec7450c991034c002a75ac7 : SUCCESS in 1h 42m 54s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/1e4b3e9bd6694af2a80dba682d15e47d : SUCCESS in 1h 24m 33s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c64a7c38daa34c8c8ec69e442cf5e9ad : SUCCESS in 1h 46m 50s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/a713e6c45fac4f0dae7d94697966a6c2 : SUCCESS in 3h 54m 42s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5b2ec1417ba048f98fee8d84d5c2f75c : SUCCESS in 1h 37m 37s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/fd8a0c4083b04128a939e5e64a9a33b3 : SUCCESS in 2h 10m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33dc8eb8a3554971b87b0952d16b13bf : SUCCESS in 5m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7fc7942b15cc4e399baa810a954595d4 : SUCCESS in 4m 38s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/8bbe049a3cab470fb04e3fbc60fd8c19 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":2},{"id":"9b4638be5e934ddbe70178515d3bc49b23499bd5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 02:39:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"511ecf9b69e3bbe64fbf9c4e34f416483d833f1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 02:42:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56faa953052b4baeab2aa7e7cb7fcbef : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e8c7d086a48429c4eade1518955fb765a74d464f","revisions":{"a4660f1b081ff9a27e49908286b3e1411ea4b213":{"kind":"REWORK","_number":1,"created":"2022-06-24 10:39:39.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/37/847437/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/37/847437/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/37/847437/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/37/847437/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/37/847437/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/37/847437/1"}}},"commit":{"parents":[{"commit":"d80478555b23d779af034cf5e76a018852b3beb8","subject":"Merge \"Add rsync-daemon to swift container\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d80478555b23d779af034cf5e76a018852b3beb8"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-06-17 09:21:49.000000000","tz":330},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-06-24 10:39:39.000000000","tz":0},"subject":"Downgrade pacemaker and resource-agents","message":"Downgrade pacemaker and resource-agents\n\nThe patch downgrades:\npacemaker pacemaker-remote resource-agents\nin container builds to avoid\nerrors at deployment step 5 with latest versions.\n\nRelated-Bug: #1978997\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ie5288864cd6f346a5bb5b481b4aa5dbd1abb9f47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a4660f1b081ff9a27e49908286b3e1411ea4b213"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a4660f1b081ff9a27e49908286b3e1411ea4b213"}]},"branch":"refs/heads/stable/wallaby"},"e8c7d086a48429c4eade1518955fb765a74d464f":{"kind":"REWORK","_number":2,"created":"2022-06-24 10:45:16.000000000","uploader":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"ref":"refs/changes/37/847437/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/37/847437/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/37/847437/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/37/847437/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/37/847437/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/37/847437/2"}}},"commit":{"parents":[{"commit":"d80478555b23d779af034cf5e76a018852b3beb8","subject":"Merge \"Add rsync-daemon to swift container\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d80478555b23d779af034cf5e76a018852b3beb8"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2022-06-17 09:21:49.000000000","tz":330},"committer":{"name":"Ronelle Landy","email":"rlandy@redhat.com","date":"2022-06-24 10:44:24.000000000","tz":-240},"subject":"Downgrade pacemaker and resource-agents","message":"Downgrade pacemaker and resource-agents\n\nThe patch downgrades:\npacemaker pacemaker-remote resource-agents\nin container builds to avoid\nerrors at deployment step 5 with latest versions.\n\nThis patch also specifies the downgrade version.\n\nRelated-Bug: #1978997\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\nChange-Id: Ie5288864cd6f346a5bb5b481b4aa5dbd1abb9f47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e8c7d086a48429c4eade1518955fb765a74d464f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e8c7d086a48429c4eade1518955fb765a74d464f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}}]}],"submit_requirements":[]}
