)]}'
{"id":"openstack%2Ftripleo-common~862249","triplet_id":"openstack%2Ftripleo-common~stable%2Fwallaby~I03342e8107a214341990584f039dd5fdca2fece4","project":"openstack/tripleo-common","branch":"stable/wallaby","topic":"blkid-stable/wallaby","attention_set":{},"removed_from_attention_set":{"8449":{"account":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"last_update":"2022-10-28 00:05:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03342e8107a214341990584f039dd5fdca2fece4","subject":"tripleo-mount-image: replace most lsblk calls with blkid","status":"MERGED","created":"2022-10-25 08:44:49.000000000","updated":"2022-10-28 00:06:40.000000000","submitted":"2022-10-28 00:05:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862249-blkid-stable/wallaby","cherry_pick_of_change":862535,"cherry_pick_of_patch_set":1,"meta_rev_id":"da7da9057a564aa6d6582dd7820d7eecb6073b18","_number":862249,"virtual_id_number":862249,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-28 00:05:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2022-10-25 12:47:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"recommended":{"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},"all":[{"value":1,"date":"2022-10-26 14:11:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},{"value":2,"date":"2022-10-27 15:49:28.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2022-10-27 16:15:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":0,"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},{"value":1,"date":"2022-10-27 20:53:39.000000000","_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}],"REVIEWER":[{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},{"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-25 08:44:49.000000000","updated_by":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"CC"},{"updated":"2022-10-25 11:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-25 12:47:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-26 14:11:13.000000000","updated_by":{"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},"reviewer":{"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},"state":"REVIEWER"},{"updated":"2022-10-27 15:49:28.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2022-10-27 16:15:23.000000000","updated_by":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"reviewer":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"state":"REVIEWER"}],"messages":[{"id":"906d8751370b2e5081fc5bf9c25711fee0ccc1c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-10-25 08:44:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"84388bfb33b705eae9f1f6f035a18ad143ccb213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2022-10-25 08:45:53.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e42609dfce19f54d778fc0e2c2e5ca10b02c1bbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 11:46:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee8af8fe161a4e1c85cdf96cd5e08739\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4f954040dc843e38bf942b54a324068 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02f694bcd93546a6895932cd443813ed : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5aa23f41a2b46959c28ddb51762a339 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9f519e5c64549f0aff0f26c49289da8 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fed20a8b1ca4d9bb6a6c6bf57bbc0d3 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1e69de344a141af825d294e8d309c76 : SUCCESS in 4m 21s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/2d40df97673843709270de8775e02db1 : SUCCESS in 2h 18m 05s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/6cdaf2cc871a4dcab1590f8edba3dd53 : SUCCESS in 2h 52m 36s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/37047382b6fb4777aeac1559195523c3 : SUCCESS in 1h 18m 20s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/7ec1cec38d2a4b919479e1a7244ad85a : SUCCESS in 1h 14m 45s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/0932abbd54d94659aa1c3d1fd02c4fd3 : FAILURE in 1h 04m 44s (non-voting)\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/481126f854c84b158fb4e14d5cd0fb24 : SUCCESS in 1h 06m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5af8d7d7741941108044acb715ff2971 : SUCCESS in 3m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c9ae0867203c4738ae46c7f814751e49 : SUCCESS in 4m 28s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/1d73b6f1e8d04569abf9a5358b7f1fca : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":2},{"id":"72f3d6ec4b1a540f4ce422dd8c7ffd44f64d42b7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-25 12:47:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5528c63f8247444488c41e16c11023e1 : SUCCESS in 2h 48m 45s\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a8eeea5b00df46eea146773d8f40ab62 : FAILURE in 25m 26s","accounts_in_message":[],"_revision_number":2},{"id":"9259299d63d20ecdc22301148a965fdd9b5d8096","author":{"_account_id":14611,"name":"dasm","display_name":"dasm","email":"dsmigiel@redhat.com","username":"dasm"},"date":"2022-10-26 14:11:13.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1ac8b68a945c9bcd9e6682c5034606a9b8186f63","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-10-27 15:49:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"510815e0c7de55eaa17e77315b5b95a5f545c6e5","author":{"_account_id":9976,"name":"Ronelle Landy","email":"rlandy@redhat.com","username":"rlandy"},"date":"2022-10-27 16:15:23.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d0ef32399cff6065abf6b2cca2c361cdca60bae0","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-10-27 20:53:39.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8da8847d227b2be8ab8ea3ed17989a32558a920e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 20:54:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3e2b053c593686b166e78e59b6373b9917d82132","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 00:05:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e42d3433ece94d08aacfc7809e14ab52\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db097e9351fc4df0b23528b766366372 : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95beb39e16bb4d3c8674071ae2e9bee0 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/317510dc13964c2e910b49d9a39661f2 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1e85ba0dc543cf899f77920d4e831f : SUCCESS in 4m 26s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/da0a1cdc6f664037bfba6ce5bae36973 : SUCCESS in 3h 04m 19s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3d47d77dd06046f4accfb4cd398870ef : SUCCESS in 3h 11m 07s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/6e425b7a59424184af1aaf17f8fe983a : SUCCESS in 1h 30m 55s\n- tripleo-ci-centos-8-9-multinode-mixed-os https://zuul.opendev.org/t/openstack/build/99cc57459b424ab3bc407f4b598c56e9 : SUCCESS in 1h 54m 25s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/2585042e5b0545218fbd48a9de16d9be : SUCCESS in 56m 52s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/8e6ecf73fe64471fb6d6c2a7f5989753 : SUCCESS in 1h 17m 56s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/886d14e760e34cf6893ef46e5399fe37 : SUCCESS in 1h 12m 01s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/2c97a8c4de3041cfa29e26ee73bf15d3 : SUCCESS in 1h 00m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6953f45be784b9a889ac8599adfe5cc : SUCCESS in 6m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3b26d54677254a709fa8948fcc62c8b3 : SUCCESS in 5m 08s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/3a827331f79044ed9877b86e1fab465d : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":2},{"id":"937936c06fbf01bb77f1f9622a2bb1252a354ac1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 00:05:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"da7da9057a564aa6d6582dd7820d7eecb6073b18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 00:06:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ae9b3dbaf1c4e889ad38993ffe632cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8d42563b33a4eb996894ed1eb1947e2 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c169beb8794d707d4d567c50cb32bc67fabeffba","revisions":{"4b024343ec9bb608871f88b3a3558775f83d94c0":{"kind":"REWORK","_number":1,"created":"2022-10-25 08:44:49.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/49/862249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/49/862249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/862249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/862249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/862249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/49/862249/1"}}},"commit":{"parents":[{"commit":"a8e39db4225754c0ed0affceca7320142e9b221b","subject":"Add chroot support to tripleo-mount-image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a8e39db4225754c0ed0affceca7320142e9b221b"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-10-24 21:43:20.000000000","tz":780},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-10-25 08:44:49.000000000","tz":0},"subject":"tripleo-mount-image: replace most lsblk calls with blkid","message":"tripleo-mount-image: replace most lsblk calls with blkid\n\nThe version of lsblk on centos8 is missing parameters which this\nscript relies on. This change switches some lsblk calls to blkid\nequivalent to improve support on hosts older than centos9.\n\nThere is a similar change proposed to diskimage-builder[1]. This\nchange addresses a specific failure on a wallaby cs8 job failure[2],\nso this change will be backported.\n\n[1] https://review.opendev.org/c/openstack/diskimage-builder/+/850882\n[2] https://review.rdoproject.org/zuul/build/3830a8cf87944db495783e43ac5d41bf/log/job-output.txt#5281\n\nChange-Id: I03342e8107a214341990584f039dd5fdca2fece4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4b024343ec9bb608871f88b3a3558775f83d94c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4b024343ec9bb608871f88b3a3558775f83d94c0"}]},"branch":"refs/heads/stable/wallaby"},"c169beb8794d707d4d567c50cb32bc67fabeffba":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-10-25 08:45:53.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/49/862249/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/49/862249/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/862249/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/862249/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/49/862249/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/49/862249/2"}}},"commit":{"parents":[{"commit":"a8e39db4225754c0ed0affceca7320142e9b221b","subject":"Add chroot support to tripleo-mount-image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/a8e39db4225754c0ed0affceca7320142e9b221b"}]}],"author":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2022-10-24 21:43:20.000000000","tz":780},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2022-10-25 08:45:53.000000000","tz":0},"subject":"tripleo-mount-image: replace most lsblk calls with blkid","message":"tripleo-mount-image: replace most lsblk calls with blkid\n\nThe version of lsblk on centos8 is missing parameters which this\nscript relies on. This change switches some lsblk calls to blkid\nequivalent to improve support on hosts older than centos9.\n\nThere is a similar change proposed to diskimage-builder[1]. This\nchange addresses a specific failure on a wallaby cs8 job failure[2],\nso this change will be backported.\n\n[1] https://review.opendev.org/c/openstack/diskimage-builder/+/850882\n[2] https://review.rdoproject.org/zuul/build/3830a8cf87944db495783e43ac5d41bf/log/job-output.txt#5281\n\n(cherry picked from 704cdb267faf67675eec86c28bf633bf29524651)\nChange-Id: I03342e8107a214341990584f039dd5fdca2fece4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c169beb8794d707d4d567c50cb32bc67fabeffba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c169beb8794d707d4d567c50cb32bc67fabeffba"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}}]}],"submit_requirements":[]}
