)]}'
{"id":"openstack%2Ftripleo-common~862887","triplet_id":"openstack%2Ftripleo-common~master~Ic96679548a553960f1242850e42f9eec6e4a45f0","project":"openstack/tripleo-common","branch":"master","topic":"fix-bearer-cache","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2022-11-11 05:21:32.000000000","reason":"\u003cGERRIT_ACCOUNT_8833\u003e replied on the change","reason_account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-14 01:54:33.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"30073":{"account":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"last_update":"2022-11-18 13:48:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic96679548a553960f1242850e42f9eec6e4a45f0","subject":"Fix token cache expires_at check","status":"MERGED","created":"2022-10-28 12:15:44.000000000","updated":"2022-11-18 13:49:53.000000000","submitted":"2022-11-18 13:48:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":27,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"862887-fix-bearer-cache","meta_rev_id":"16971d40cf7cdf133c5b0bdfb11bbdc03d524740","_number":862887,"virtual_id_number":862887,"owner":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-18 13:48:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-11-18 08:54:45.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"all":[{"value":2,"date":"2022-11-18 06:28:40.000000000","_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":2,"date":"2022-11-18 10:24:32.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"value":1,"date":"2022-11-18 10:24:32.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-28 14:55:49.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-28 15:22:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-31 01:36:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2022-10-31 04:16:25.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"CC"},{"updated":"2022-10-31 04:55:53.000000000","updated_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"reviewer":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"state":"REVIEWER"}],"messages":[{"id":"4d54a5d10e6776ec8f5f94b92977d17ef609f04a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-28 12:15:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"324c8d9e542918c9ba3453149a62f11df216a227","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-28 12:54:51.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d22137513dcb3b1d0d4e2dd004e3981a8d1cb44f","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-28 12:56:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dfb114c2daf3ab88b65d57f50cc34e2e4448180f","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-28 14:55:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/b74759fb560d4ec0a19d10295a98dea4 : SUCCESS in 2h 38m 42s","accounts_in_message":[],"_revision_number":1},{"id":"a863952f290835c7c872b6e2bd8acccd6cd8ed94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 15:22:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f8d4f03482f473aaec9328c653c484d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ee3fb752fcc44a1b53604f54fbff227 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd76cd202be406fbfa4ed621c74309b : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e887278aefa425ba2488c3c15ab007f : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b207d605d294f55bcd2e53b404ecbcb : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0dec69ab0b6a4f34adbd6286c1b41e30 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a580f4643a0e48389ddd667b2d73e7d3 : SUCCESS in 3m 09s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/1ccbd08de28d4f6f9b6e2ae02129aee0 : SUCCESS in 2h 59m 59s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/a7c94f411dbc48cfaf651a5e9e25f049 : SUCCESS in 1h 39m 18s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/e304cbe9358a4ff0b199bd9f93676869 : SUCCESS in 1h 52m 28s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/8b8ba50180e54076a4b0185b66b645d2 : SUCCESS in 41m 17s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/5e88b02354ba452e9245c3a80f2e2c09 : SUCCESS in 1h 21m 13s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/16dffa31f02140229c08bd6a38ebbb61 : SUCCESS in 1h 36m 57s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ac3fb5e45f554f429b84dcd7d191872d : SUCCESS in 1h 25m 11s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/a76f4a8f6fe5490091f2ebd4b8d903a7 : FAILURE in 1h 04m 53s (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/ea2d1a7d40a648d09995e1fcb8ad3bb3 : SUCCESS in 2h 22m 26s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/d04e90703a13443f83372e192eb6f3fa : SUCCESS in 1h 25m 37s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/2f1fe5e169f547588a85ff68b92a3499 : SUCCESS in 1h 12m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fa4aad4556d74077952693419c27d67f : SUCCESS in 4m 07s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/ad7e77ba4fb749e49f63360d9406616d : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":1},{"id":"93a5f9ec1fdb45c5f8130b43e4c3e1b228726784","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-29 06:38:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6705284c1eb276fe6da5d93c8ab5f67cb488ef54","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-29 06:41:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"16ffc70353feb0fad7633a6b2ffbab9b4cf7e0a5","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-29 06:42:02.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"ad057e23a6b224e93d588cffae0021bfa3b1ce20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-29 06:44:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"65334bdc59f7b5741948188ad8d1e182ce05346b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-29 06:52:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7534c83c7ecd1652bd2d70748e0dbb27411714ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-29 07:02:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f432b16ca59856731afef76ad4a3a10795ceeb86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-29 09:23:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41d31f308a77418bb55d59f4265e913c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fe5c50a310e4581b0a80c899e7e5894 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f49bedd90de4cb1ab571823a228760f : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70cdcf1a86e64120a0aa3b019e88c2e1 : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6927540aca54c18b323d0d59af6daa1 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a4ad519e40a4ab9a30652eb551a90fa : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a376b250b520418097fa2276f14610e0 : SUCCESS in 2m 52s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/71a60d4d8bee4d5a9a9bee9255897cc7 : SUCCESS in 2h 16m 15s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/2fce15cf40a144de92c6e19f3a1d278c : SUCCESS in 1h 23m 13s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/9a769a98ecec4d6483a3b73c439ec4bb : SUCCESS in 1h 19m 55s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/6dd95258559a4592b476e5bc2f049251 : SUCCESS in 35m 42s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/062cc34ad5a347cba3f871938173bee5 : SUCCESS in 1h 07m 20s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/5a76d4c8824346ecbe0b7689c5a152a8 : SUCCESS in 1h 23m 37s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ef60566a84f14c70baaf8d1e9e9d4645 : SUCCESS in 1h 14m 36s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/5a9558f991a446e386e92aee1c8cfdd0 : FAILURE in 1h 02m 19s (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/bf60c98babf14a04a48ae0cc4ceaf701 : SUCCESS in 1h 52m 58s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/9e55b9c06a324e338259ff536be2554a : SUCCESS in 1h 16m 01s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/83548d7c34cf4dbe8b24ebcfda2f8721 : SUCCESS in 50m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bbf2ccb646d24399ac637f42d80bff55 : SUCCESS in 3m 30s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/0ceb239cafff4faea4fc2ef05d39bb2e : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":6},{"id":"c46a06afd58647940391136752ffc16c675659d7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-29 09:29:20.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a025ae8c21c74031a5edc854ca9c8a38 : SUCCESS in 2h 25m 56s","accounts_in_message":[],"_revision_number":6},{"id":"a1aa3389492d1bcfed986c85b04c73de3d5612b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-29 23:56:20.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2a32e331d2ec8651231637988cdaedb58f1eb372","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-30 00:00:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"e8ca257ac180b5df4922df3aa76daf562609841c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-30 02:44:49.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/7a1a5ff1b41342239e72bdd4311142d9 : SUCCESS in 2h 31m 35s","accounts_in_message":[],"_revision_number":8},{"id":"4a557e8aa597713630e2551e51b83f240a94fd21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-30 03:23:03.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28b1a7a16a3941a8b3e5ec03aa81b765\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aaf2b53ffc954781961e9804358ef262 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/331f7a699507412593a703ea7f47a99e : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f32b5a643744387afa0cebe323b806b : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14426c8803334f4aa9b9b5ee46d02259 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6784ca6470641e983600069ad2c83d2 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9790d098ac5447c6847765ea67964006 : SUCCESS in 3m 26s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/be268eabbd784f7f8affe4f9eb843bb6 : SUCCESS in 3h 13m 45s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/2405ceae341f43efa3d4ccf78e36271d : SUCCESS in 1h 58m 39s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/37ac9c9183fd432882f32cf7eb2ae42b : SUCCESS in 1h 37m 12s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/e20b5bc4426c4a228f089f1f35716d68 : SUCCESS in 39m 48s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/5730b1901be948e4a6ae64c9a3b5fa75 : SUCCESS in 1h 15m 18s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/127cbc82049c453f8261e935e14855ec : SUCCESS in 1h 39m 31s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/bf182f0e89b2414795b20ce43c26b4f6 : SUCCESS in 1h 26m 51s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/07674aec7bbb4584a37f436688781555 : FAILURE in 1h 13m 47s (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/383a0024dfad448482055b665798899c : SUCCESS in 2h 17m 08s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f7f0393b3a8d4926b1a239e4857ef3d1 : SUCCESS in 1h 26m 29s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/be74cfb4408c45aa9bf1f2d8750db9e6 : SUCCESS in 1h 01m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7c95989136f4468ba9823d0ac5892853 : SUCCESS in 3m 28s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/122c34ec715642988bb04ec05c5bb912 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":8},{"id":"06ace7ad9621db2a57575bbdd99d64ec332f32dc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-31 01:36:43.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f3fbe4a84a797cffe88dfe70366ce5a0e2831655","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 01:51:43.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3747a9510cac5235992842ea73ba6bc0d799d4c9","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 02:01:46.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"af27e2175d331f9c27e2d7143e032976d944f220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 02:09:30.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"31123125866e523a8c60c195a20de0edae9c10e3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-31 02:15:29.000000000","message":"Patch Set 10: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"6a2e76bc9d1a636b6815aea95111aca46833fee0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 03:54:00.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"abc1b2e1f049e7cb0585d67d4564e3ac61bf681d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 04:09:07.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6d563bdcb0f92e1b50e05cf97ce6ef81c64ec3e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 04:11:19.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"8e4d31092f77df98f2cc234f00ba91cd84a747d3","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 04:11:46.000000000","message":"Patch Set 13:\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"3cfe4e0695c6e0abe6ac6baf3716a56d29fc874e","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 04:16:25.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"7e65586ced000cfa0eebe1c3b0d32728686ddc88","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 04:38:06.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d50cec41aa12950dfcab57cec8df834e0b929f9e","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-10-31 04:55:53.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"27e77593c2f5ed6ae3b1efc7d1f7761aae9758aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 05:29:31.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"5237b2d86df917983fba3ae9a582aeba8141a607","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 05:30:06.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"c4ad309494f7634f7b2e6dc3519e4b8176b3dfbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 05:32:48.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"52c32dccdbd20d1022184f494e65af2169881bf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 07:04:35.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"f6639ddadc7e6907b163e718fbda1b0f0eb37d17","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-10-31 09:08:22.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d001b9f714234f3a9b3b2af7eecafcd0 : SUCCESS in 2h 02m 39s","accounts_in_message":[],"_revision_number":16},{"id":"4cb78eb0463a9c9e60bf53c9e266fb06007d498b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 09:25:33.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bfbbde435104234836855621dd371b5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c4b984bde3c41ac9e76323807ec5def : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bae64595aea4459dbab488f53cde6e3f : SUCCESS in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0217047b23d40b29851fc1aaa01c235 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0eb12f40dc0443d887fc20a42039d5de : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/384cd2add32d46e0a2dd14dd27cfdb59 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046f4fe9b279473483688da177640e25 : SUCCESS in 2m 53s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/da9cf55ea3d64d708e7f45cddca2c089 : SUCCESS in 2h 12m 22s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/919d4ef7c30a4d40a2adac36773ee339 : SUCCESS in 1h 16m 02s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b107a23f7acd475ea9251eaaf6da1184 : SUCCESS in 1h 13m 26s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/eb761cb61f67423496296c4eaed68216 : SUCCESS in 31m 47s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/a8a79b8002c84825af9fa42f23b37933 : SUCCESS in 1h 19m 44s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/e7be6d4c68ea4aac9b0245f244c687aa : SUCCESS in 1h 15m 34s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/32b2ba5512de472eacca24cb2ad4f934 : SUCCESS in 1h 03m 02s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/2ba6ae600ac54b13a4086ea96ac944eb : FAILURE in 59m 17s (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/c0659847a3c3496cb96ce596b062e227 : SUCCESS in 1h 55m 40s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/7ee3fccaccb14341bc8bd29059754717 : SUCCESS in 1h 07m 29s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/e1aeb4330c3d4ae5aebcb5ed7d30f880 : SUCCESS in 47m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7cf03b8ebedb4f1ca0dd5fa7b29bcd08 : SUCCESS in 3m 40s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/08bc99d3273d44edbee04fe2dc5d3b42 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":16},{"id":"d133120167c54d18c509c1a47e005b962180119f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-10-31 15:42:49.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"c92cc9823a17e018c862c680be082f012efe4d23","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 23:10:04.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"f454a35a9ed4f3ef7eb9ee1090dc56663c149d9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 23:42:30.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"8e0bccc8482f2ce6f4ab23e4a571f56cfc35b465","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-10-31 23:43:08.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"6a015ca25090a7a1098c54c0b6d919de2d94c0e4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-11-01 01:48:46.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/39065fe23b094b9096b32ce1f88c705d : SUCCESS in 2h 05m 10s","accounts_in_message":[],"_revision_number":17},{"id":"2fbd7e3e061425821cc7a76bf45462c9225d8a32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 03:31:35.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07968170c167470b8da04c823bd3a020\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1058c4e8c68c4e39a1b4fcf008de8a1d : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd853709607649e19fd5a1b76f73893c : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27752d445ecf4b20a025d679e6a89308 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffeb758c078748358d34ce57bdaa6ec7 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/179cf2ad5d2642f8acea920eb4d1cef1 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b81e17af5f245adb81d2fbea02d1530 : SUCCESS in 3m 17s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/d2e2aa9352d244b7bb2541f8f8cadfc6 : SUCCESS in 3h 39m 18s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/f960fe6b45f84f6997e6d7fb67687b79 : SUCCESS in 1h 52m 53s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/b9fbbfdf2f3641b2849a39fe496db5d0 : SUCCESS in 1h 49m 03s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/c809a488825146b2ad60d066280cd706 : SUCCESS in 49m 10s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/261ad8d0646f42f09cd4aace346d58f4 : SUCCESS in 1h 25m 54s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/4e6bcccd8e384ac187c56f99cd58e616 : SUCCESS in 1h 52m 25s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/ebead5a50bab4e02a95512f231d3ca02 : SUCCESS in 1h 34m 43s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/df6d97fc92da4a6fa7ba267a324826e7 : FAILURE in 1h 11m 33s (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/951fd614f8cf4e77bb66bcee00b471a0 : SUCCESS in 3h 14m 43s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/f734bca9ded24ea8b83a79d095087037 : SUCCESS in 1h 53m 29s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/1809d822a06c4d8c975f946e205ea90c : SUCCESS in 1h 25m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1f1d97b7e385494ca51fa9374c152f48 : SUCCESS in 3m 38s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/98778d8ac6194a2ba91495f4bcb0f872 : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":17},{"id":"2a8b75ff6ec119e492cdde9d96e54a98b87dd15a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-01 04:03:06.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"c962ef831f0d0b0574c73f99d1fda2b38b967215","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-11-01 05:50:44.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"b181e7d5f6149be7a0413105cd152321595a5153","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-11-01 05:53:50.000000000","message":"Patch Set 19: Commit message was updated.","accounts_in_message":[],"_revision_number":19},{"id":"45713d59d7a87b7c03982c2ae2e159db7251195d","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-11-01 05:54:01.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"4302db1f35ab3f473651414677ad0fd5104ea0f9","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2022-11-11 05:21:32.000000000","message":"Patch Set 19: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"a67e85908c53e2b82d3bbdc3dc2885967c5a1df0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-11-11 07:45:18.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/9d59904518674ca78d13631837a93bad : SUCCESS in 2h 17m 26s","accounts_in_message":[],"_revision_number":19},{"id":"15dfc8b20c88c1a012b64bcf97076c2db67bf82a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-11 08:03:48.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/638db1ee55a0468c8c6e29325b9c581b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/680f9b65ff97495f91ffda8c6d8178e4 : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0b18a3d66e74e1d8d01eaec160681e2 : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbd9f2bbf4df494394a0aaf5009d1757 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e1511c6370c4a26acc00b654b5253b1 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8529d732b8cc4f8e8fc84139df986120 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81bfab9674974a37926ee12a0c9bc32b : SUCCESS in 3m 01s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/c7c5d931f34e41b7830e44fc9c0b351f : SUCCESS in 2h 37m 21s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/b22e1275c678433d87014b9082c079b4 : SUCCESS in 1h 11m 38s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c0efde306d6e4b95b1fd47f902748276 : SUCCESS in 1h 16m 07s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/237bcaa7cc0f4647adfd2e0bc0c1e6ef : SUCCESS in 32m 50s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/dcec10c37db64619a5dcdee6eff5fc8a : SUCCESS in 1h 19m 57s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/2f562a7d36f04b1a9ec3af5984c41141 : SUCCESS in 1h 47m 58s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/56f3d01066044ab5955fd2b7af25ce18 : SUCCESS in 1h 19m 03s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/9d4e4732368e40158c63707b04be27e6 : FAILURE in 52m 36s (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/4e89764a2f1641c48981ed3015551481 : SUCCESS in 1h 57m 35s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2a95b744ded44de89119f7332201186c : SUCCESS in 1h 23m 26s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/4b97e34e43ff4ae3904b77d77a53433f : SUCCESS in 1h 00m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/00906d883f19468b926fdd94ccec2e2b : SUCCESS in 3m 09s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/7b5e10b583d64af895e0648f319b5199 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":19},{"id":"f96762cdf85cfd388bfe1c99773bb0359e7a3505","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-14 01:54:33.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":19},{"id":"5531f74deaa31ceb028ae1e0d8ad6e52d1e3fc21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 01:55:10.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"04c45f3f10aaa4b771337a7352a5cc839900c6e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 05:05:35.000000000","message":"Patch Set 19: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2388e218fe30469eaee8a734109a2f30\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d86e6c349d5e496ca65d5cf9a73a9745 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9688e86c4e794c75ace54a35f2030ec6 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/240cb267426f4f3b9c07659d5985e102 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dbbcd578a4a4e9bb456bafb6addfcc5 : SUCCESS in 3m 21s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/b86b5737be914d55956b4d6c38c81d17 : TIMED_OUT in 3h 06m 08s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/2425fe4811aa44ca9e738eb39b587a48 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/626ea58882074a44b10167fe48e3a68b : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/90281889075543e38ded213ad15255a8 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/661a23b22a28436ca95dc5be7cad06da : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/da3c4e2435424336b84a67001ec522c3 : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/03b6be3309424071bac8a46cb9b5933b : TIMED_OUT in 3h 10m 20s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/9a1af64fd9104c719e3a2cddc6d43d4e : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/e714ed0d358440e88a2af4c08204f42c : SKIPPED Skipped due to failed job tripleo-ci-centos-9-content-provider\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f74c86c6932f427ba0541f1a8bf9ba8a : SUCCESS in 3m 30s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/e338ea3a8cf24cdd8755f791a4b5c2d5 : SUCCESS in 3m 40s","accounts_in_message":[],"_revision_number":19},{"id":"6edb468754145b631b653e42bebc8fe6d4945dfc","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-11-15 02:25:09.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"fcc6601f7d9859e99bf8274006ccb92ab10cd119","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-11-15 02:58:14.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c5f86906591b48adbbf74e241a9a59e9 : FAILURE in 32m 10s","accounts_in_message":[],"_revision_number":19},{"id":"2919c75288eb4315cd2e832c70e81cf58c2a7e76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 03:28:55.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc4b2b42e1bb4406b00527fb63077554\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c95e84b9a47849cca2c13b581c5b49a4 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fbf036b200d470ba5568c7f14ab128e : FAILURE in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea6df7e87d3c43a48e5248fca744bced : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/944e7900abb64ee88076aada6a88e18a : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35776089baad40f687bb2ed979cd24dc : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1926bcd4ac39487c86796b460d5306e8 : SUCCESS in 3m 57s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/55638f1f9e0546c191a44f9e7694554d : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/108554260f2b41b1b1edc9075ee4bcc2 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/5dfac118019245368a56c18952007633 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/f2f03be0172d4206b8e12471abcca57b : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/8f237fbb5ad14959b1d53522276d3508 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d56c586fb47b49d6b73454e1a840ef14 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/b195a5abb1eb4a049f282ff391765416 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/7e6e63ea4a9648909af96a1f597dbfaf : SKIPPED Skipped due to failed job openstack-tox-linters (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/4d4955497f874e029b04111d9adb0915 : SUCCESS in 58m 22s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c975bf6ba5bc4302a50b63deb65ce6cd : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/b27d59a7fa1145a3a917219595f90e2b : SKIPPED Skipped due to failed job openstack-tox-linters\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d92f10198d2244e0899e2171b2c53266 : FAILURE in 2m 46s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/a04ef9afb9584134911993d47aacd2b8 : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":19},{"id":"2170b81c5c28bee013a9e99b5e1197ec609427c2","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-11-15 03:32:53.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"67ec75e5f563419a5662ccdac22863f2a465b8be","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-11-15 04:06:47.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a1c3396063694133936306762d341405 : FAILURE in 32m 51s","accounts_in_message":[],"_revision_number":19},{"id":"3030cf2781ad2cf886057ec951f64af55f03527f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 06:46:58.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6c833d23d734122aa1973100c2b6ee6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ca028b9531441c9bce597ec36237c61 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527b081f372b4fa5a7b3be825d3a073f : FAILURE in 2m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21ec60ba4a1748f58039ff22b59fce1f : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a9f478caeb245e78dbf747e63d661cb : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a094d93c988d45e68b325cab7bc2d655 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a3d4317870540dc93b555cc6cc80772 : SUCCESS in 3m 10s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/a2158ff18a1d4fedb34688a9f470ce20 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/3ff68f9a66654664b11abf4e5339c1b0 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/87859fe56df34ca5b63d5f2c577e0985 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/c6bf5f4c121d49ed8faebc1d5ace0616 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/3c2f8a0f48ae42d39d5f55c00b5a5607 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/92cf629e824e41799472ebd73d0bd7cc : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/c0005dbbc45d40c89bf30cc114ebbe22 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/e68792f2b26c4cd8a6650f3451dc972a : SKIPPED Skipped due to failed job openstack-tox-linters (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/6c97eda276dd40d98122d70f12999ed2 : TIMED_OUT in 3h 07m 52s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/c8fb53b8c3c0453a8ffe195bc2d0a7dd : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/2ad305aa33ef443a94475d9c0e88d4f7 : SKIPPED Skipped due to failed job openstack-tox-linters\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4e36226c4ad146019f9ef5d47368135b : FAILURE in 2m 29s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/311e10423d8f46c69b20c92b7cc329d5 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":19},{"id":"95aa5b5942b58da0e7440d563f0bbe7cebdcf3c9","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-11-15 22:48:38.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"83836411ebf5adc6666f53099db1cdfbd7a8a4d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 23:37:08.000000000","message":"Patch Set 19:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/068297eeeaa04b239fd8fcab7fd469aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09bf3e91491044c6b06996999b563e14 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b6c9120e6d54d568bdffaf976d04893 : FAILURE in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a5452f4f48c41fdaa61c78f86eb4a15 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5497dc7f5344bf393ff3fa47df0d28f : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01c0bc5aa5ae45709dc619288b1a87c4 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc52f45009724abfacb8290156cd84fd : SUCCESS in 3m 52s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/e9010b194af442ad82943a6ae6f6d749 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/66966d1ed56e419ea3c486e98f74ebc0 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2b741b37f08d4ac2902239e5fce683f2 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/0e02b997258f432fadd09bccff776e3b : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/ff373cc9adcf4fd78ac3c289eb447f78 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/ce819ab0e89a4293bb1efd594d6e34b2 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/30e866e580c041ccb7599bdc03c9793e : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/9daaabed49664a4483d52f5df7c0cf49 : SKIPPED Skipped due to failed job openstack-tox-linters (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/1a2f195c08b641a494e2e3d0bad84261 : SUCCESS in 46m 32s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/1ddce5f0885544979e52901bfdbb7720 : SKIPPED Skipped due to failed job openstack-tox-linters\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/7bcba33eb6534347ae396710fd772dd8 : SKIPPED Skipped due to failed job openstack-tox-linters\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b58760895abf4209a07caa849b56af12 : FAILURE in 3m 47s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/bddfe7aa7d974d1993be942706ac2b47 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":19},{"id":"305b98511589c232d294718c649de0dd4fcdc552","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-11-16 01:06:08.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/58c48783af064e779e4adc114c363caa : SUCCESS in 2h 16m 27s","accounts_in_message":[],"_revision_number":19},{"id":"93ab0336d178bfdc86e6d2d98a346e493ed63437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2022-11-18 06:28:40.000000000","message":"Patch Set 20: Patch Set 19 was rebased","accounts_in_message":[],"_revision_number":20},{"id":"3d05eaf5c1b706f694ee4eb9c1dbd5b195944ea3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2022-11-18 08:54:45.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-9-ovb-3ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/c3f9e5c0eab34bf595ebce329b342e4e : SUCCESS in 2h 24m 41s","accounts_in_message":[],"_revision_number":20},{"id":"811031ee85c0a211f59e48de782423fd6b70ac5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 09:56:19.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b05441de2b348e79f5e28f77a497164\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2900f5ffe894d9eac411521d90660dc : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f5511acc63e4b7d85237a24f022de9e : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6a1aff714f2457c9b7064a876eb22a2 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13139162b72e4bfd9b29bf00845fe447 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbbee1d728544f6989ef9d0e4043f718 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb3edef44a2646b78d802c86e239b73a : SUCCESS in 4m 21s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/15ac75f5abf245a2bb67aa5607dca26f : SUCCESS in 3h 17m 20s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/6d4d8045d78b430fb0ee412c74427feb : SUCCESS in 1h 47m 19s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/c2e1a83994124e1894d4986916b3f7b7 : SUCCESS in 1h 47m 35s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/43e7f92c897a4fd0abc0f771168cece6 : SUCCESS in 55m 18s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/8408d6ac05244b97803f897431ed715c : SUCCESS in 1h 34m 27s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c8c8c7a2f7b34c969ab8b8c9ca4fe4d1 : SUCCESS in 1h 54m 21s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/adb5b8abbb8845b09f284069e2c596d2 : SUCCESS in 1h 40m 05s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/4d20b20d4216422cbd18daac7017469a : FAILURE in 1h 16m 09s (non-voting)\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/72f998546e60438883a0b755ca667cc5 : SUCCESS in 2h 46m 11s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/00423fa47004493eb93034b89159c9c8 : SUCCESS in 1h 38m 18s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/1ea15b2c7d0046e2b5954f95d4e9c561 : SUCCESS in 1h 13m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ba64a9f86ca54e20b17fbd3a6add1bf7 : SUCCESS in 4m 44s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/fb17152f8f8345469c4180774572f89f : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":20},{"id":"7dec03518f7912ee254bfff60202505b0354190a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-18 10:24:32.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"20d020b87df754ae889a50afb6637ca9e7a2fdee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 10:25:20.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"b13de6e4dad2bb6a845743b212a7d95b93e48718","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 13:48:54.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/633cd7ae8af646a5bbc07afa232193f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713ea3700d6744feb488768202f47e9a : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b187b68c5e494babbb1c83b22ce06b7b : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7bfdb30cc5f64d0097c33eb2a32e265d : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d4819fe82804e82918873cc1d270a4d : SUCCESS in 4m 11s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/11d46380f02f45f8abd73cfa7f219b29 : SUCCESS in 3h 17m 35s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/7ba172b9e39b4da2a5d7135796075dff : SUCCESS in 1h 52m 11s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/128bd4f3bedb4d7bb7980e1ae2a12abd : SUCCESS in 48m 29s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/177bd998137f4821bd6b9bcff72dd4b0 : SUCCESS in 1h 27m 18s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/18911f54be1f4163a40c042302275b22 : SUCCESS in 1h 55m 14s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/05bf62e5d6cc447989cc9d293b1f6d9b : SUCCESS in 1h 43m 32s\n- tripleo-ci-centos-9-content-provider-zed https://zuul.opendev.org/t/openstack/build/daf211b840aa4692a705156471211a35 : SUCCESS in 2h 35m 45s\n- tripleo-ci-centos-9-undercloud-upgrade https://zuul.opendev.org/t/openstack/build/c4fabce70a38432daae0de966bd9a7c8 : SUCCESS in 1h 06m 29s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/43a6bcdbf3304951abef6eed33a6d2f1 : SUCCESS in 1h 36m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f166930d4c584f49bbf4911e50100b6f : SUCCESS in 3m 50s\n- tripleo-cross-tripleoclient-py39 https://zuul.opendev.org/t/openstack/build/fb507c0799cb477bb8692ac754ac585a : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":20},{"id":"c3dc466e76eee4e970c0428b85dceb9c9f893b4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 13:48:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":20},{"id":"16971d40cf7cdf133c5b0bdfb11bbdc03d524740","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-18 13:49:53.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/771d6a3e7bd64a66a05ef6a99d8120f7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbe580cf7b8f4f74897168d58563cc51 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"6e4e430b46cfdce21663fa1f2417ae925aede97e","revisions":{"b3dc3faf6adab8bcc2e4ac8a0bb37175a0702a12":{"kind":"REWORK","_number":1,"created":"2022-10-28 12:15:44.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/1"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:15:19.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens have a expires_at field, not a expires_in. This patch\ncorrects the key lookup from the token to resolve cached token\nauthentication issues\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b3dc3faf6adab8bcc2e4ac8a0bb37175a0702a12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b3dc3faf6adab8bcc2e4ac8a0bb37175a0702a12"}]},"branch":"refs/heads/master"},"68ecfbc4e28f37aa649d337b4885edb1785bb187":{"kind":"REWORK","_number":2,"created":"2022-10-29 06:38:23.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/2"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-29 06:38:13.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens have a expires_at field, not a expires_in. This patch\ncorrects the key lookup from the token to resolve cached token\nauthentication issues\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/68ecfbc4e28f37aa649d337b4885edb1785bb187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/68ecfbc4e28f37aa649d337b4885edb1785bb187"}]},"branch":"refs/heads/master"},"f4763bffbcff136c30f3f5cc221b4acbb18218be":{"kind":"REWORK","_number":3,"created":"2022-10-29 06:41:51.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/3"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-29 06:41:42.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens have a expires_at field, not a expires_in. This patch\ncorrects the key lookup from the token to resolve cached token\nauthentication issues\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f4763bffbcff136c30f3f5cc221b4acbb18218be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/f4763bffbcff136c30f3f5cc221b4acbb18218be"}]},"branch":"refs/heads/master"},"497b6f9269cd0ff31c9cc555126771f41f419843":{"kind":"REWORK","_number":4,"created":"2022-10-29 06:44:24.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/4"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-29 06:44:14.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens have a expires_at field, not a expires_in. This patch\ncorrects the key lookup from the token to resolve cached token\nauthentication issues\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/497b6f9269cd0ff31c9cc555126771f41f419843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/497b6f9269cd0ff31c9cc555126771f41f419843"}]},"branch":"refs/heads/master"},"c49bbef7b6305f12fdb4229b5cd736555cff34aa":{"kind":"REWORK","_number":5,"created":"2022-10-29 06:52:12.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/5"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-29 06:52:01.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens have a expires_at field, not a expires_in. This patch\ncorrects the key lookup from the token to resolve cached token\nauthentication issues\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c49bbef7b6305f12fdb4229b5cd736555cff34aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c49bbef7b6305f12fdb4229b5cd736555cff34aa"}]},"branch":"refs/heads/master"},"c33813e383d624688c8c10faf3c1051ce58ccf11":{"kind":"REWORK","_number":6,"created":"2022-10-29 07:02:15.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/6"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-29 07:02:02.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens have a expires_at field, not a expires_in. This patch\ncorrects the key lookup from the token to resolve cached token\nauthentication issues\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c33813e383d624688c8c10faf3c1051ce58ccf11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c33813e383d624688c8c10faf3c1051ce58ccf11"}]},"branch":"refs/heads/master"},"d6de1c77ca4d37ea0d803d389c660dcf6f327d57":{"kind":"REWORK","_number":7,"created":"2022-10-29 23:56:20.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/7"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-29 23:55:56.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens have a expires_at field, not a expires_in. This patch\ncorrects the key lookup from the token to resolve cached token\nauthentication issues\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d6de1c77ca4d37ea0d803d389c660dcf6f327d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d6de1c77ca4d37ea0d803d389c660dcf6f327d57"}]},"branch":"refs/heads/master"},"02742af7ea4b4bb717f530299dfd37ab9b4d1722":{"kind":"REWORK","_number":8,"created":"2022-10-30 00:00:44.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/8"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-30 00:00:35.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens have a expires_at field, not a expires_in. This patch\ncorrects the key lookup from the token to resolve cached token\nauthentication issues\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/02742af7ea4b4bb717f530299dfd37ab9b4d1722"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/02742af7ea4b4bb717f530299dfd37ab9b4d1722"}]},"branch":"refs/heads/master"},"17e6142a5f54f85716a7a0d8171111a92d3f8653":{"kind":"REWORK","_number":9,"created":"2022-10-31 01:51:43.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/9"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-31 01:51:33.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens have a expires_at field, not a expires_in. This patch\ncorrects the key lookup from the token to resolve cached token\nauthentication issues\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/17e6142a5f54f85716a7a0d8171111a92d3f8653"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/17e6142a5f54f85716a7a0d8171111a92d3f8653"}]},"branch":"refs/heads/master"},"e13b0b734e257d678898f8ac63291a0bd2d82023":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2022-10-31 02:09:30.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/10"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-31 02:08:37.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens can have a expires_at field, and not a expires_in. This patch\ncorrects the key lookup to allow for both fields from the token to resolve\ncached token authentication issues.\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e13b0b734e257d678898f8ac63291a0bd2d82023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e13b0b734e257d678898f8ac63291a0bd2d82023"}]},"branch":"refs/heads/master"},"39d5216cafd92cc2c5ff6b101925623ec2b82ae6":{"kind":"REWORK","_number":11,"created":"2022-10-31 03:54:00.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/11"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-31 03:53:48.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens can have a expires_at field, and not a expires_in. This patch\ncorrects the key lookup to allow for both fields from the token to resolve\ncached token authentication issues.\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/39d5216cafd92cc2c5ff6b101925623ec2b82ae6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/39d5216cafd92cc2c5ff6b101925623ec2b82ae6"}]},"branch":"refs/heads/master"},"9b02f1b2da24dfd99821e0ea2d7c5102014ccb20":{"kind":"REWORK","_number":12,"created":"2022-10-31 04:09:07.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/12"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-31 04:08:57.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens can have a expires_at field, and not a expires_in. This patch\ncorrects the key lookup to allow for both fields from the token to resolve\ncached token authentication issues.\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9b02f1b2da24dfd99821e0ea2d7c5102014ccb20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/9b02f1b2da24dfd99821e0ea2d7c5102014ccb20"}]},"branch":"refs/heads/master"},"317e791e6123d27b274a9772d4738ab6eb79caae":{"kind":"REWORK","_number":13,"created":"2022-10-31 04:11:19.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/13"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-31 04:11:09.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens can have a expires_at field, and not a expires_in. This patch\ncorrects the key lookup to allow for both fields from the token to resolve\ncached token authentication issues.\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/317e791e6123d27b274a9772d4738ab6eb79caae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/317e791e6123d27b274a9772d4738ab6eb79caae"}]},"branch":"refs/heads/master"},"5402def585ec556f5e688b522a96949ded522ebd":{"kind":"REWORK","_number":14,"created":"2022-10-31 05:29:31.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/14"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-31 05:29:21.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens can have a expires_at field, and not a expires_in. This patch\ncorrects the key lookup to allow for both fields from the token to resolve\ncached token authentication issues.\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5402def585ec556f5e688b522a96949ded522ebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/5402def585ec556f5e688b522a96949ded522ebd"}]},"branch":"refs/heads/master"},"d326f0fe930c2e644321b6776405b415f05d1119":{"kind":"REWORK","_number":15,"created":"2022-10-31 05:32:48.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/15"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-31 05:32:38.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens can have a expires_at field, and not a expires_in. This patch\ncorrects the key lookup to allow for both fields from the token to resolve\ncached token authentication issues.\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d326f0fe930c2e644321b6776405b415f05d1119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/d326f0fe930c2e644321b6776405b415f05d1119"}]},"branch":"refs/heads/master"},"00366aa4d04fe61bd181a22c6666d30a072f72dd":{"kind":"REWORK","_number":16,"created":"2022-10-31 07:04:35.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/16"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-31 07:04:24.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens can have a expires_at field, and not a expires_in. This patch\ncorrects the key lookup to allow for both fields from the token to resolve\ncached token authentication issues.\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/00366aa4d04fe61bd181a22c6666d30a072f72dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/00366aa4d04fe61bd181a22c6666d30a072f72dd"}]},"branch":"refs/heads/master"},"6c290ff70a46e79cb66a6ad4cafc345cce741377":{"kind":"REWORK","_number":17,"created":"2022-10-31 23:42:30.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/17"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-31 23:42:11.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens can have a expires_at field, and not a expires_in. This patch\ncorrects the key lookup to allow for both fields from the token to resolve\ncached token authentication issues.\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6c290ff70a46e79cb66a6ad4cafc345cce741377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6c290ff70a46e79cb66a6ad4cafc345cce741377"}]},"branch":"refs/heads/master"},"e4dbe9e2466e25bfe3cfd89572f4f81e584c2247":{"kind":"REWORK","_number":18,"created":"2022-11-01 05:50:44.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/18"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-11-01 05:50:32.000000000","tz":600},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens can have a expires_at field, and not a expires_in. This patch\ncorrects the key lookup to allow for both fields from the token to resolve\ncached token authentication issues.\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e4dbe9e2466e25bfe3cfd89572f4f81e584c2247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e4dbe9e2466e25bfe3cfd89572f4f81e584c2247"}]},"branch":"refs/heads/master"},"08e890ee61869b32e1604b86cb466e7d5596ee8e":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2022-11-01 05:53:50.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/19"}}},"commit":{"parents":[{"commit":"c0f8e094a52e0573c3f95e39f873f15c2fc9c371","subject":"Merge \"tripleo-mount-image: replace most lsblk calls with blkid\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/c0f8e094a52e0573c3f95e39f873f15c2fc9c371"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-11-01 05:53:50.000000000","tz":0},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens can have a expires_at field (specifically, we\u0027re seeing this issue\nwith Satellite). The token does not an expires_in. This patch allows for\nthe lookup lookup of either key from the token to resolve\ncached token authentication issues.\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/08e890ee61869b32e1604b86cb466e7d5596ee8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/08e890ee61869b32e1604b86cb466e7d5596ee8e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6e4e430b46cfdce21663fa1f2417ae925aede97e":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2022-11-18 06:28:40.000000000","uploader":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"ref":"refs/changes/87/862887/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/87/862887/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/87/862887/20"}}},"commit":{"parents":[{"commit":"0cb5354521ff5786824c431766e92da6f2ff1c5a","subject":"Pull flake8 from Github","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/0cb5354521ff5786824c431766e92da6f2ff1c5a"}]}],"author":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-10-28 12:11:55.000000000","tz":600},"committer":{"name":"Brendan Shephard","email":"bshephar@redhat.com","date":"2022-11-18 06:28:40.000000000","tz":0},"subject":"Fix token cache expires_at check","message":"Fix token cache expires_at check\n\nTokens can have a expires_at field (specifically, we\u0027re seeing this issue\nwith Satellite). The token does not an expires_in. This patch allows for\nthe lookup lookup of either key from the token to resolve\ncached token authentication issues.\n\nExample token:\n```\n{\u0027token\u0027: \u0027$2a$09$1b6453892473a467d0737uJ56J02CLEXEkOaX0pdLWgLMlg.qWDSi\u0027,\n \u0027expires_at\u0027: \u00272022-10-28T04:54:58.296Z\u0027,\n \u0027issued_at\u0027: \u00272022-10-17T23:45:09.306Z\u0027}\n```\n\nIt\u0027s unnecessary to get the local time, since tokens have Zulu time. So just\ndatetime.now() is sufficient. But to perform the comparison, we need to convert\nthe string datetime from the return value into a valid datetime object.\n\nThis change corrects the describe behavior.\n\nResolves: rhbz#2134075\nChange-Id: Ic96679548a553960f1242850e42f9eec6e4a45f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6e4e430b46cfdce21663fa1f2417ae925aede97e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/6e4e430b46cfdce21663fa1f2417ae925aede97e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[]}
