)]}'
{"id":"openstack%2Ftripleo-common~892548","triplet_id":"openstack%2Ftripleo-common~stable%2Fwallaby~Idbcfbf29626d9a66427b78067813915caf0e9cd5","project":"openstack/tripleo-common","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2023-09-04 07:40:43.000000000","reason":"Change was submitted"},"6681":{"account":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"last_update":"2023-09-04 07:40:43.000000000","reason":"Change was submitted"},"36316":{"account":{"_account_id":36316,"name":"Alexon Ferreira de Oliveira","display_name":"Alexon Oliveira","email":"alolivei@redhat.com","username":"alexonoliveira"},"last_update":"2023-09-04 07:40:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idbcfbf29626d9a66427b78067813915caf0e9cd5","subject":"Changed the DEFAULT_COMPUTE_API_VERSION value in the constants.py","status":"MERGED","created":"2023-08-23 15:12:40.000000000","updated":"2023-09-04 17:28:05.000000000","submitted":"2023-09-04 07:40:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892548","meta_rev_id":"b109472e22572bd53e668de0bced91fbeeb0c908","_number":892548,"virtual_id_number":892548,"owner":{"_account_id":36316,"name":"Alexon Ferreira de Oliveira","display_name":"Alexon Oliveira","email":"alolivei@redhat.com","username":"alexonoliveira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-04 07:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2023-08-23 17:32:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":2,"date":"2023-08-23 16:47:02.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2023-09-04 02:49:18.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"}],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 16:15:33.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"},{"updated":"2023-08-23 16:47:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2023-08-23 17:32:12.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2023-08-23 19:35:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"791d8dd6ecf77da97626a186e4b47afcfd45de07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36316,"name":"Alexon Ferreira de Oliveira","display_name":"Alexon Oliveira","email":"alolivei@redhat.com","username":"alexonoliveira"},"date":"2023-08-23 15:12:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"123cc41cd4170abd07fc90f1241c7bd4c00cc395","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-08-23 16:47:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa286cb43e16d88acaf282dec59e2e4dc8f23018","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-08-23 17:32:12.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f6e16660072679720c9452b119de96e2daba2b1","author":{"_account_id":36316,"name":"Alexon Ferreira de Oliveira","display_name":"Alexon Oliveira","email":"alolivei@redhat.com","username":"alexonoliveira"},"date":"2023-08-23 17:45:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2fc0e963d7602c8b97acd5904482215b2b168db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 19:35:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e2d7b8aae494236a9871139253e3020\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d10a5c5a0715477a9fc41fe79ac457fe : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699cbb3b952b4636b3ef18cfd72f96ae : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/775dbcecc69f4c749098449f31c037da : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4409bf6803e743d1bad04cddb320b96f : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af4d5173644b42aebb0cad2c896e6caa : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27ff770688d54d07936c8fb479c90e11 : SUCCESS in 3m 30s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/03c98ab3ffdb49ddbe8396a3cfe73241 : SUCCESS in 3h 43m 35s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/402c75bdcf8449948385c5d20306e67c : SUCCESS in 4h 17m 06s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/96ebf6b61540416ca7d25df8741aa80b : SUCCESS in 1h 51m 10s\n- tripleo-ci-centos-8-9-multinode-mixed-os https://zuul.opendev.org/t/openstack/build/791f82a1739a4546a7e743b0c8d0596b : SUCCESS in 2h 09m 16s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/2c238aaf2b7a45418b394750c040e076 : SUCCESS in 1h 45m 10s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/b5e621509f5540edafe2bbd68f9006f7 : SUCCESS in 1h 06m 10s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/6b5764b0c5db40179fdc36bcb12b4bab : SUCCESS in 1h 36m 33s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/fd66d2c42bea44cc87e36540c073a632 : SUCCESS in 1h 27m 32s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/3c3cf6de08e4484f86d9155581a5deec : SUCCESS in 1h 55m 56s (non-voting)\n- tripleo-ci-centos-9-content-provider-current https://zuul.opendev.org/t/openstack/build/a3397773fef84a46b755073ce9f9aa0d : SUCCESS in 3h 38m 46s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/5ac5a0fd21d34468a2fde118df744c0d : SUCCESS in 2h 06m 13s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/42f121174d874e1d8496b190930cb952 : SUCCESS in 4h 18m 33s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/84bbbeb170a64ebd9bcba7810318e06d : SUCCESS in 2h 38m 33s\n- tripleo-ci-centos-8-standalone-ffu-wallaby https://zuul.opendev.org/t/openstack/build/f746a9e57c9948f2b7a13278004a4b6f : SUCCESS in 2h 30m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8fa46db3e11d4335bc66d4286e9d7576 : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/51f0e248a28044a48a7eabfb04de3a4b : SUCCESS in 4m 18s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/0d51716cc8a547f2be6eec1e36de79ec : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0ebe741ec3c398a95ad7ceb5f59d347ed874e186","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-09-04 02:49:18.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"464d5a154531c9247fe167617895a877e70ad783","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 02:50:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4755b424ced396285d557068ad41acddca967fc1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 07:40:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fe946d6709847228c88671f476bc7f1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd18f5caa03c44bd8e4a26d080fe4cab : SUCCESS in 3m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/123f23cbeef5455ea94c744fc067391f : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f96131fe6d6a4f40afa2ac62b2a660dd : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8169e78e0900468e97b42e3b8824ae59 : SUCCESS in 3m 30s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/1df0a3aed64e41c18748b07218be2ecf : SUCCESS in 3h 40m 22s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3f0dec9ed5b74b4d91f0a599b42278e7 : SUCCESS in 4h 41m 54s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/d3ce681a46384edb8436033f507caa8e : SUCCESS in 1h 28m 09s\n- tripleo-ci-centos-8-9-multinode-mixed-os https://zuul.opendev.org/t/openstack/build/2340848ff17b4e01b359a99c05162899 : SUCCESS in 1h 42m 16s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/b2cb1ff2413e4fe9b1c628ce6da34df9 : SUCCESS in 53m 56s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/bd332fcb19f749a2bc272eeb1cdc01a2 : SUCCESS in 1h 38m 18s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/6fc792ac2e0542518179aca1373fe4ec : SUCCESS in 1h 13m 01s\n- tripleo-ci-centos-9-content-provider-current https://zuul.opendev.org/t/openstack/build/d977c08b04604373b243c7183748f276 : SUCCESS in 3h 35m 28s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/f4e009b938664549a8677b940850ca95 : SUCCESS in 4h 45m 53s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/ed28af9453d449e6b8d59c6b27517e6c : SUCCESS in 2h 13m 39s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/ebeea1ef5e9542a7bd314634dfe25af8 : SUCCESS in 2h 50m 04s\n- tripleo-ci-centos-8-standalone-ffu-wallaby https://zuul.opendev.org/t/openstack/build/5b0b533bbac34dcc82e12ece5453925f : SUCCESS in 2h 50m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1e4aa1057cf84b2ebf34674ccbe445f5 : SUCCESS in 3m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/785f4a842b774958ad0a667049927b97 : SUCCESS in 4m 18s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/422bb1f6aa484902a0a18d1ef9e0062d : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2629362d195940e41621b2041ada132060dd7585","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 07:40:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5269f286dd9bd9be0eb00c99a6a3128519c0eeac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 07:41:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfcae908e2764cf190333f351f4f4289\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dec4fcd514aa4df6b6e7e7d8200b9fce : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1},{"id":"b109472e22572bd53e668de0bced91fbeeb0c908","author":{"_account_id":36316,"name":"Alexon Ferreira de Oliveira","display_name":"Alexon Oliveira","email":"alolivei@redhat.com","username":"alexonoliveira"},"date":"2023-09-04 17:28:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df8edc662feebd6700618a5d68d7c966ecae9256","revisions":{"df8edc662feebd6700618a5d68d7c966ecae9256":{"kind":"REWORK","_number":1,"created":"2023-08-23 15:12:40.000000000","uploader":{"_account_id":36316,"name":"Alexon Ferreira de Oliveira","display_name":"Alexon Oliveira","email":"alolivei@redhat.com","username":"alexonoliveira"},"ref":"refs/changes/48/892548/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/48/892548/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/48/892548/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/48/892548/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/48/892548/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/48/892548/1"}}},"commit":{"parents":[{"commit":"4de2a324761ce79fdd2af4657acc6b07468ad6f2","subject":"Merge \"Handle different format in expires_at field\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/4de2a324761ce79fdd2af4657acc6b07468ad6f2"}]}],"author":{"name":"Alexon Oliveira","email":"alolivei@redhat.com","date":"2023-08-23 15:11:57.000000000","tz":-180},"committer":{"name":"Alexon Oliveira","email":"alolivei@redhat.com","date":"2023-08-23 15:11:57.000000000","tz":-180},"subject":"Changed the DEFAULT_COMPUTE_API_VERSION value in the constants.py","message":"Changed the DEFAULT_COMPUTE_API_VERSION value in the constants.py\n\nThe default value defined for the variable DEFAULT_COMPUTE_API_VERSION\nin the ~/.overcloudrc file generated by overcloudrc.py, which pulls\nthis variable value from constants.py, is \u00272.latest\u0027, but this\ngenerates a warning message everytime a user uses this standard config\nfile, like this one below:\n\n/usr/lib/python3.9/site-packages/openstack/config/cloud_region.py:452:\nUserWarning: You have a configured API_VERSION with \u0027latest\u0027 in it.\nIn the context of openstacksdk this doesn\u0027t make any sense.\n  warnings.warn(\n\nTo avoid that, I changed the value \u00272.latest\u0027 to simply \u00272.1\u0027,\nwhich is the current API version in use after a fresh install.\n\nChange-Id: Idbcfbf29626d9a66427b78067813915caf0e9cd5\nSigned-off-by: Alexon Oliveira \u003calolivei@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/df8edc662feebd6700618a5d68d7c966ecae9256"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/df8edc662feebd6700618a5d68d7c966ecae9256"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
