)]}'
{"id":"openstack%2Ftripleo-common~896319","triplet_id":"openstack%2Ftripleo-common~stable%2Fwallaby~I36336f858eb93447fa01ccaec6b33f7f7430bc9d","project":"openstack/tripleo-common","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"8833":{"account":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"last_update":"2023-09-25 19:28:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I36336f858eb93447fa01ccaec6b33f7f7430bc9d","subject":"Remove preference param for media range","status":"MERGED","created":"2023-09-25 03:28:00.000000000","updated":"2023-09-25 19:29:27.000000000","submitted":"2023-09-25 19:28:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896319","meta_rev_id":"13ae40022abcf0940918d16354f58412bf19131d","_number":896319,"virtual_id_number":896319,"owner":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34419,"name":"Ricardo Diaz Campos","email":"rdiazcam@redhat.com","username":"rdiazcam"},{"value":0,"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-25 19:28:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"recommended":{"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"},"all":[{"value":1,"date":"2023-09-25 12:56:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34419,"name":"Ricardo Diaz Campos","email":"rdiazcam@redhat.com","username":"rdiazcam"},{"value":1,"date":"2023-09-25 13:45:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"},{"value":2,"date":"2023-09-25 14:43:33.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2023-09-25 13:57:17.000000000","_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":34419,"name":"Ricardo Diaz Campos","email":"rdiazcam@redhat.com","username":"rdiazcam"},{"value":0,"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"},{"value":1,"date":"2023-09-25 14:43:33.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},{"_account_id":34419,"name":"Ricardo Diaz Campos","email":"rdiazcam@redhat.com","username":"rdiazcam"},{"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 08:10:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-25 12:56:57.000000000","updated_by":{"_account_id":34419,"name":"Ricardo Diaz Campos","email":"rdiazcam@redhat.com","username":"rdiazcam"},"reviewer":{"_account_id":34419,"name":"Ricardo Diaz Campos","email":"rdiazcam@redhat.com","username":"rdiazcam"},"state":"REVIEWER"},{"updated":"2023-09-25 13:45:34.000000000","updated_by":{"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"},"reviewer":{"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"},"state":"REVIEWER"},{"updated":"2023-09-25 13:57:17.000000000","updated_by":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"reviewer":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"state":"REVIEWER"},{"updated":"2023-09-25 14:43:33.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"}],"messages":[{"id":"49e81f57c4dc6591b7fa68bb3241156396bc7eca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-09-25 03:28:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f74456be3f89432dc93996a8768ddb038829eb8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 08:10:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5fc2d1c75b14591aced0626c4886ace\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9671b3d695aa46bfb06c2c66ea43db71 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0d0f701df6441b99b942ac96c355827 : SUCCESS in 2m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc2928dfbea84a24b9ad85ba6a1a8af8 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/815cc9b5a96645d68fa17c60efadfa7a : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b5ab2c39078486bb32a37015f7c7846 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b04e54618ec2415b91658e253c07e0d6 : SUCCESS in 3m 48s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/7be9f877525b46abb4125af2e3b38a33 : SUCCESS in 3h 17m 26s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8ff8c8262b2d499484c16fc9c3e5dcd0 : SUCCESS in 4h 37m 18s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/2b0ccf08441a4b1489af6c545fc91d0e : SUCCESS in 1h 13m 50s\n- tripleo-ci-centos-8-9-multinode-mixed-os https://zuul.opendev.org/t/openstack/build/3b8818690f7c456593a9ac71a9908083 : RETRY_LIMIT in 10s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6ec51aa589a54d47902c4f37bb27b2e8 : SUCCESS in 1h 10m 12s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/704c382cc271452e9513479a7ee2cb4c : SUCCESS in 43m 42s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/434f6b2222e241cca2980cf431a1dbd4 : SUCCESS in 1h 37m 46s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/1c8478b258844a41a1c8d7ee9bddcfbc : SUCCESS in 59m 34s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/d999362366f543bbbfcca7f4cf84c159 : SUCCESS in 1h 10m 22s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/908ba89318f44e1ba3f51f3047fb7157 : SUCCESS in 1h 08m 18s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/7697e7d07a964b559d2f9a686061572e : SUCCESS in 1h 16m 47s (non-voting)\n- tripleo-ci-centos-9-content-provider-current https://zuul.opendev.org/t/openstack/build/f174208aed764055893abd89ef01e056 : SUCCESS in 3h 23m 16s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/3f878037f03542a1ac484804a8ec2924 : SUCCESS in 2h 03m 38s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/e715720f3fbc48abbe15a05cfa08f64b : SUCCESS in 4h 36m 46s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/a03f227ee16c4bfdbe0702f7692bab10 : SUCCESS in 2h 47m 20s\n- tripleo-ci-centos-8-standalone-ffu-wallaby https://zuul.opendev.org/t/openstack/build/8343aecd082a4312904cf17205aa6338 : SUCCESS in 2h 58m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4c425160f34f4e4b88f366a6838d96ad : SUCCESS in 3m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/72bdd7ccaad2499facadbf115c7ccd63 : SUCCESS in 4m 13s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/42d6ea3ace1644a698b762c0c7f74138 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"bb94774220890c033394136bfdbda6c7601c10a2","author":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"date":"2023-09-25 08:18:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1ccedda89937cca35ce666cd45f580f186b486dd","author":{"_account_id":34419,"name":"Ricardo Diaz Campos","email":"rdiazcam@redhat.com","username":"rdiazcam"},"date":"2023-09-25 12:56:57.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30ad0da2dd591cdd8eb68337e895dc3f5503049d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 13:18:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb09ea13af98487e88f6f28279dd1181\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdb87c98d44c455fa6dc19a1a231f15a : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de32d86e767349d5a89df7802d1c1ae9 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9704d52a3c0e400f9fa798d103ce3344 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6fd8df9c8afa48849982246554faaca4 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08385d68f1c9460f8b9c793284eba388 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3576d5ef9a7d4e2480c80f063516cc84 : SUCCESS in 3m 44s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/e730aa1fa03e42a0971b37ec20cf0095 : SUCCESS in 2h 49m 58s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/69f9c9db05d140f8b17095eac8f07645 : SUCCESS in 4h 58m 21s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/d27668c5f9fd44838da639b040cec756 : SUCCESS in 1h 28m 39s\n- tripleo-ci-centos-8-9-multinode-mixed-os https://zuul.opendev.org/t/openstack/build/4be8b4b3d5e440b3a557cca41b4a1956 : SUCCESS in 1h 41m 41s\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/d6b5468508d44b77a209862ba9a7ffda : SUCCESS in 1h 36m 33s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/706f27f7f84947f3b9fb8b6dcaf55b82 : SUCCESS in 53m 10s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/215d0ae5037c471aab8a95ec0cce1155 : SUCCESS in 1h 18m 14s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/50e78ffc133b4990ae7091047f86597e : SUCCESS in 1h 13m 04s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/cafac712e3f94e6fbf3ef9804b39be6c : SUCCESS in 1h 25m 40s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/42d1418493bd4ddd8a62f0f1732ffd2c : SUCCESS in 1h 18m 11s\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/45e5d7a4097f4b61866bd5a06a50a4cb : SUCCESS in 1h 35m 30s (non-voting)\n- tripleo-ci-centos-9-content-provider-current https://zuul.opendev.org/t/openstack/build/28f10bd5658448bea5e6af98d90c7ee5 : SUCCESS in 2h 53m 00s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/32b6cc7add924a458f41526836ea45d0 : SUCCESS in 1h 45m 56s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/1c6812e59caf481780c842a9561ed9d0 : SUCCESS in 4h 55m 19s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/708e35de1b18416c95f8ad1e8f316b6d : SUCCESS in 3h 00m 40s\n- tripleo-ci-centos-8-standalone-ffu-wallaby https://zuul.opendev.org/t/openstack/build/cef8b6125d5347029b64c5ea5ea6da20 : SUCCESS in 3h 09m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b3211de3ceeb43b2a5cc27cdfc511ebb : SUCCESS in 3m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e1cc13c0e5cb45ec9223b7828a06a631 : SUCCESS in 4m 10s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/44c3dad29ee94406b69faa71c30bec90 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"fcdfd5d4e78d596c7d0870153931113ffc04133b","author":{"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"},"date":"2023-09-25 13:45:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c6e14432fdf575523261e84b48f99e6c53ee4d1a","author":{"_account_id":30073,"name":"Brendan Shephard","email":"bshephar@bne-home.net","username":"bshephar"},"date":"2023-09-25 13:57:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"248ab9e59b2633128a9931eb72600945c44ea706","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2023-09-25 14:43:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f073d8c8a48bdeafc71e50aa5698ad3849e81097","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 14:44:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ba868b623e0380824a2407d54ed97dfe79c175ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 19:28:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e9b55813ede4b6883949c23088c0f33\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e60a1995ae16487181e10080ab5cff2d : SUCCESS in 2m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3b98c5d3229440a5b35798dc211ae36e : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/498a9950d03640bf81ca3c4c25f2bb6e : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657c73ec2bb1426e8ed4ebaadc309aff : SUCCESS in 3m 28s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/573c4315fd3249958d4b28efd3e8934b : SUCCESS in 4h 00m 34s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/419183c5c4f04a3fbf8c441bb7f62934 : SUCCESS in 4h 39m 54s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/fdb6d5d5e2e948669c35051e2e6d967e : SUCCESS in 1h 44m 39s\n- tripleo-ci-centos-8-9-multinode-mixed-os https://zuul.opendev.org/t/openstack/build/6e38df31be5f453aa1479cca0fd0389b : SUCCESS in 2h 13m 29s\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/92983d81031d4ad38e1e9dd68633ada0 : SUCCESS in 1h 16m 02s\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/16ebff302e0840fc90b3cbe900e1020a : SUCCESS in 1h 41m 30s\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/09ee7cdfff014ac188f1c61261fbcdad : SUCCESS in 1h 32m 42s\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/da2ef161b0a04fa0a42ef1e358f6f111 : SUCCESS in 2h 12m 15s\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/1713380463324c3fa3388c887aed210b : SUCCESS in 1h 46m 07s\n- tripleo-ci-centos-9-content-provider-current https://zuul.opendev.org/t/openstack/build/d4ee8bb64bed45d5be6bffe1ee3ed22b : SUCCESS in 2h 47m 33s\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/9c27d209046e464fa95e5dd1433de34c : SUCCESS in 4h 42m 39s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/111dda34483a446190d46038eb465ae9 : SUCCESS in 1h 33m 29s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/f74186bcb5674227ac80c257b6503d5c : SUCCESS in 2h 50m 12s\n- tripleo-ci-centos-8-standalone-ffu-wallaby https://zuul.opendev.org/t/openstack/build/e862a02a7e06443491727fe7b58350c8 : SUCCESS in 3h 00m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a15c2a53515647d0a211ca354dc7cab4 : SUCCESS in 4m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dcf1a311a576455f85ca9b1c36520c5e : SUCCESS in 3m 50s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/9fc1fe20c5484330afa7e553f281d3c2 : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ac3708ed310be38b0ac975e6ab76073850d8d102","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 19:28:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"13ae40022abcf0940918d16354f58412bf19131d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 19:29:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a06a1dcc06e14ec3a2c3b4c87aa5e9b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c9570b9033745a3ba5e401d0f55a1bf : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5b18f28399d0eb89bdfcd41424e2fca510d857d","revisions":{"e5b18f28399d0eb89bdfcd41424e2fca510d857d":{"kind":"REWORK","_number":1,"created":"2023-09-25 03:28:00.000000000","uploader":{"_account_id":8833,"name":"Rabi Mishra","email":"ramishra@redhat.com","username":"rabi"},"ref":"refs/changes/19/896319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/19/896319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/19/896319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/19/896319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/19/896319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/19/896319/1"}}},"commit":{"parents":[{"commit":"935dad181a926b1bf3c747d6661bc01d56c7eac9","subject":"Ironic PXE: Use the specific healthcheck file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/935dad181a926b1bf3c747d6661bc01d56c7eac9"}]}],"author":{"name":"rabi","email":"ramishra@redhat.com","date":"2023-09-25 03:16:45.000000000","tz":330},"committer":{"name":"rabi","email":"ramishra@redhat.com","date":"2023-09-25 03:27:36.000000000","tz":330},"subject":"Remove preference param for media range","message":"Remove preference param for media range\n\nLooks like most of the registries we interact does not understand\nit. Let\u0027s drop it.\n\nResolves: rhbz#2240163\nChange-Id: I36336f858eb93447fa01ccaec6b33f7f7430bc9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e5b18f28399d0eb89bdfcd41424e2fca510d857d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/e5b18f28399d0eb89bdfcd41424e2fca510d857d"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35491,"name":"Roberto Alfieri","display_name":"Roberto Alfieri","email":"me@rebtoor.com","username":"ralfieri"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
