)]}'
{"id":"openstack%2Ftripleo-common~896374","triplet_id":"openstack%2Ftripleo-common~stable%2Fwallaby~If0dcd9170b36951e4b0d1dc8123844275ad68470","project":"openstack/tripleo-common","branch":"stable/wallaby","topic":"do_not_submit_remove_media_range","hashtags":[],"change_id":"If0dcd9170b36951e4b0d1dc8123844275ad68470","subject":"do not submit this","status":"ABANDONED","created":"2023-09-25 10:17:20.000000000","updated":"2024-02-26 05:29:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"4f5f28c2d3db913559482660df9df0d35f1fb1ea","_number":896374,"virtual_id_number":896374,"owner":{"_account_id":34419,"name":"Ricardo Diaz Campos","email":"rdiazcam@redhat.com","username":"rdiazcam"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-25 12:03:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-09-25 12:03:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fca2bfe964d6e7cb1222f11f06ff7b65f0e664ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34419,"name":"Ricardo Diaz Campos","email":"rdiazcam@redhat.com","username":"rdiazcam"},"date":"2023-09-25 10:17:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bcd6f61303415c4a67ccbfb1aa0fd4318f86bfc","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34419,"name":"Ricardo Diaz Campos","email":"rdiazcam@redhat.com","username":"rdiazcam"},"date":"2023-09-25 10:17:38.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"594892fac1784cc63eed439d7f6804a98160c1c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 12:03:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2662e9e2a794d62bdca2efc473ae8ea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be09051d45fc488993431be3998b1f38 : FAILURE in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6de8c0e99114eb7ba796c362ba669a5 : SUCCESS in 2m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26f296ea68974d57a5ff85e3116680de : FAILURE in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e90e1d31be249ea85090027c30e6596 : FAILURE in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f560085c1b3746e69d9d84b44345e565 : FAILURE in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/006306fb4e734124aef78abf47b43548 : SUCCESS in 3m 26s\n- tripleo-ci-centos-9-content-provider https://zuul.opendev.org/t/openstack/build/4d0d5df678e0410d9f68f827c6280d32 : SKIPPED Skipped due to failed job openstack-tox-py36\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c43fbad4b697439ab88783a6b97eb149 : FAILURE in 41m 38s\n- tripleo-ci-centos-9-containers-multinode https://zuul.opendev.org/t/openstack/build/827c2592cd974b51ac0eb57d49e9ec8f : SKIPPED Skipped due to failed job openstack-tox-py36\n- tripleo-ci-centos-8-9-multinode-mixed-os https://zuul.opendev.org/t/openstack/build/9289d1d8bb9b4c0eaf4e27552cba5416 : SKIPPED Skipped due to failed job openstack-tox-py36\n- tripleo-ci-centos-9-scenario007-multinode-oooq-container https://zuul.opendev.org/t/openstack/build/6d1d96e5180f4e8fbab27968ba0abcbf : SKIPPED Skipped due to failed job openstack-tox-py36\n- tripleo-ci-centos-9-undercloud-containers https://zuul.opendev.org/t/openstack/build/b50d6489c19a4bc6921d34cbe1b9c601 : SKIPPED Skipped due to failed job openstack-tox-py36\n- tripleo-ci-centos-8-standalone https://zuul.opendev.org/t/openstack/build/96f8fd92a5204135a7ec4ca91fcb0fd8 : SKIPPED Skipped due to failed job tripleo-ci-centos-8-content-provider\n- tripleo-ci-centos-9-standalone https://zuul.opendev.org/t/openstack/build/eba7b974e00a4b4d9f7d29e6d379cffc : SKIPPED Skipped due to failed job openstack-tox-py36\n- tripleo-ci-centos-9-scenario001-standalone https://zuul.opendev.org/t/openstack/build/c44a4c701d094b2bb69c1456216ca453 : SKIPPED Skipped due to failed job openstack-tox-py36\n- tripleo-ci-centos-9-scenario004-standalone https://zuul.opendev.org/t/openstack/build/28321ee2031345fb822ece314451dbce : SKIPPED Skipped due to failed job openstack-tox-py36\n- tripleo-ci-centos-9-standalone-fips https://zuul.opendev.org/t/openstack/build/333a634492bb49b8a47ae532851f47d9 : SKIPPED Skipped due to failed job openstack-tox-py36 (non-voting)\n- tripleo-ci-centos-9-content-provider-current https://zuul.opendev.org/t/openstack/build/5c5b5c4e2f4645cc948d6f6dba587f60 : FAILURE in 33m 13s\n- tripleo-ci-centos-9-scenario000-multinode-oooq-container-updates https://zuul.opendev.org/t/openstack/build/2f481da10f7a4e4f8844929bfc182728 : SKIPPED Skipped due to failed job openstack-tox-py36\n- tripleo-ci-centos-8-content-provider-train https://zuul.opendev.org/t/openstack/build/c3b0dc620bbd499ca73b634822999b12 : SUCCESS in 1h 40m 32s\n- tripleo-ci-centos-8-undercloud-ffu-wallaby https://zuul.opendev.org/t/openstack/build/867afccae65340dcb8b2889228ba4784 : SKIPPED Skipped due to failed job tripleo-ci-centos-8-content-provider\n- tripleo-ci-centos-8-standalone-ffu-wallaby https://zuul.opendev.org/t/openstack/build/785fbeb0dbd64343b07949da63cefd6a : SKIPPED Skipped due to failed job tripleo-ci-centos-8-content-provider\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1315a9c1d08a47f0868412b473451f89 : SUCCESS in 3m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/99a7c82ba2c04f0db0bc66e6044a8dae : FAILURE in 5m 03s\n- tripleo-cross-tripleoclient-py37 https://zuul.opendev.org/t/openstack/build/39b8ecefe5d14653ad76dbd990ba7f49 : SUCCESS in 5m 21s","accounts_in_message":[],"_revision_number":1},{"id":"4f5f28c2d3db913559482660df9df0d35f1fb1ea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-26 05:29:30.000000000","message":"Abandoned\n\nTrieplO project is retiring now, for details, please see https://review.opendev.org/c/openstack/governance/+/905145  or reach out to OpenStack TC.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b8eb0eb7926fd0ea65795b3a501e3885100d9bb3","revisions":{"b8eb0eb7926fd0ea65795b3a501e3885100d9bb3":{"kind":"REWORK","_number":1,"created":"2023-09-25 10:17:20.000000000","uploader":{"_account_id":34419,"name":"Ricardo Diaz Campos","email":"rdiazcam@redhat.com","username":"rdiazcam"},"ref":"refs/changes/74/896374/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-common","ref":"refs/changes/74/896374/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/896374/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/896374/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-common refs/changes/74/896374/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-common refs/changes/74/896374/1"}}},"commit":{"parents":[{"commit":"935dad181a926b1bf3c747d6661bc01d56c7eac9","subject":"Ironic PXE: Use the specific healthcheck file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/935dad181a926b1bf3c747d6661bc01d56c7eac9"}]}],"author":{"name":"Ricardo Diaz","email":"rdiazcam@redhat.com","date":"2023-09-25 10:15:41.000000000","tz":120},"committer":{"name":"Ricardo Diaz","email":"rdiazcam@redhat.com","date":"2023-09-25 10:15:41.000000000","tz":120},"subject":"do not submit this","message":"do not submit this\n\nClone of I36336f858eb93447fa01ccaec6b33f7f7430bc9d for testing\npurposes\n\nChange-Id: If0dcd9170b36951e4b0d1dc8123844275ad68470\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b8eb0eb7926fd0ea65795b3a501e3885100d9bb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-common/commit/b8eb0eb7926fd0ea65795b3a501e3885100d9bb3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
