)]}'
{"id":"openstack%2Ftripleo-docs~424577","triplet_id":"openstack%2Ftripleo-docs~master~I52a01302d3a8010dba5ae2fad662c58fe5bbcc56","project":"openstack/tripleo-docs","branch":"master","hashtags":[],"change_id":"I52a01302d3a8010dba5ae2fad662c58fe5bbcc56","subject":"Update manila deployoment steps","status":"MERGED","created":"2017-01-24 10:51:42.000000000","updated":"2017-01-26 20:38:38.000000000","submitted":"2017-01-26 20:38:38.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6197f383521e30a3023e9a881660f129123183fa","_number":424577,"virtual_id_number":424577,"owner":{"_account_id":7582,"name":"Jan Provaznik","email":"jan.provaznik@gmail.com","username":"jprovazn"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2017-01-26 20:38:38.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"},{"value":0,"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2017-01-26 18:51:14.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-01-25 19:30:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"},{"value":1,"date":"2017-01-24 12:01:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":2,"date":"2017-01-26 20:33:12.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"},{"value":0,"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"value":1,"date":"2017-01-26 20:33:12.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-01-24 10:54:52.000000000","updated_by":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"reviewer":{"_account_id":9003,"name":"Tom Barron","email":"tpb@dyncloud.net","username":"tbarron"},"state":"REVIEWER"},{"updated":"2017-01-24 12:01:49.000000000","updated_by":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"reviewer":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"state":"REVIEWER"},{"updated":"2017-01-25 19:30:55.000000000","updated_by":{"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"},"reviewer":{"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"},"state":"REVIEWER"},{"updated":"2017-01-26 18:51:14.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-01-26 20:33:12.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2017-01-26 20:38:38.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"eedfe9304b54cdb22bbd30166354a79b94eebf3d","author":{"_account_id":7582,"name":"Jan Provaznik","email":"jan.provaznik@gmail.com","username":"jprovazn"},"date":"2017-01-24 10:51:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3e4c2ca91d73eb28fbe75f97217c54a394083ab","author":{"_account_id":7582,"name":"Jan Provaznik","email":"jan.provaznik@gmail.com","username":"jprovazn"},"date":"2017-01-24 10:53:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d4c77b5e7c8800db0cb52df8a3949d8a83d56ffc","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2017-01-24 10:57:38.000000000","message":"Patch Set 2:\n\n(1 comment)\n\ncool, I think we need a little amend but thanks :)","accounts_in_message":[],"_revision_number":2},{"id":"1321d0c5d69408fa52faab262b062c837fa5086c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 10:58:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-docs-docs-ubuntu-xenial http://docs-draft.openstack.org/77/424577/2/check/gate-tripleo-docs-docs-ubuntu-xenial/f8c6f51//doc/build/html/ : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":2},{"id":"7608535444b65159ab4139af66fe2c78b6ae884f","author":{"_account_id":7582,"name":"Jan Provaznik","email":"jan.provaznik@gmail.com","username":"jprovazn"},"date":"2017-01-24 11:00:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6b4e855f57d0b21a76f37450b0bdce3f44c133e7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 11:04:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-docs-docs-ubuntu-xenial http://docs-draft.openstack.org/77/424577/3/check/gate-tripleo-docs-docs-ubuntu-xenial/8d4ba17//doc/build/html/ : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":3},{"id":"ba0d9e53011c1d9c016d665b245f2027a8ca8fc0","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2017-01-24 11:11:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ac9d91214f6b240c95d6ad91373d3f509ed91cb6","author":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"date":"2017-01-24 11:46:27.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a2bfbf027d344a67a595dc1b252bd0ba2c689259","author":{"_account_id":7582,"name":"Jan Provaznik","email":"jan.provaznik@gmail.com","username":"jprovazn"},"date":"2017-01-24 11:51:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7df7823a45980a6aac3bca86a7cbd74b670cc602","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-24 11:54:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-docs-docs-ubuntu-xenial http://docs-draft.openstack.org/77/424577/4/check/gate-tripleo-docs-docs-ubuntu-xenial/46c6d78//doc/build/html/ : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":4},{"id":"409d9c4ddc2bce9adc924e66e5216125f8003000","author":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"date":"2017-01-24 12:01:49.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"9d608d7526bfdc48ace1173e111f19011c0d8461","author":{"_account_id":17780,"name":"Dustin Schoenbrun","email":"dschoenb@redhat.com","username":"dschoenb"},"date":"2017-01-25 19:30:55.000000000","message":"Patch Set 4: Code-Review+1\n\nLooks good to me, thanks, Jan!","accounts_in_message":[],"_revision_number":4},{"id":"1f1dbf65677ff4d64e186f4f56d5f49a56253f91","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-01-26 18:51:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"52357224c0ed3b9deadeb5f09b9b13616f896a3f","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2017-01-26 20:33:12.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5188d5c998e48feb490881d037b35426b75bb1ae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 20:33:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"77e175deec8cfc7949fef93150e0e1eeb6e52148","date":"2017-01-26 20:38:38.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":4},{"id":"6197f383521e30a3023e9a881660f129123183fa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-01-26 20:38:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tripleo-docs-docs-ubuntu-xenial http://docs-draft.openstack.org/77/424577/4/gate/gate-tripleo-docs-docs-ubuntu-xenial/d2ccf66//doc/build/html/ : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2ec99d99b34cc6bddcccee66105813f4d11f78f9","revisions":{"1964d04534ef2e24df7bc6f726877da88c7ef250":{"kind":"REWORK","_number":1,"created":"2017-01-24 10:51:42.000000000","uploader":{"_account_id":7582,"name":"Jan Provaznik","email":"jan.provaznik@gmail.com","username":"jprovazn"},"ref":"refs/changes/77/424577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/77/424577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/1"}}},"commit":{"parents":[{"commit":"c8697e4fce5ba4248328f656432a66dc78ca0db0","subject":"Merge \"Virtual deployments are not production-ready\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/c8697e4fce5ba4248328f656432a66dc78ca0db0"}]}],"author":{"name":"Jan Provaznik","email":"jprovazn@redhat.com","date":"2017-01-24 10:50:37.000000000","tz":60},"committer":{"name":"Jan Provaznik","email":"jprovazn@redhat.com","date":"2017-01-24 10:50:37.000000000","tz":60},"subject":"Update manila deployoment steps","message":"Update manila deployoment steps\n\nAdds a section how to deploy manila with ceph deployed by overcloud.\n\nChange-Id: I52a01302d3a8010dba5ae2fad662c58fe5bbcc56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/1964d04534ef2e24df7bc6f726877da88c7ef250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/1964d04534ef2e24df7bc6f726877da88c7ef250"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c8697e4fce5ba4248328f656432a66dc78ca0db0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"7a3f6ec035b89ef395738ef8d93872d95ba92cc9":{"kind":"REWORK","_number":2,"created":"2017-01-24 10:53:23.000000000","uploader":{"_account_id":7582,"name":"Jan Provaznik","email":"jan.provaznik@gmail.com","username":"jprovazn"},"ref":"refs/changes/77/424577/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/77/424577/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/2"}}},"commit":{"parents":[{"commit":"c8697e4fce5ba4248328f656432a66dc78ca0db0","subject":"Merge \"Virtual deployments are not production-ready\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/c8697e4fce5ba4248328f656432a66dc78ca0db0"}]}],"author":{"name":"Jan Provaznik","email":"jprovazn@redhat.com","date":"2017-01-24 10:50:37.000000000","tz":60},"committer":{"name":"Jan Provaznik","email":"jprovazn@redhat.com","date":"2017-01-24 10:53:15.000000000","tz":60},"subject":"Update manila deployoment steps","message":"Update manila deployoment steps\n\nAdds a section how to deploy manila with ceph deployed by overcloud.\n\nChange-Id: I52a01302d3a8010dba5ae2fad662c58fe5bbcc56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/7a3f6ec035b89ef395738ef8d93872d95ba92cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/7a3f6ec035b89ef395738ef8d93872d95ba92cc9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c8697e4fce5ba4248328f656432a66dc78ca0db0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"7ec404b0f14ff37922a32de92b617d7cdd9e269f":{"kind":"REWORK","_number":3,"created":"2017-01-24 11:00:30.000000000","uploader":{"_account_id":7582,"name":"Jan Provaznik","email":"jan.provaznik@gmail.com","username":"jprovazn"},"ref":"refs/changes/77/424577/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/77/424577/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/3"}}},"commit":{"parents":[{"commit":"c8697e4fce5ba4248328f656432a66dc78ca0db0","subject":"Merge \"Virtual deployments are not production-ready\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/c8697e4fce5ba4248328f656432a66dc78ca0db0"}]}],"author":{"name":"Jan Provaznik","email":"jprovazn@redhat.com","date":"2017-01-24 10:50:37.000000000","tz":60},"committer":{"name":"Jan Provaznik","email":"jprovazn@redhat.com","date":"2017-01-24 11:00:15.000000000","tz":60},"subject":"Update manila deployoment steps","message":"Update manila deployoment steps\n\nAdds a section how to deploy manila with ceph deployed by overcloud.\n\nChange-Id: I52a01302d3a8010dba5ae2fad662c58fe5bbcc56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/7ec404b0f14ff37922a32de92b617d7cdd9e269f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/7ec404b0f14ff37922a32de92b617d7cdd9e269f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c8697e4fce5ba4248328f656432a66dc78ca0db0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"2ec99d99b34cc6bddcccee66105813f4d11f78f9":{"kind":"REWORK","_number":4,"created":"2017-01-24 11:51:23.000000000","uploader":{"_account_id":7582,"name":"Jan Provaznik","email":"jan.provaznik@gmail.com","username":"jprovazn"},"ref":"refs/changes/77/424577/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/77/424577/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/77/424577/4"}}},"commit":{"parents":[{"commit":"c8697e4fce5ba4248328f656432a66dc78ca0db0","subject":"Merge \"Virtual deployments are not production-ready\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/c8697e4fce5ba4248328f656432a66dc78ca0db0"}]}],"author":{"name":"Jan Provaznik","email":"jprovazn@redhat.com","date":"2017-01-24 10:50:37.000000000","tz":60},"committer":{"name":"Jan Provaznik","email":"jprovazn@redhat.com","date":"2017-01-24 11:51:15.000000000","tz":60},"subject":"Update manila deployoment steps","message":"Update manila deployoment steps\n\nAdds a section how to deploy manila with ceph deployed by overcloud.\n\nChange-Id: I52a01302d3a8010dba5ae2fad662c58fe5bbcc56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/2ec99d99b34cc6bddcccee66105813f4d11f78f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/2ec99d99b34cc6bddcccee66105813f4d11f78f9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c8697e4fce5ba4248328f656432a66dc78ca0db0","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
