)]}'
{"id":"openstack%2Ftripleo-docs~446997","triplet_id":"openstack%2Ftripleo-docs~master~I2a56e997378fc870a43eb072ff3cd0ea5903e8c5","project":"openstack/tripleo-docs","branch":"master","hashtags":[],"change_id":"I2a56e997378fc870a43eb072ff3cd0ea5903e8c5","subject":"Add warning about undercloud.conf configuration","status":"ABANDONED","created":"2017-03-17 13:30:37.000000000","updated":"2017-05-05 18:25:39.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"892ce68e643cc1089036580687641f87552c8f9c","_number":446997,"virtual_id_number":446997,"owner":{"_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"date":"2017-03-20 19:03:45.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"date":"2017-03-20 11:44:21.000000000","_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"},{"value":1,"date":"2017-03-17 18:09:48.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":-1,"date":"2017-03-20 08:49:32.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},{"_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-03-17 13:50:03.000000000","updated_by":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"reviewer":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"state":"REVIEWER"},{"updated":"2017-03-17 17:42:47.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2017-03-17 17:52:59.000000000","updated_by":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"reviewer":{"_account_id":17888,"name":"Florian Fuchs","email":"f@florianfuchs.com","username":"flfuchs"},"state":"REVIEWER"},{"updated":"2017-03-17 18:09:48.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-03-20 08:49:32.000000000","updated_by":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"reviewer":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"state":"REVIEWER"},{"updated":"2017-03-20 19:03:45.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"}],"messages":[{"id":"7ee12d356131c1f3ee9a5f87fba1036d3e7768fe","author":{"_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"},"date":"2017-03-17 13:30:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34c847de6d33451669b708a99d646ac128210649","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-17 13:32:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-docs-docs-ubuntu-xenial http://docs-draft.openstack.org/97/446997/1/check/gate-tripleo-docs-docs-ubuntu-xenial/0b4ca53//doc/build/html/ : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1},{"id":"c411909eac9090a9bd3812cdfb268563f87f08f3","author":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"date":"2017-03-17 13:50:03.000000000","message":"Patch Set 1:\n\nCan we add validations to help prevent these issues?","accounts_in_message":[],"_revision_number":1},{"id":"0c18b50dc8ddc68b113cb0ba0dcc16e13d46ffa3","author":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"date":"2017-03-17 13:50:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7bc0464db9e5c2aac911bb53486baadabe12fb9c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2017-03-17 17:42:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3885cf11f87622dd554216da75d8dcf3b92b7275","author":{"_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"},"date":"2017-03-17 18:04:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7530469022098465e679c52d43fd4cbbf471d069","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-03-17 18:09:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-docs-docs-ubuntu-xenial http://docs-draft.openstack.org/97/446997/2/check/gate-tripleo-docs-docs-ubuntu-xenial/219f43f//doc/build/html/ : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":2},{"id":"989bc46de83bd48970a46fc7e495e6caede67886","author":{"_account_id":13039,"name":"Martin André","email":"m.andre@redhat.com","username":"mandre"},"date":"2017-03-20 08:49:32.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9187f01a6fa1ce335dfebaa268575cbb3dc96528","author":{"_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"},"date":"2017-03-20 11:44:21.000000000","message":"Patch Set 2:\n\n\u003e (2 comments)\n\n \u003e (2 comments)\n\nlooking at the documentation, TripleO validations needs some work on this subject. It might be a good place to put it, but currently it\u0027s idea of a \"big enough\" dhcp range is 25 nodes with no warnings based on detected number of nodes in the instackenv. Same thing for the introspection ip range or node scheduling settings. \n\nWould you rather I go and put a patch in against validations and maybe add a stronger warning to use it here instead? To be honest I don\u0027t foresee anyone that needs this advice using a GUI installer. The larger customer installations generally come with expectations that preclude a GUI just from a features and configuration required.","accounts_in_message":[],"_revision_number":2},{"id":"9eb654c5a60f990037b2f288e65f6b76f96a8207","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2017-03-20 19:03:45.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nWall of text inline","accounts_in_message":[],"_revision_number":2},{"id":"b79e103a53697a2c5f8476c23efcfce7db3615ff","author":{"_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"},"date":"2017-03-20 19:06:59.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n \u003e \n \u003e Wall of text inline\n\nMaybe we\u0027re barking up the wrong tree entirely? Should I be adding checks in the command workflows to error out on this sort of stuff? The problem in general is that Openstack is hard to debug not because it\u0027s impossible or particularly difficult but because the tools don\u0027t have sanity checks all over the place. I know sanity checks aren\u0027t super pretty to put everywhere but we need to have good ones somewhere. \n\nCould we \u0027validate\u0027 the users instackenv mostly for the purpose of figuring out how many nodes they have and warning them appropriately?","accounts_in_message":[],"_revision_number":2},{"id":"cb2b086daee9dd55167c4a24fa3df904192fe760","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2017-03-20 21:16:33.000000000","message":"Patch Set 2:\n\n\u003e \u003e (1 comment)\n \u003e \u003e\n \u003e \u003e Wall of text inline\n \u003e \n \u003e Maybe we\u0027re barking up the wrong tree entirely? Should I be adding\n \u003e checks in the command workflows to error out on this sort of stuff?\n \u003e The problem in general is that Openstack is hard to debug not\n \u003e because it\u0027s impossible or particularly difficult but because the\n \u003e tools don\u0027t have sanity checks all over the place. I know sanity\n \u003e checks aren\u0027t super pretty to put everywhere but we need to have\n \u003e good ones somewhere.\n \u003e \n \u003e Could we \u0027validate\u0027 the users instackenv mostly for the purpose of\n \u003e figuring out how many nodes they have and warning them\n \u003e appropriately?\n\nUnfortunately I don\u0027t think we can.  I\u0027ve seen plenty of cases where an undercloud is deployed before all the overcloud hardware is available, or where they want to deploy hardware in phases.  In either case, the instackenv they start with is not going to be a complete description of the anticipated environment.  I think we\u0027re better off just asking for an anticipated node count explicitly rather than trying to guess based on other data.\n\nMore sanity checks are good IMHO, but they are tricky to write because a lot of times there are valid use cases that get broken in the process.  For example, it seems obvious that you need to have the overcloud-full image in glance before doing a deployment.  Except we also have the deployed-server method that doesn\u0027t use any glance images.  So how do you validate that?\n\nIn this case an example might be someone who doesn\u0027t use bulk introspection, so they only need a handful of IPs in the inspection range.  If you disallow that then they\u0027re stuck putting more IPs than necessary in the inspection range, or disabling the validations completely and losing that benefit.\n\nSo maybe the answer is to make individual validations toggleable.  Default them all to on, and if a user has a good reason to ignore the failure then they can just turn off that specific validation and keep the others.  The undercloud install doesn\u0027t currently have a way to do that, but it might be a good addition.","accounts_in_message":[],"_revision_number":2},{"id":"6e8a4a166060dd7965ddaa00074b950b266d0235","author":{"_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"},"date":"2017-03-21 11:32:43.000000000","message":"Patch Set 2:\n\nThat seems to be the way to go to me as well. Have lots of detailed warnings and then some way to ignore them. \n\nMaybe a standard setup for sanity checks where it prints a message, links some sort of documentation and has a \"yes this is fine\" option. Then we just provide a --no-warnings option at the top level for people who often perform workflows that trigger sanity checks (or go into a config file and turn off warnings). \n\nI think the problem with validations is that they don\u0027t get run every time you run openstack overcloud deploy or some other baremetal command, we probably don\u0027t need strict sanity checking on mundane vm and stack creation but long running hardware operations will net save time if they spend 30 seconds doing sanity checks before a deploy. Maybe have the workflow call out to validations at runtime so validations can be sure of things like node count and other setup properties?","accounts_in_message":[],"_revision_number":2},{"id":"892ce68e643cc1089036580687641f87552c8f9c","author":{"_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"},"date":"2017-05-05 18:25:39.000000000","message":"Abandoned\n\npursuant to discussion in the comments, this is more of a validations thing.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"17902cbd141d7c6de79916e568196b69dc740e28","revisions":{"4eea0c8ae94121836bbde9e4bb77dbb368cdc656":{"kind":"REWORK","_number":1,"created":"2017-03-17 13:30:37.000000000","uploader":{"_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"},"ref":"refs/changes/97/446997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/97/446997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/446997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/446997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/446997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/97/446997/1"}}},"commit":{"parents":[{"commit":"294fc51c971720aeee6273c93ac694b540b942cd","subject":"Merge \"Add Ansible tasks to container service yaml structure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/294fc51c971720aeee6273c93ac694b540b942cd"}]}],"author":{"name":"jkilpatr","email":"jkilpatr@redhat.com","date":"2017-03-17 13:20:52.000000000","tz":-240},"committer":{"name":"jkilpatr","email":"jkilpatr@redhat.com","date":"2017-03-17 13:30:37.000000000","tz":-240},"subject":"Add warning about undercloud.conf configuration","message":"Add warning about undercloud.conf configuration\n\nRelated bug:\nhttps://bugs.launchpad.net/tripleo/+bug/1673600\n\nThis is essentially trying to call attention to deployment mistakes\nthat tend to be time consuming and difficult to debug. If you fail to\nset scheduler_max_attempts your overcloud deploy will run for hours\nthen fail with error messages that send you off debugging node states\nor hardware itself.\n\nSimilarly users may set a small introspection range to not waste ip\u0027s\nthat won\u0027t be used often and then try to use the bulk introspection workflow\nwhich will fail because they don\u0027t have a introspection range \u003e\u003d their number\nof nodes. Once again debugging this is time consuming and obscure.\n\nI\u0027m open to suggestion as to how we should try to call attention to these\nsorts of gotchas in the documentation.\n\nChange-Id: I2a56e997378fc870a43eb072ff3cd0ea5903e8c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/4eea0c8ae94121836bbde9e4bb77dbb368cdc656"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/4eea0c8ae94121836bbde9e4bb77dbb368cdc656"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"294fc51c971720aeee6273c93ac694b540b942cd","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"17902cbd141d7c6de79916e568196b69dc740e28":{"kind":"REWORK","_number":2,"created":"2017-03-17 18:04:09.000000000","uploader":{"_account_id":22306,"name":"Justin Kilpatrick","email":"jkilpatr@redhat.com","username":"jkilpatr"},"ref":"refs/changes/97/446997/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/97/446997/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/446997/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/446997/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/446997/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/97/446997/2"}}},"commit":{"parents":[{"commit":"294fc51c971720aeee6273c93ac694b540b942cd","subject":"Merge \"Add Ansible tasks to container service yaml structure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/294fc51c971720aeee6273c93ac694b540b942cd"}]}],"author":{"name":"jkilpatr","email":"jkilpatr@redhat.com","date":"2017-03-17 13:20:52.000000000","tz":-240},"committer":{"name":"jkilpatr","email":"jkilpatr@redhat.com","date":"2017-03-17 18:04:05.000000000","tz":-240},"subject":"Add warning about undercloud.conf configuration","message":"Add warning about undercloud.conf configuration\n\nRelated bug:\nhttps://bugs.launchpad.net/tripleo/+bug/1673600\n\nThis is essentially trying to call attention to deployment mistakes\nthat tend to be time consuming and difficult to debug. If you fail to\nset scheduler_max_attempts your overcloud deploy will run for hours\nthen fail with error messages that send you off debugging node states\nor hardware itself.\n\nSimilarly users may set a small introspection range to not waste ip\u0027s\nthat won\u0027t be used often and then try to use the bulk introspection workflow\nwhich will fail because they don\u0027t have a introspection range \u003e\u003d their number\nof nodes. Once again debugging this is time consuming and obscure.\n\nI\u0027m open to suggestion as to how we should try to call attention to these\nsorts of gotchas in the documentation.\n\nChange-Id: I2a56e997378fc870a43eb072ff3cd0ea5903e8c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/17902cbd141d7c6de79916e568196b69dc740e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/17902cbd141d7c6de79916e568196b69dc740e28"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"294fc51c971720aeee6273c93ac694b540b942cd","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
