)]}'
{"id":"openstack%2Ftripleo-docs~452974","triplet_id":"openstack%2Ftripleo-docs~master~I49348b0bdf90a089fd2797d779abbeb135c5d436","project":"openstack/tripleo-docs","branch":"master","topic":"selinux_note","hashtags":[],"change_id":"I49348b0bdf90a089fd2797d779abbeb135c5d436","subject":"Accessing the ramdisk - selinux note","status":"MERGED","created":"2017-04-03 22:12:43.000000000","updated":"2017-04-05 10:24:56.000000000","submitted":"2017-04-05 10:24:56.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"03e27006dc59817bf1b925aed4c8d74b881bf96b","_number":452974,"virtual_id_number":452974,"owner":{"_account_id":22600,"name":"Alexander Chuzhoy","email":"sasha@redhat.com","username":"achuzhoy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-04-05 10:24:55.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},{"value":0,"_account_id":18893,"name":"Miles Gould","email":"mgould@redhat.com","username":"pozorvlak"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-04-05 10:22:37.000000000","_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},{"value":1,"date":"2017-04-04 17:24:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":18893,"name":"Miles Gould","email":"mgould@redhat.com","username":"pozorvlak"},{"value":2,"date":"2017-04-05 09:12:46.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-04-05 10:22:37.000000000","_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},{"value":0,"_account_id":18893,"name":"Miles Gould","email":"mgould@redhat.com","username":"pozorvlak"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":18893,"name":"Miles Gould","email":"mgould@redhat.com","username":"pozorvlak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-04-04 17:24:06.000000000","updated_by":{"_account_id":18893,"name":"Miles Gould","email":"mgould@redhat.com","username":"pozorvlak"},"reviewer":{"_account_id":18893,"name":"Miles Gould","email":"mgould@redhat.com","username":"pozorvlak"},"state":"REVIEWER"},{"updated":"2017-04-05 09:12:46.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2017-04-05 10:22:37.000000000","updated_by":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"reviewer":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"state":"REVIEWER"},{"updated":"2017-04-05 10:24:55.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"ebe4d68440957e959767f7f3858b9fd516e8a7da","author":{"_account_id":22600,"name":"Alexander Chuzhoy","email":"sasha@redhat.com","username":"achuzhoy"},"date":"2017-04-03 22:12:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76d934676ba9f1226ed8cce27346b25e410b753a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-03 22:16:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-docs-docs-ubuntu-xenial http://docs-draft.openstack.org/74/452974/1/check/gate-tripleo-docs-docs-ubuntu-xenial/a1ddef3//doc/build/html/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6214629568cc101a1db57ad00508deccdfeffaf8","author":{"_account_id":18893,"name":"Miles Gould","email":"mgould@redhat.com","username":"pozorvlak"},"date":"2017-04-04 16:53:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)\n\nLooks good apart from some minor nits inline.","accounts_in_message":[],"_revision_number":1},{"id":"4433f45374b131c314eb085c5e49f5febc069b1e","author":{"_account_id":22600,"name":"Alexander Chuzhoy","email":"sasha@redhat.com","username":"achuzhoy"},"date":"2017-04-04 17:10:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ca9bc3efb2fad7fb6b10718de1c25a7d4f9f8ef","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-04 17:12:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-docs-docs-ubuntu-xenial http://docs-draft.openstack.org/74/452974/2/check/gate-tripleo-docs-docs-ubuntu-xenial/58516d3//doc/build/html/ : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":2},{"id":"dfd899230a6d2a8098c4acb52ed1f8359aede783","author":{"_account_id":18893,"name":"Miles Gould","email":"mgould@redhat.com","username":"pozorvlak"},"date":"2017-04-04 17:13:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)\n\nGetting there; a couple more SPAG nits inline.","accounts_in_message":[],"_revision_number":2},{"id":"9cca38a65679e15049462bcff0b2456f45a6a432","author":{"_account_id":22600,"name":"Alexander Chuzhoy","email":"sasha@redhat.com","username":"achuzhoy"},"date":"2017-04-04 17:16:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"65223984d9ba2fdaaaefdac4e088cf768271e10a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-04 17:18:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-docs-docs-ubuntu-xenial http://docs-draft.openstack.org/74/452974/3/check/gate-tripleo-docs-docs-ubuntu-xenial/0b91297//doc/build/html/ : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":3},{"id":"d1ea9f6bee3c3a5f7b67d99cfeb7b457b5add2cd","author":{"_account_id":18893,"name":"Miles Gould","email":"mgould@redhat.com","username":"pozorvlak"},"date":"2017-04-04 17:24:06.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM.","accounts_in_message":[],"_revision_number":3},{"id":"60686399eb3b9b875f71423116aa41724cbe876a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2017-04-05 09:12:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ee5948489b62652de256168f591256a6ce91526a","author":{"_account_id":4328,"name":"Steven Hardy","email":"steven.hardy@suse.com","username":"shardy"},"date":"2017-04-05 10:22:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4a0ed140a742d8273f1eeaf3b8c401fd69d52e83","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-05 10:22:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"8b7f6ee526a37c6c83a5ec292aa200827c8da15a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-05 10:24:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-tripleo-docs-docs-ubuntu-xenial http://docs-draft.openstack.org/74/452974/3/gate/gate-tripleo-docs-docs-ubuntu-xenial/e56746c//doc/build/html/ : SUCCESS in 1m 42s","accounts_in_message":[],"_revision_number":3},{"id":"4c65432834ad408e98c1a27763c7efb502860280","date":"2017-04-05 10:24:56.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"26728800e353a93142e456c9b83b1c5e6062a56d","revisions":{"df2a2c37cf2a648705fb08da1026a8a71f67eef4":{"kind":"REWORK","_number":1,"created":"2017-04-03 22:12:43.000000000","uploader":{"_account_id":22600,"name":"Alexander Chuzhoy","email":"sasha@redhat.com","username":"achuzhoy"},"ref":"refs/changes/74/452974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/74/452974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/74/452974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/74/452974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/74/452974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/74/452974/1"}}},"commit":{"parents":[{"commit":"aedac83765a576f09955a293b8b9a5dbeea1c2b7","subject":"Remove os-cloud-config bits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/aedac83765a576f09955a293b8b9a5dbeea1c2b7"}]}],"author":{"name":"Sasha Chuzhoy","email":"sasha@redhat.com","date":"2017-04-03 22:12:14.000000000","tz":-240},"committer":{"name":"Sasha Chuzhoy","email":"sasha@redhat.com","date":"2017-04-03 22:12:42.000000000","tz":-240},"subject":"Accessing the ramdisk - selinux note","message":"Accessing the ramdisk - selinux note\n\nChange-Id: I49348b0bdf90a089fd2797d779abbeb135c5d436\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/df2a2c37cf2a648705fb08da1026a8a71f67eef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/df2a2c37cf2a648705fb08da1026a8a71f67eef4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aedac83765a576f09955a293b8b9a5dbeea1c2b7","is_merged_in_target_branch":true,"change_id":"If4a9dc29f7d9daac221222587cbc8309b7acf6c5","change_number":451088,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"05f6e98304cf9b7de881cfa7887386a194ae315e":{"kind":"REWORK","_number":2,"created":"2017-04-04 17:10:49.000000000","uploader":{"_account_id":22600,"name":"Alexander Chuzhoy","email":"sasha@redhat.com","username":"achuzhoy"},"ref":"refs/changes/74/452974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/74/452974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/74/452974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/74/452974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/74/452974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/74/452974/2"}}},"commit":{"parents":[{"commit":"aedac83765a576f09955a293b8b9a5dbeea1c2b7","subject":"Remove os-cloud-config bits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/aedac83765a576f09955a293b8b9a5dbeea1c2b7"}]}],"author":{"name":"Sasha Chuzhoy","email":"sasha@redhat.com","date":"2017-04-03 22:12:14.000000000","tz":-240},"committer":{"name":"Sasha Chuzhoy","email":"sasha@redhat.com","date":"2017-04-04 17:10:40.000000000","tz":-240},"subject":"Accessing the ramdisk - selinux note","message":"Accessing the ramdisk - selinux note\n\nChange-Id: I49348b0bdf90a089fd2797d779abbeb135c5d436\nSigned-off-by: Sasha Chuzhoy \u003csasha@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/05f6e98304cf9b7de881cfa7887386a194ae315e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/05f6e98304cf9b7de881cfa7887386a194ae315e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aedac83765a576f09955a293b8b9a5dbeea1c2b7","is_merged_in_target_branch":true,"change_id":"If4a9dc29f7d9daac221222587cbc8309b7acf6c5","change_number":451088,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"26728800e353a93142e456c9b83b1c5e6062a56d":{"kind":"REWORK","_number":3,"created":"2017-04-04 17:16:11.000000000","uploader":{"_account_id":22600,"name":"Alexander Chuzhoy","email":"sasha@redhat.com","username":"achuzhoy"},"ref":"refs/changes/74/452974/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/74/452974/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/74/452974/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/74/452974/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/74/452974/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/74/452974/3"}}},"commit":{"parents":[{"commit":"aedac83765a576f09955a293b8b9a5dbeea1c2b7","subject":"Remove os-cloud-config bits","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/aedac83765a576f09955a293b8b9a5dbeea1c2b7"}]}],"author":{"name":"Sasha Chuzhoy","email":"sasha@redhat.com","date":"2017-04-03 22:12:14.000000000","tz":-240},"committer":{"name":"Sasha Chuzhoy","email":"sasha@redhat.com","date":"2017-04-04 17:16:04.000000000","tz":-240},"subject":"Accessing the ramdisk - selinux note","message":"Accessing the ramdisk - selinux note\n\nChange-Id: I49348b0bdf90a089fd2797d779abbeb135c5d436\nSigned-off-by: Sasha Chuzhoy \u003csasha@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/26728800e353a93142e456c9b83b1c5e6062a56d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/26728800e353a93142e456c9b83b1c5e6062a56d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"aedac83765a576f09955a293b8b9a5dbeea1c2b7","is_merged_in_target_branch":true,"change_id":"If4a9dc29f7d9daac221222587cbc8309b7acf6c5","change_number":451088,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
