)]}'
{"id":"openstack%2Ftripleo-docs~478975","triplet_id":"openstack%2Ftripleo-docs~master~I466300231897b91ca6b08c7afb475a7bade6f3c4","project":"openstack/tripleo-docs","branch":"master","topic":"bp/tripleo-ceph-ansible","hashtags":[],"change_id":"I466300231897b91ca6b08c7afb475a7bade6f3c4","subject":"Document need of ceph-ansible in the undercloud","status":"ABANDONED","created":"2017-06-29 15:39:09.000000000","updated":"2017-07-25 22:11:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"97dff87c6a1de667754e2ad3699131b3aa8a3a0c","_number":478975,"virtual_id_number":478975,"owner":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"date":"2017-07-25 19:18:21.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":-1,"date":"2017-07-07 20:54:17.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2017-07-05 17:24:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},{"value":2,"date":"2017-06-30 08:35:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-06-29 16:01:58.000000000","updated_by":{"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},"reviewer":{"_account_id":8652,"name":"Attila Darazs","email":"adarazs@redhat.com","username":"adarazs"},"state":"REVIEWER"},{"updated":"2017-06-30 08:35:07.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2017-07-05 17:24:47.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2017-07-07 20:54:17.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-07-25 19:18:21.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"}],"messages":[{"id":"5135b36cc7d82b59fe0bd2e385e16f26752bb833","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2017-06-29 15:39:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fce747f81a1b285279110d517a93176ec0ea094e","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2017-06-29 15:46:59.000000000","message":"Topic set to bp/tripleo-ceph-ansible","accounts_in_message":[],"_revision_number":1},{"id":"9bb6a78085a2c2f89f46739e3fbff9e57a8bd367","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-29 15:53:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-docs-docs-ubuntu-xenial http://docs-draft.openstack.org/75/478975/1/check/gate-tripleo-docs-docs-ubuntu-xenial/1e2e68a//doc/build/html/ : SUCCESS in 8m 41s","accounts_in_message":[],"_revision_number":1},{"id":"11f7ba3c9cdab45e2737bad337717ff3528ea6d9","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2017-06-29 16:06:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"79f99b23dd0db029b7280db1f8fc39dd98bbbed2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-29 16:17:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-tripleo-docs-docs-ubuntu-xenial http://docs-draft.openstack.org/75/478975/2/check/gate-tripleo-docs-docs-ubuntu-xenial/5570a8e//doc/build/html/ : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":2},{"id":"4e7008284657e6852cd074249b1658193447a928","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2017-06-30 08:35:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3978affb83efc3ef3399490df23eee2889734dff","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2017-07-05 17:24:47.000000000","message":"Patch Set 2: Code-Review+2\n\nWhen the dep merges this should be good to go.","accounts_in_message":[],"_revision_number":2},{"id":"7e237d85a323d62fa4eb4c984084f70145d331cd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-07-07 20:54:17.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"8d2ec19512cd974a45b324c352855d17d1480494","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2017-07-25 19:18:21.000000000","message":"Patch Set 2:\n\nCould this be abandoned in place of https://review.openstack.org/487155","accounts_in_message":[],"_revision_number":2},{"id":"97dff87c6a1de667754e2ad3699131b3aa8a3a0c","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-07-25 22:11:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7c0b11631fb8c450af46e06020b7276d33c4fc0e","revisions":{"4c7af70a3a260ada79f58046d7842c98bf9271bb":{"kind":"REWORK","_number":1,"created":"2017-06-29 15:39:09.000000000","uploader":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"ref":"refs/changes/75/478975/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/75/478975/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/75/478975/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/75/478975/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/75/478975/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/75/478975/1"}}},"commit":{"parents":[{"commit":"85a74fc39653887260f0347c5cd9d03b4422ffd6","subject":"Merge \"Add important note for scaleout of custom roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/85a74fc39653887260f0347c5cd9d03b4422ffd6"}]}],"author":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2017-06-29 15:37:34.000000000","tz":120},"committer":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2017-06-29 15:37:34.000000000","tz":120},"subject":"Document need of ceph-ansible in the undercloud","message":"Document need of ceph-ansible in the undercloud\n\nWith the migration to ceph-ansible for the deployment of Ceph, it\nis necessary to install it manually in the undercloud to deploy\nCeph in the overcloud. This is on purpose so that people not\nusing Ceph won\u0027t need to install it. The deployment workflow is\nalways installed instead, via tripleo-common.\n\nChange-Id: I466300231897b91ca6b08c7afb475a7bade6f3c4\nDepends-On: I81d44a1e198c83a4ef8b109b4eb6c611555dcdc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/4c7af70a3a260ada79f58046d7842c98bf9271bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/4c7af70a3a260ada79f58046d7842c98bf9271bb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"85a74fc39653887260f0347c5cd9d03b4422ffd6","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"7c0b11631fb8c450af46e06020b7276d33c4fc0e":{"kind":"REWORK","_number":2,"created":"2017-06-29 16:06:47.000000000","uploader":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"ref":"refs/changes/75/478975/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/75/478975/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/75/478975/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/75/478975/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/75/478975/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/75/478975/2"}}},"commit":{"parents":[{"commit":"85a74fc39653887260f0347c5cd9d03b4422ffd6","subject":"Merge \"Add important note for scaleout of custom roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/85a74fc39653887260f0347c5cd9d03b4422ffd6"}]}],"author":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2017-06-29 15:37:34.000000000","tz":120},"committer":{"name":"Giulio Fidente","email":"gfidente@redhat.com","date":"2017-06-29 16:06:39.000000000","tz":120},"subject":"Document need of ceph-ansible in the undercloud","message":"Document need of ceph-ansible in the undercloud\n\nWith the migration to ceph-ansible for the deployment of Ceph, it\nis necessary to install it manually in the undercloud to deploy\nCeph in the overcloud. This is on purpose so that people not\nusing Ceph won\u0027t need to install it. The deployment workflow is\nalways installed instead, via tripleo-common.\n\nChange-Id: I466300231897b91ca6b08c7afb475a7bade6f3c4\nDepends-On: I81d44a1e198c83a4ef8b109b4eb6c611555dcdc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/7c0b11631fb8c450af46e06020b7276d33c4fc0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/7c0b11631fb8c450af46e06020b7276d33c4fc0e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"85a74fc39653887260f0347c5cd9d03b4422ffd6","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
