)]}'
{"id":"openstack%2Ftripleo-docs~655880","triplet_id":"openstack%2Ftripleo-docs~master~I6f1ebe3c2c24f968e118e0a61b1f8dcacae7a26f","project":"openstack/tripleo-docs","branch":"master","topic":"repositories_basic_deployment","hashtags":[],"change_id":"I6f1ebe3c2c24f968e118e0a61b1f8dcacae7a26f","subject":"Update the example.","status":"ABANDONED","created":"2019-04-26 11:31:38.000000000","updated":"2019-06-12 14:24:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c9a5324a38dfeb9319dbc132ad04776c66a8aa75","_number":655880,"virtual_id_number":655880,"owner":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"actions":{},"labels":{"Verified":{"rejected":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":-2,"date":"2019-06-12 01:15:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-06-12 01:10:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-06-11 07:39:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2019-06-11 07:39:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"value":2,"date":"2019-06-11 07:39:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-06-11 19:09:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-06-12 01:10:53.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-26 15:58:00.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-29 04:47:02.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-05-31 06:27:26.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-06-11 19:09:26.000000000","updated_by":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"reviewer":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"state":"REVIEWER"},{"updated":"2019-06-12 01:10:53.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-06-12 01:15:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8063c135f5df8e30c6b55347da42b107ddca1fb5","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-04-26 11:31:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f925cd22835dd1990babb52d6194fb33fb641ca1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-26 11:41:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/655880/1/check/openstack-tox-docs/51f87c0/html/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"eb686d6468cae129d128bccbc33e5555916091d0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-26 15:58:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"112f9cf7c094e9bcf0932b451ce61d3dbe4b1cbf","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-20 17:19:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"404672440f6279567865e325f92d130901a5520a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 17:36:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/655880/2/check/openstack-tox-docs/6e94c1d/html/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2},{"id":"fc749e9e099d1a1c88d013b5d79e440c7305acfa","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-05-21 21:35:07.000000000","message":"Patch Set 2:\n\nI think this patch can be merged.","accounts_in_message":[],"_revision_number":2},{"id":"e13c3ce22c6e017e778c8f78d82878800c888e33","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-05-29 04:47:02.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8330500b97232a0c6aa1434eabb563a2ff704aa1","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-05-31 06:27:26.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"502f53ebf66f2e81616b9938fff4946f8259811a","author":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"date":"2019-06-11 07:39:01.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"713e17b3905fb7ef01c31625fe69b5dd4846899e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 07:47:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/80/655880/3/check/openstack-tox-docs/6a02acf/html/ : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":3},{"id":"5de2f0cdfbb686d154aa59708da7e14b441eef9f","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-06-11 19:09:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"58ba20bbef589252811ed6bac4df241ae39f709d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-06-12 01:10:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"76ade8905b7436c473420bc88eb8946f3cd94f7d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 01:15:54.000000000","message":"Patch Set 3: Verified-2\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"7a61b90ad824d545456878c6635f7b9ca29a41c8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-12 14:24:46.000000000","message":"Abandoned\n\nI4b0f167f5cf9ee3b964ef809c435dc7403d260c7 took care of this","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"72d94aae00866f16f95f5ea05134aa17e382951b","revisions":{"c6773f4770a0ea04f298538b364f5df3c339e46c":{"kind":"REWORK","_number":1,"created":"2019-04-26 11:31:38.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/80/655880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/80/655880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/80/655880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/80/655880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/80/655880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/80/655880/1"}}},"commit":{"parents":[{"commit":"d5adb8342ece67e0a9cdcf185434c24b6a42d7b0","subject":"Merge \"Clarify DCN requirements for provider networks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/d5adb8342ece67e0a9cdcf185434c24b6a42d7b0"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-04-26 11:31:20.000000000","tz":120},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-04-26 11:31:20.000000000","tz":120},"subject":"Update the example.","message":"Update the example.\n\nUpdate the example variable environment with maintained versions.\n\nChange-Id: I6f1ebe3c2c24f968e118e0a61b1f8dcacae7a26f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/c6773f4770a0ea04f298538b364f5df3c339e46c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/c6773f4770a0ea04f298538b364f5df3c339e46c"}]},"branch":"refs/heads/master"},"b391b826237d3826fd55861d7b7146e2a2889587":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-20 17:19:07.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/80/655880/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/80/655880/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/80/655880/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/80/655880/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/80/655880/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/80/655880/2"}}},"commit":{"parents":[{"commit":"db9b91497e0b1527740888ec98cafd148fcfa29d","subject":"Merge \"Update the repositories file with stein.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/db9b91497e0b1527740888ec98cafd148fcfa29d"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-04-26 11:31:20.000000000","tz":120},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-05-20 17:18:56.000000000","tz":120},"subject":"Update the example.","message":"Update the example.\n\nUpdate the example variable environment with maintained versions.\n\nChange-Id: I6f1ebe3c2c24f968e118e0a61b1f8dcacae7a26f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/b391b826237d3826fd55861d7b7146e2a2889587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/b391b826237d3826fd55861d7b7146e2a2889587"}]},"branch":"refs/heads/master"},"72d94aae00866f16f95f5ea05134aa17e382951b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-06-11 07:39:01.000000000","uploader":{"_account_id":29222,"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","username":"hobbestigrou"},"ref":"refs/changes/80/655880/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/80/655880/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/80/655880/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/80/655880/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/80/655880/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/80/655880/3"}}},"commit":{"parents":[{"commit":"fe8548151c293be534b3e3d06e7355086e84b592","subject":"Fix formatting for mon_host_v1 example","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/fe8548151c293be534b3e3d06e7355086e84b592"}]}],"author":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-04-26 11:31:20.000000000","tz":120},"committer":{"name":"Natal Ngétal","email":"hobbestigrou@erakis.eu","date":"2019-06-11 07:38:48.000000000","tz":120},"subject":"Update the example.","message":"Update the example.\n\nUpdate the example variable environment with maintained versions.\n\nChange-Id: I6f1ebe3c2c24f968e118e0a61b1f8dcacae7a26f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/72d94aae00866f16f95f5ea05134aa17e382951b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/72d94aae00866f16f95f5ea05134aa17e382951b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
