)]}'
{"id":"openstack%2Ftripleo-docs~659297","triplet_id":"openstack%2Ftripleo-docs~master~Ide6f67a4643e9a7943a1271403def7de6721b890","project":"openstack/tripleo-docs","branch":"master","topic":"659297","hashtags":[],"change_id":"Ide6f67a4643e9a7943a1271403def7de6721b890","subject":"Update tag_from_label section for ceph containers","status":"MERGED","created":"2019-05-15 14:12:52.000000000","updated":"2019-05-28 12:18:36.000000000","submitted":"2019-05-28 12:18:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659297-1559045916927-399657d8","meta_rev_id":"f3ae771a9386b9ef1fbc1bc8acfdec91634e4331","_number":659297,"virtual_id_number":659297,"owner":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":2,"date":"2019-05-28 12:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"all":[{"value":2,"date":"2019-05-20 19:59:08.000000000","_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":2,"date":"2019-05-28 12:11:35.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"all":[{"value":0,"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"value":0,"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":1,"date":"2019-05-28 12:11:35.000000000","_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-16 01:46:26.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2019-05-20 19:59:08.000000000","updated_by":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"reviewer":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"state":"REVIEWER"},{"updated":"2019-05-28 12:11:35.000000000","updated_by":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"reviewer":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"state":"REVIEWER"},{"updated":"2019-05-28 12:18:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cb3c6d24cd190b3a600a85a70fba6d203a56b5ea","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-05-15 14:12:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"823a07450a541e6d7bdba65a7c58e18371957d46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 14:17:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/659297/1/check/openstack-tox-docs/24f4d1f/html/ : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"6b3bed86550eef71fd13de9c2a51a285e82dfc00","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-05-15 14:21:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"93a5c594137bc5a26876745cb3671d3b7f474ecf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 14:44:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/659297/2/check/openstack-tox-docs/6867cec/html/ : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"2f4414b7c8d1f7dbaf271d0c336a29d1e18b240d","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-05-15 15:52:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e8ebbe865ff429e13caa5d7825a11e9af3cf739a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-15 16:09:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/659297/3/check/openstack-tox-docs/0c4786c/html/ : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":3},{"id":"8905876ab3e8339eed96af996372e179f0b027ce","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-16 00:58:21.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"b1dcb28775cda56da571c216dd47f9209d281d8a","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2019-05-16 01:46:26.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c0b444515a6a799e68e9657d44f2ad471a7e56fb","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-05-16 07:39:07.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e017616082a56ec98c93b21a85bdb32b1dfb41f2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 07:43:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/659297/4/check/openstack-tox-docs/0a57692/html/ : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":4},{"id":"d9a335ebfaca1565427a6eb6df7e709cc86a3cd7","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-05-16 09:49:20.000000000","message":"Patch Set 4:\n\nI\u0027m not sure that providing an example in an admonition section is the best choice.\nAs john suggested, reminding to the next section and simply specifying that Ceph images in general doesn\u0027t follow the \"openstack way\" could be enough for us, but I\u0027ll think about a redesign and provide an example for ceph.","accounts_in_message":[],"_revision_number":4},{"id":"14def294d2c485e650d2388d41fc83fc2e917bf3","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-16 14:21:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f16ac743bac6aba1fe1bcd5db0217472c9a7a263","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-05-16 19:11:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1fb1fd8b506a224852f02ee7e645643b112e3c50","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 19:24:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/659297/5/check/openstack-tox-docs/cdec28f/html/ : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":5},{"id":"a94e3f2d13b722a1df2a19678e2c0f0fc0feecb8","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-16 20:49:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b6b4d280969aa3b28cd9663f26b31d633608284b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 20:54:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/659297/6/check/openstack-tox-docs/486138b/html/ : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":6},{"id":"fe7b741f68638f2892df36257c959abe07cde3e3","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-05-17 08:04:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"9c474d98afe68520e6a4f7c678542150e385b4d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 08:11:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/659297/7/check/openstack-tox-docs/05e5120/html/ : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":7},{"id":"53f2825ca3e2b0d14a0bbaeab4e095e2519bdc13","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-20 12:24:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"992112c5251cb160cf4a2fe148849691d9ab3140","author":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"date":"2019-05-20 18:54:44.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8e81d2a18165a46e1adb073ae192dd303beb323c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 19:04:57.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/659297/8/check/openstack-tox-docs/02a6705/html/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":8},{"id":"dd6c6384053193f585adb22d81d5d06e9d11e6bd","author":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"date":"2019-05-20 19:59:08.000000000","message":"Patch Set 8: Code-Review+2\n\n@stevebaker, Curious what you think of this revision.","accounts_in_message":[],"_revision_number":8},{"id":"6b8541de8641ebc973ca5cb2365812c758208fde","author":{"_account_id":6796,"name":"Giulio Fidente","email":"gfidente@redhat.com","username":"gfidente"},"date":"2019-05-28 12:11:35.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"e9e43fd20417220c2e01b3b4f71ee570cf03c8d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 12:11:47.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"5ffae9495f67f5daddf5503772faac0be7dd0ddf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 12:18:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"789817adffa4bfe416d9822ecccb899a6313d897","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-28 12:18:36.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/97/659297/8/gate/openstack-tox-docs/656c612/html/ : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"930b0a6d1609344bd1fa363f0efadbd936e6d2fa","revisions":{"6d9d05643cc7083b8ba7834a8c5aabe66d3f49b2":{"kind":"REWORK","_number":1,"created":"2019-05-15 14:12:52.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/97/659297/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/97/659297/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/1"}}},"commit":{"parents":[{"commit":"db9b91497e0b1527740888ec98cafd148fcfa29d","subject":"Merge \"Update the repositories file with stein.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/db9b91497e0b1527740888ec98cafd148fcfa29d"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-15 13:01:27.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-15 14:15:26.000000000","tz":120},"subject":"Update tag_from_label section for ceph containers","message":"Update tag_from_label section for ceph containers\n\nAdd admonition about how tag_from_label mechanism\nworks for external projects and clarify that it\ndoesn\u0027t work as expected for ceph containers, so\nin this case the ceph_tag defined in the container\nimage_prepare yaml file will be used.\n\nDepends-On: https://review.opendev.org/#/c/659236\nChange-Id: Ide6f67a4643e9a7943a1271403def7de6721b890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/6d9d05643cc7083b8ba7834a8c5aabe66d3f49b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/6d9d05643cc7083b8ba7834a8c5aabe66d3f49b2"}]},"branch":"refs/heads/master"},"adafa8df0852ef8637044a0442cdaf8a022f3581":{"kind":"REWORK","_number":2,"created":"2019-05-15 14:21:15.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/97/659297/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/97/659297/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/2"}}},"commit":{"parents":[{"commit":"db9b91497e0b1527740888ec98cafd148fcfa29d","subject":"Merge \"Update the repositories file with stein.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/db9b91497e0b1527740888ec98cafd148fcfa29d"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-15 13:01:27.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-15 14:23:43.000000000","tz":120},"subject":"Update tag_from_label section for ceph containers","message":"Update tag_from_label section for ceph containers\n\nAdd admonition about how tag_from_label mechanism\nworks for external projects and clarify that it\ndoesn\u0027t work as expected for ceph containers, so\nin this case the ceph_tag defined in the container\nimage_prepare yaml file will be used.\n\nDepends-On: https://review.opendev.org/#/c/659236\nChange-Id: Ide6f67a4643e9a7943a1271403def7de6721b890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/adafa8df0852ef8637044a0442cdaf8a022f3581"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/adafa8df0852ef8637044a0442cdaf8a022f3581"}]},"branch":"refs/heads/master"},"670ff4b2609de9b9499659c87a0bbec5e970fb5d":{"kind":"REWORK","_number":3,"created":"2019-05-15 15:52:29.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/97/659297/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/97/659297/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/3"}}},"commit":{"parents":[{"commit":"db9b91497e0b1527740888ec98cafd148fcfa29d","subject":"Merge \"Update the repositories file with stein.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/db9b91497e0b1527740888ec98cafd148fcfa29d"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-15 13:01:27.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-15 15:54:57.000000000","tz":120},"subject":"Update tag_from_label section for ceph containers","message":"Update tag_from_label section for ceph containers\n\nAdd admonition about how tag_from_label mechanism\nworks for external projects and clarify that it\ndoesn\u0027t work as expected for ceph containers, so\nin this case the ceph_tag defined in the container\nimage_prepare yaml file will be used.\n\nDepends-On: https://review.opendev.org/#/c/659236\nChange-Id: Ide6f67a4643e9a7943a1271403def7de6721b890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/670ff4b2609de9b9499659c87a0bbec5e970fb5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/670ff4b2609de9b9499659c87a0bbec5e970fb5d"}]},"branch":"refs/heads/master"},"a9367e53a98e4adf475ed4951ef567cc38117661":{"kind":"REWORK","_number":4,"created":"2019-05-16 07:39:07.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/97/659297/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/97/659297/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/4"}}},"commit":{"parents":[{"commit":"db9b91497e0b1527740888ec98cafd148fcfa29d","subject":"Merge \"Update the repositories file with stein.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/db9b91497e0b1527740888ec98cafd148fcfa29d"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-15 13:01:27.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-16 07:41:16.000000000","tz":120},"subject":"Update tag_from_label section for ceph containers","message":"Update tag_from_label section for ceph containers\n\nAdd admonition about how tag_from_label mechanism\nworks for external projects and clarify that it\ndoesn\u0027t work as expected for ceph containers, so\nin this case the ceph_tag defined in the container\nimage_prepare yaml file will be used.\n\nChange-Id: Ide6f67a4643e9a7943a1271403def7de6721b890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/a9367e53a98e4adf475ed4951ef567cc38117661"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/a9367e53a98e4adf475ed4951ef567cc38117661"}]},"branch":"refs/heads/master"},"4ffd8ee5195a065d6cc2eb9d99f96af8b1b7b705":{"kind":"REWORK","_number":5,"created":"2019-05-16 19:11:17.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/97/659297/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/97/659297/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/5"}}},"commit":{"parents":[{"commit":"db9b91497e0b1527740888ec98cafd148fcfa29d","subject":"Merge \"Update the repositories file with stein.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/db9b91497e0b1527740888ec98cafd148fcfa29d"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-15 13:01:27.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-16 19:13:47.000000000","tz":120},"subject":"Update tag_from_label section for ceph containers","message":"Update tag_from_label section for ceph containers\n\nAdd admonition about how tag_from_label mechanism\nworks for external projects and clarify that it\ndoesn\u0027t work as expected for ceph containers, so\nin this case the ceph_tag defined in the container\nimage_prepare yaml file will be used.\n\nChange-Id: Ide6f67a4643e9a7943a1271403def7de6721b890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/4ffd8ee5195a065d6cc2eb9d99f96af8b1b7b705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/4ffd8ee5195a065d6cc2eb9d99f96af8b1b7b705"}]},"branch":"refs/heads/master"},"ae9de8c3f1a1c17447395ce2002b6b3b2b7c4f5e":{"kind":"REWORK","_number":6,"created":"2019-05-16 20:49:31.000000000","uploader":{"_account_id":18002,"name":"John Fulton","email":"fulton@redhat.com","username":"fultonj"},"ref":"refs/changes/97/659297/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/97/659297/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/6"}}},"commit":{"parents":[{"commit":"db9b91497e0b1527740888ec98cafd148fcfa29d","subject":"Merge \"Update the repositories file with stein.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/db9b91497e0b1527740888ec98cafd148fcfa29d"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-15 13:01:27.000000000","tz":120},"committer":{"name":"John Fulton","email":"fulton@redhat.com","date":"2019-05-16 20:48:57.000000000","tz":-240},"subject":"Update tag_from_label section for ceph containers","message":"Update tag_from_label section for ceph containers\n\nAdd admonition about how tag_from_label mechanism\nworks for external projects and clarify that it\ndoesn\u0027t work as expected for ceph containers, so\nin this case the ceph_tag defined in the container\nimage_prepare yaml file will be used.\n\nChange-Id: Ide6f67a4643e9a7943a1271403def7de6721b890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/ae9de8c3f1a1c17447395ce2002b6b3b2b7c4f5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/ae9de8c3f1a1c17447395ce2002b6b3b2b7c4f5e"}]},"branch":"refs/heads/master"},"c1f931042756ca5e48622c4a6622edf11ebfd07d":{"kind":"REWORK","_number":7,"created":"2019-05-17 08:04:17.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/97/659297/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/97/659297/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/7"}}},"commit":{"parents":[{"commit":"db9b91497e0b1527740888ec98cafd148fcfa29d","subject":"Merge \"Update the repositories file with stein.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/db9b91497e0b1527740888ec98cafd148fcfa29d"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-15 13:01:27.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-17 08:06:44.000000000","tz":120},"subject":"Update tag_from_label section for ceph containers","message":"Update tag_from_label section for ceph containers\n\nAdd admonition about how tag_from_label mechanism\nworks for external projects and clarify that it\ndoesn\u0027t work as expected for ceph containers, so\nin this case the ceph_tag defined in the container\nimage_prepare yaml file will be used.\n\nChange-Id: Ide6f67a4643e9a7943a1271403def7de6721b890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/c1f931042756ca5e48622c4a6622edf11ebfd07d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/c1f931042756ca5e48622c4a6622edf11ebfd07d"}]},"branch":"refs/heads/master"},"930b0a6d1609344bd1fa363f0efadbd936e6d2fa":{"kind":"REWORK","_number":8,"created":"2019-05-20 18:54:44.000000000","uploader":{"_account_id":25402,"name":"Francesco Pantano","email":"fpantano@redhat.com","username":"fmount"},"ref":"refs/changes/97/659297/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/97/659297/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/97/659297/8"}}},"commit":{"parents":[{"commit":"db9b91497e0b1527740888ec98cafd148fcfa29d","subject":"Merge \"Update the repositories file with stein.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/db9b91497e0b1527740888ec98cafd148fcfa29d"}]}],"author":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-15 13:01:27.000000000","tz":120},"committer":{"name":"fpantano","email":"fpantano@redhat.com","date":"2019-05-20 18:56:29.000000000","tz":120},"subject":"Update tag_from_label section for ceph containers","message":"Update tag_from_label section for ceph containers\n\nAdd admonition about how tag_from_label mechanism\nworks for external projects and clarify that it\ndoesn\u0027t work as expected for ceph containers, so\nin this case the ceph_tag defined in the container\nimage_prepare yaml file will be used.\n\nChange-Id: Ide6f67a4643e9a7943a1271403def7de6721b890\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/930b0a6d1609344bd1fa363f0efadbd936e6d2fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/930b0a6d1609344bd1fa363f0efadbd936e6d2fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
