)]}'
{"id":"openstack%2Ftripleo-docs~672044","triplet_id":"openstack%2Ftripleo-docs~master~I465949fb71ca3963942da042f0ec6d2d4341061c","project":"openstack/tripleo-docs","branch":"master","hashtags":[],"change_id":"I465949fb71ca3963942da042f0ec6d2d4341061c","subject":"[tripleo-ci][docs] Update Grafana link after recent migration","status":"MERGED","created":"2019-07-22 13:30:18.000000000","updated":"2019-07-24 10:06:05.000000000","submitted":"2019-07-24 10:06:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672044-1563962765358-d6eb9d84","meta_rev_id":"b87463d281e5f6ef5303471cf9b63401af2adab5","_number":672044,"virtual_id_number":672044,"owner":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":2,"date":"2019-07-24 10:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":2,"date":"2019-07-22 14:05:01.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-24 09:56:07.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"all":[{"value":0,"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-24 09:56:07.000000000","_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 13:32:58.000000000","updated_by":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"reviewer":{"_account_id":8175,"name":"Rafael Folco","email":"rfolco@redhat.com","username":"rfolco"},"state":"REVIEWER"},{"updated":"2019-07-22 13:33:05.000000000","updated_by":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2019-07-22 14:05:01.000000000","updated_by":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"reviewer":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"state":"REVIEWER"},{"updated":"2019-07-24 09:56:07.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-07-24 10:06:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ceada10d9c2a3e66052ce3e1163fd7b6ea37639","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-22 13:30:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7bbbdeae87c21be615043cc25f035c5bbde3e732","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 13:37:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/672044/1/check/openstack-tox-docs/85e7193/html/ : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"9422261176b3af053d2b664c3ca720a1c0aba2d9","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-22 13:47:22.000000000","message":"Patch Set 1: Code-Review-1\n\n16:34 #tripleo: \u003c sshnaidm\u003e marios|ruck|call, use cockpit-ci.tripleo.org it should point to cockpit dashboard","accounts_in_message":[],"_revision_number":1},{"id":"fe8630c02000a04758bbd7d40a7d2e48a0d65157","author":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"date":"2019-07-22 14:02:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d0934a6808281f59851cc244cd2a504e9bd111d","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2019-07-22 14:05:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d8b12386618ae32569adaa6e279d5dba31a8485a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 14:13:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/672044/2/check/openstack-tox-docs/fabfcac/html/ : SUCCESS in 4m 30s","accounts_in_message":[],"_revision_number":2},{"id":"3b06a68ea064f6315b5c7909366b14d633717aff","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-24 09:56:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ac419ab0a78f6d2c42d13a050e476ecaad23e55f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 09:56:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e67bdce59e1af8d9ab2f921cb090e5f62964df06","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 10:06:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/44/672044/2/gate/openstack-tox-docs/a35fd49/html/ : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":2},{"id":"b87463d281e5f6ef5303471cf9b63401af2adab5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 10:06:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"53eacdcd05d24f59eb779a0e57416c3f7ef8d87e","revisions":{"1495e3c98f78488125d0f08125eb5667a18a6b7b":{"kind":"REWORK","_number":1,"created":"2019-07-22 13:30:18.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/44/672044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/44/672044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/44/672044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/44/672044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/44/672044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/44/672044/1"}}},"commit":{"parents":[{"commit":"7338a81c3f9428598bc5f3fc7b9f0ae81998cb71","subject":"Merge \"RHSM: updates about the service path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/7338a81c3f9428598bc5f3fc7b9f0ae81998cb71"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-07-22 13:28:28.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-07-22 13:28:28.000000000","tz":180},"subject":"[tripleo-ci][docs] Update Grafana link after recent migration","message":"[tripleo-ci][docs] Update Grafana link after recent migration\n\nThe tripleo-ci squad grafana instance was just moved to a new host\nand we need to update the url\n\nChange-Id: I465949fb71ca3963942da042f0ec6d2d4341061c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/1495e3c98f78488125d0f08125eb5667a18a6b7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/1495e3c98f78488125d0f08125eb5667a18a6b7b"}]},"branch":"refs/heads/master"},"53eacdcd05d24f59eb779a0e57416c3f7ef8d87e":{"kind":"REWORK","_number":2,"created":"2019-07-22 14:02:12.000000000","uploader":{"_account_id":8449,"name":"Marios Andreou","email":"marios.andreou@gmail.com","username":"marios"},"ref":"refs/changes/44/672044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/44/672044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/44/672044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/44/672044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/44/672044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/44/672044/2"}}},"commit":{"parents":[{"commit":"7338a81c3f9428598bc5f3fc7b9f0ae81998cb71","subject":"Merge \"RHSM: updates about the service path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/7338a81c3f9428598bc5f3fc7b9f0ae81998cb71"}]}],"author":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-07-22 13:28:28.000000000","tz":180},"committer":{"name":"Marios Andreou","email":"marios@redhat.com","date":"2019-07-22 14:03:31.000000000","tz":180},"subject":"[tripleo-ci][docs] Update Grafana link after recent migration","message":"[tripleo-ci][docs] Update Grafana link after recent migration\n\nThe tripleo-ci squad grafana instance was just moved to a new host\nand we need to update the url http://cockpit-ci.tripleo.org/\n\nChange-Id: I465949fb71ca3963942da042f0ec6d2d4341061c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/53eacdcd05d24f59eb779a0e57416c3f7ef8d87e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/53eacdcd05d24f59eb779a0e57416c3f7ef8d87e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
