)]}'
{"id":"openstack%2Ftripleo-docs~672189","triplet_id":"openstack%2Ftripleo-docs~master~Idbcd60c89ed3e078c4ee41743e4d9955bae2e86d","project":"openstack/tripleo-docs","branch":"master","topic":"undercloud-minion","hashtags":[],"change_id":"Idbcd60c89ed3e078c4ee41743e4d9955bae2e86d","subject":"Minion doc","status":"MERGED","created":"2019-07-22 22:43:55.000000000","updated":"2019-08-12 15:15:25.000000000","submitted":"2019-08-12 15:15:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672189-1565622925777-d258f855","meta_rev_id":"b64f7f9de1cc7191b276922ac3cc4cd15dec3878","_number":672189,"virtual_id_number":672189,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-08-12 15:15:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-05 19:46:22.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-08 18:10:24.000000000","_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-08-09 11:09:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":1,"date":"2019-08-12 15:06:50.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-05 19:46:22.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-08 18:10:24.000000000","updated_by":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2019-08-09 11:09:37.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-08-12 15:15:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9189f154d24d2aa6a92fcadf81b9014c53f5203","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-22 22:43:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e56a4b3de4d26bf34099a68049819b95a4ae4d57","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-22 22:44:17.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"866c4c5a8857c45150887f54b10249145cad6cb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 22:52:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/672189/2/check/openstack-tox-docs/1ec5e38/ : FAILURE in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"a4caf672924a73191228bde73f85c746fc567cbf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-25 14:52:42.000000000","message":"Topic changed from undercloud-minion to bp/undercloud-minion","accounts_in_message":[],"_revision_number":2},{"id":"68d88239a9d5da83a46a21b71e9df77fb4a24f19","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-25 23:26:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cbb72bf4533f9560b64a00c5da31c55a2ffaa6f5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-25 23:34:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/89/672189/3/check/openstack-tox-docs/266865d/ : FAILURE in 3m 22s","accounts_in_message":[],"_revision_number":3},{"id":"0a75e83860d7e2fd9afef8164997bddcaa5b57f2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 19:21:19.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c5f7a5a9be8b22a7894f5b80269de1c9bdceaba6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 19:25:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/89/672189/4/check/openstack-tox-docs/c964a6f/html/ : SUCCESS in 3m 04s","accounts_in_message":[],"_revision_number":4},{"id":"2cdb299b9709a049cf0b774c1640549acb0df0c6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 19:30:16.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ef2987988512f1adfbf9f6f5f34a229a8f9a0725","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 19:35:11.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a4b763bf36f9fbb5acbbbc3da77de041205d314c","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 19:35:18.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3fffc9456fd8898d1e1a597aa80fc0b9542ddc20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 19:40:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/89/672189/5/check/openstack-tox-docs/00feae3/html/ : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":5},{"id":"8247a58da87bccb9b8a8a0b3c5af0cf2cc52df69","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 19:46:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5c46a4ec606be5eb8026a8da6c5bd916c3eeac37","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 19:46:22.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"69e268f00461fed63842deb2d668ea64aa6c64aa","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-05 20:24:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"af4b4535167be74447cea99c86c1c55d3793e5ee","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-08-08 18:10:24.000000000","message":"Patch Set 5: Code-Review+2\n\nlooks good to me, I would consider removing anything f28 at this point though.","accounts_in_message":[],"_revision_number":5},{"id":"7e44e89417b897d6f3e41df6dcefa81df943cdfd","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-08-09 11:09:37.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"28ec2d60eec07a03d64576f082b996107f73d79e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-12 15:06:50.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"a504aa551738457e9af1240ddae1dc7eb9a7331d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:07:01.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8f441eb6300803360144a5d69aec0054db98bcd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:15:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"c9d75cd8784f78bfa5e5409c46e4f16ddea41e30","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-12 15:15:25.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://logs.opendev.org/89/672189/5/gate/openstack-tox-docs/cdcf33d/html/ : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"70bf8832131942008847756f7b302ff27997c5ed","revisions":{"fae421a8ccc951a6431600a20f2dcdc953bef28c":{"kind":"REWORK","_number":1,"created":"2019-07-22 22:43:55.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/89/672189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/89/672189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/1"}}},"commit":{"parents":[{"commit":"62cb1470f7685fe37adb8e62e623558356814361","subject":"Merge \"Remove midonet manual\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/62cb1470f7685fe37adb8e62e623558356814361"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-22 22:43:48.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-22 22:43:48.000000000","tz":-360},"subject":"WIP: Minion doc","message":"WIP: Minion doc\n\nChange-Id: Idbcd60c89ed3e078c4ee41743e4d9955bae2e86d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/fae421a8ccc951a6431600a20f2dcdc953bef28c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/fae421a8ccc951a6431600a20f2dcdc953bef28c"}]},"branch":"refs/heads/master"},"66ee6e8a0e385123d39ee6e01d1268c806a9d405":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-07-22 22:44:17.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/89/672189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/89/672189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/2"}}},"commit":{"parents":[{"commit":"62cb1470f7685fe37adb8e62e623558356814361","subject":"Merge \"Remove midonet manual\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/62cb1470f7685fe37adb8e62e623558356814361"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-22 22:43:48.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-22 22:44:03.000000000","tz":-360},"subject":"WIP: Minion doc","message":"WIP: Minion doc\n\nRelated-Blueprint: undercloud-minion\nChange-Id: Idbcd60c89ed3e078c4ee41743e4d9955bae2e86d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/66ee6e8a0e385123d39ee6e01d1268c806a9d405"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/66ee6e8a0e385123d39ee6e01d1268c806a9d405"}]},"branch":"refs/heads/master"},"37e45a197b1a0a0c628d855e04744c64907e74bf":{"kind":"REWORK","_number":3,"created":"2019-07-25 23:26:17.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/89/672189/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/89/672189/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/3"}}},"commit":{"parents":[{"commit":"62cb1470f7685fe37adb8e62e623558356814361","subject":"Merge \"Remove midonet manual\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/62cb1470f7685fe37adb8e62e623558356814361"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-22 22:43:48.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-25 23:26:13.000000000","tz":-360},"subject":"WIP: Minion doc","message":"WIP: Minion doc\n\nRelated-Blueprint: undercloud-minion\nChange-Id: Idbcd60c89ed3e078c4ee41743e4d9955bae2e86d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/37e45a197b1a0a0c628d855e04744c64907e74bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/37e45a197b1a0a0c628d855e04744c64907e74bf"}]},"branch":"refs/heads/master"},"f336c139f68609de9dd3907b37ad3c01857dd800":{"kind":"REWORK","_number":4,"created":"2019-08-05 19:21:19.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/89/672189/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/89/672189/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/4"}}},"commit":{"parents":[{"commit":"130aa09fdba2e8dd99ef192e70fd6bf3146bbd65","subject":"Correcting typo in achieve spelling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/130aa09fdba2e8dd99ef192e70fd6bf3146bbd65"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-22 22:43:48.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-05 19:18:56.000000000","tz":-360},"subject":"Minion doc","message":"Minion doc\n\nDocument the basic install for the undercloud minion\n\nRelated-Blueprint: undercloud-minion\nChange-Id: Idbcd60c89ed3e078c4ee41743e4d9955bae2e86d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/f336c139f68609de9dd3907b37ad3c01857dd800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/f336c139f68609de9dd3907b37ad3c01857dd800"}]},"branch":"refs/heads/master"},"70bf8832131942008847756f7b302ff27997c5ed":{"kind":"REWORK","_number":5,"created":"2019-08-05 19:35:11.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/89/672189/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tripleo-docs","ref":"refs/changes/89/672189/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tripleo-docs refs/changes/89/672189/5"}}},"commit":{"parents":[{"commit":"130aa09fdba2e8dd99ef192e70fd6bf3146bbd65","subject":"Correcting typo in achieve spelling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/130aa09fdba2e8dd99ef192e70fd6bf3146bbd65"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-07-22 22:43:48.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-05 19:35:05.000000000","tz":-360},"subject":"Minion doc","message":"Minion doc\n\nDocument the basic install for the undercloud minion\n\nRelated-Blueprint: undercloud-minion\nChange-Id: Idbcd60c89ed3e078c4ee41743e4d9955bae2e86d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/70bf8832131942008847756f7b302ff27997c5ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tripleo-docs/commit/70bf8832131942008847756f7b302ff27997c5ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
